)]}'
{"id":"openstack%2Fkayobe~852103","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I7e10b851016253bf85625e9daa89416c738242e8","project":"openstack/kayobe","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-08-08 10:43:19.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-08-08 07:59:53.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I7e10b851016253bf85625e9daa89416c738242e8","subject":"ovn: add network group to neutron-ovn-metadata-agent","status":"MERGED","created":"2022-08-05 15:14:46.000000000","updated":"2022-08-08 10:44:40.000000000","submitted":"2022-08-08 10:43:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852103","cherry_pick_of_change":850608,"cherry_pick_of_patch_set":1,"meta_rev_id":"20e45356ddd538885d8e5bc939f0a4cdf2fb0e42","_number":852103,"virtual_id_number":852103,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-08 10:43:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-08-08 07:40:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-08 07:59:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-08 07:59:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-05 15:14:46.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-08-05 17:00:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-08 07:40:21.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-08-08 07:40:21.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"153397b587a99c5575bba11d4107dcd12449706d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-08-05 15:14:46.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d129448e833408e44c02f912feaed7a9db4fee07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 17:00:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/baefa70431d64a6cbd9d1a97ad285b58\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db8420f5e2004b2ca6463f9d266c6305 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de91d2a75d98461ca391286e1d089945 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e0ccc88bd36478384a5a8a428d3db75 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed521b5a044e49aebb1283c5b04246f3 : SUCCESS in 7m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5869ad4ee88e4624ba83bdfef537dbd7 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56cc25a3daff4ecdb4473afa103cabba : SUCCESS in 9m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6277a51c09864388a3ad682c2d8332b8 : SUCCESS in 6m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1533ef3c37574bc39061770130c331ae : SUCCESS in 11m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/32b83e73e2554ed9899b2132082862be : SUCCESS in 19m 16s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/984cf05aedf9418cb0b88f70feebd241 : SUCCESS in 59m 59s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7716e4513b44ada937876ba829f13e8 : SUCCESS in 52m 40s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/881e9648a5b546889960d09f454c681c : SUCCESS in 1h 10m 22s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a8261392886745708ca28adf4fd5f026 : SUCCESS in 23m 29s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c90340fb8fec45fb8a548642942885ad : SUCCESS in 15m 25s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6ca9d4b515d74cf2b8362aae7b7a164b : SUCCESS in 1h 36m 58s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4e3790acca0e4927bc62bc8340729ff1 : SUCCESS in 21m 03s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc2301c42d254f6d8f7d60cd9584bab7 : SUCCESS in 15m 55s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/f87a69aa6bd94af6a035964cae94b943 : SUCCESS in 50m 19s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7e952faae8ea4e52ba8af52034b03545 : SUCCESS in 27m 18s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c8a388ef7642419aaf5df5204ac0738f : SUCCESS in 13m 04s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74096257331147079d1c82404966ad19 : SUCCESS in 12m 16s","accounts_in_message":[],"_revision_number":1},{"id":"a7fb947a73bee8321f86f2de0fa180b2864883f4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-08-08 07:40:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ea35566f084ff6f9b48869c58564e73925971c9f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-08-08 07:59:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c22bac0e8212dfbe4f89c494fb8c670b16ceecb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 08:00:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f8b2885d131b771445d52592fbbb2f3e47fc9e41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 10:43:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/501280d5c5c64d31ab5f49f1e503534d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffcf4f00011c456dbab77838e511bdee : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c90dfc3084d4241be89d5ec8e7385dc : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bcc565dfc29433f9e70f9b8e9b74cb0 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55fd9175f5f940f594d84a493bc20608 : SUCCESS in 8m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1d511a1897354aaa96b30715979167e6 : SUCCESS in 6m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a94c735fd7f1411ea9e7a06a980e2067 : SUCCESS in 12m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ea5d9398274b4a3db639d9af0c6c4005 : SUCCESS in 15m 53s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a14fdc0fdd3b4319994010536af61570 : SUCCESS in 1h 18m 36s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a1bbc255bf04a17ae2932a146a5d8bf : SUCCESS in 54m 07s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/110d4ec3007246439fadff7c3bed65e9 : SUCCESS in 1h 18m 59s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5c61f2115f704d7bafdf466d7484b80d : SUCCESS in 18m 31s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f66f77c2d4c942a1bde9bdd53409f45a : SUCCESS in 20m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b3a22afba5ec4b03890198b88fec69e9 : SUCCESS in 1h 46m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b264166543f5406db1fb95426c30c62d : SUCCESS in 17m 31s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/754326ca57dd41dcbb64d9662a474694 : SUCCESS in 17m 07s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c6861b2f2351428aa15025123a700b95 : SUCCESS in 31m 38s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/28a7ea0987b44f0cbfb71dc94a1169bc : SUCCESS in 19m 51s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4d6797fea9a42ffbe8d9ac7f60a8ef7 : SUCCESS in 10m 13s","accounts_in_message":[],"_revision_number":1},{"id":"fc860d73a3a881a63f519f84ea7836b7ad590098","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 10:43:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"20e45356ddd538885d8e5bc939f0a4cdf2fb0e42","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 10:44:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cca601ae89654342a3564fd5dc40c636\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b245d5d095c4e09bcbf0e74b440e555 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dcdbceac8d5a1f9aa06bc44d68a6804b9f49790b","revisions":{"dcdbceac8d5a1f9aa06bc44d68a6804b9f49790b":{"kind":"REWORK","_number":1,"created":"2022-08-05 15:14:46.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/03/852103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/852103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/852103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/852103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/852103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/852103/1"}}},"commit":{"parents":[{"commit":"98fbc2f2d3bb089785f22538b270402044be34a5","subject":"Add missing declaration of kolla_enable_hacluster","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/98fbc2f2d3bb089785f22538b270402044be34a5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-07-21 06:18:00.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-08-05 15:14:46.000000000","tz":0},"subject":"ovn: add network group to neutron-ovn-metadata-agent","message":"ovn: add network group to neutron-ovn-metadata-agent\n\nFollow up after I111df79f08eb76a735f33c6edf03b5a302fb6cc3 in kolla-ansible\n\nChange-Id: I7e10b851016253bf85625e9daa89416c738242e8\n(cherry picked from commit 5ee99c59481e4fb31119a418248fb4823d6a63ae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dcdbceac8d5a1f9aa06bc44d68a6804b9f49790b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dcdbceac8d5a1f9aa06bc44d68a6804b9f49790b"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
