)]}'
{"id":"openstack%2Fkayobe~852104","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I7e10b851016253bf85625e9daa89416c738242e8","project":"openstack/kayobe","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-08-08 10:43:21.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-08-08 07:59:57.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I7e10b851016253bf85625e9daa89416c738242e8","subject":"ovn: add network group to neutron-ovn-metadata-agent","status":"MERGED","created":"2022-08-05 15:14:57.000000000","updated":"2022-08-08 10:44:40.000000000","submitted":"2022-08-08 10:43:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852104","cherry_pick_of_change":850608,"cherry_pick_of_patch_set":1,"meta_rev_id":"0b469a9798b8819e0d6f98e634d17ac6d7e5f184","_number":852104,"virtual_id_number":852104,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-08 10:43:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-08-08 07:40:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-08 07:59:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-08 07:59:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-05 15:14:57.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-08-05 16:28:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-08 07:40:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-08-08 07:40:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"5bb9df7b6fff277b63a464003a4204c0478c5940","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-08-05 15:14:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"136581557e768eafd94118db577cc8326350c37f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 16:28:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a54191973ada49e48d3f5b069b1e56a3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5b47563e50043188be1fd523228d7b3 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1627377bf60d4b7a9d116b0d560b7bc9 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/414428ae5c63404fbc8049cbb4bef527 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d72c44a4f36c452e9d2371c8d4c2a20b : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab3501dec6a3428abac56c0b4cb60463 : SUCCESS in 7m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/73ddee4c685a45a0b879dd75eb16d385 : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f0bb3459d450482f81a43caf63d040df : SUCCESS in 9m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/06d6ae5376f44b398c615a37b76a34b0 : SUCCESS in 17m 34s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7971fc41c48e4aaaac4d377ea427d19b : SUCCESS in 1h 07m 23s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0fe627a57f1d47a98dca4f40c7dd21ae : SUCCESS in 1h 03m 15s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c0aa79dedc7847f991d3014abbdd254c : SUCCESS in 15m 32s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/715de2018aff479d8d40eef0d17aae7a : FAILURE in 16m 40s (non-voting)\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/644a76dde18d49f88a8eb4d2be49f480 : SUCCESS in 17m 10s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6bc61cff7d394a69b558467f85dce9ab : FAILURE in 15m 55s (non-voting)\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/31624a390e3946e9967173c9426ea1c1 : SUCCESS in 16m 17s","accounts_in_message":[],"_revision_number":1},{"id":"e2bafb114c6af1274f0c1452deb316a6587cfeb3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-08-08 07:40:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"610f076c7c86f3367375b7643b9d3d04674e23dc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-08-08 07:59:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7f2e18060cb05cd6b5eddb3afec607dee91b0605","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 08:00:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7afe3a0a371ed10a5bf5ecf0cea709d66a9d20fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 10:43:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9acd547b09f43e095691b3e7811c27b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6262586699d946d89842cf8616705022 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89cd7d0cd9b646c2a2508f0e5df1d9b7 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa82570a9f2842acafff27a910902e08 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98b2fb9662684d7482605e47dc1bf02b : SUCCESS in 7m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/500e4281818643ddb5ce41a800105278 : SUCCESS in 7m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/21edcc254fef4a7186a688050604b1c1 : SUCCESS in 9m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b3da6094e60c42b2a1bd73dfe94f4908 : SUCCESS in 13m 53s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/539be78eb54f44ea909e1130335eacec : SUCCESS in 1h 06m 04s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/eb95890d42a64adb98e35f7be06b4865 : SUCCESS in 1h 18m 06s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/070febc3167d4bc5adf7cbeef0503f45 : SUCCESS in 18m 04s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6604794a019e481089dd24a4330d17b9 : SUCCESS in 14m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7d56e030cbed48219fb8680880f04900 : SUCCESS in 16m 26s","accounts_in_message":[],"_revision_number":1},{"id":"127586519d70b423c5fd7ca6aceb118346c6edf2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 10:43:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0b469a9798b8819e0d6f98e634d17ac6d7e5f184","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 10:44:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33b9575d243f4d2fa3a26d495db9a5a3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a7db157839b45bd9a9d6cdcde19eba0 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5853c7a2039b11d33723c2caf14c7dc7a065e2a8","revisions":{"5853c7a2039b11d33723c2caf14c7dc7a065e2a8":{"kind":"REWORK","_number":1,"created":"2022-08-05 15:14:57.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/04/852104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/852104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/852104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/852104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/852104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/852104/1"}}},"commit":{"parents":[{"commit":"7faade5db6e715b03f45382e706fca09b175e51b","subject":"Use net_mask filter instead of ansible\u0027s ipaddr","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7faade5db6e715b03f45382e706fca09b175e51b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-07-21 06:18:00.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-08-05 15:14:57.000000000","tz":0},"subject":"ovn: add network group to neutron-ovn-metadata-agent","message":"ovn: add network group to neutron-ovn-metadata-agent\n\nFollow up after I111df79f08eb76a735f33c6edf03b5a302fb6cc3 in kolla-ansible\n\nChange-Id: I7e10b851016253bf85625e9daa89416c738242e8\n(cherry picked from commit 5ee99c59481e4fb31119a418248fb4823d6a63ae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5853c7a2039b11d33723c2caf14c7dc7a065e2a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5853c7a2039b11d33723c2caf14c7dc7a065e2a8"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
