)]}'
{"id":"openstack%2Fkayobe~852106","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I86c92a1b0b8cf5e4dc534e04709e38bd90f7660e","project":"openstack/kayobe","branch":"stable/xena","topic":"rocky-image-stable/xena","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-08-08 09:39:23.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I86c92a1b0b8cf5e4dc534e04709e38bd90f7660e","subject":"Update to latest Rocky Linux cloud image","status":"MERGED","created":"2022-08-05 21:49:08.000000000","updated":"2022-08-08 11:54:04.000000000","submitted":"2022-08-08 11:52:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852106-rocky-image-stable/xena","cherry_pick_of_change":848649,"cherry_pick_of_patch_set":2,"meta_rev_id":"a3b3154fca8e1c7e13260cc2dab9f7d72d05935d","_number":852106,"virtual_id_number":852106,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-08 11:52:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-08-08 09:31:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-08 09:39:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-08 09:39:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-05 23:36:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-08 09:31:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"dfddf2a0fc9c5d99013001f36baef6658401b99c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-08-05 21:49:08.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7d157a53440db524227783220f1c27c4deb169e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 23:36:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2412c696eb1442a8ba454bb3384dc4c9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a5140a0f3c14ee1839bf0cfb9447d55 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e2375b898a4460ea617611c03d2907a : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fb97f76eaba4c1d99a5caec732dfc27 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00998d046565406084837a1b7a14b4fa : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8738364990f24360b3e88681a114a797 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75f6ac1905a447328ea0eef62371d3c0 : SUCCESS in 8m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27645a652e284a67b54c06ea9e44b343 : SUCCESS in 3m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5c7418f92e0f4fe19a97357884941e40 : SUCCESS in 5m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/50c22c57ebf6431cb563f27fdc5f2108 : SUCCESS in 13m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6d2350dde2b4454d8bfefa403888afc0 : SUCCESS in 17m 59s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/132f5b96f0104c68b5dc00909970530b : SUCCESS in 1h 12m 26s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/d8e1d081579a4c1998e59fba2e21a4cc : SUCCESS in 1h 15m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e5b360ab58048f0af1a43142d602e13 : SUCCESS in 54m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f6aacd85022e4967a19412a1d9589247 : SUCCESS in 1h 05m 32s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d486b94fd3244fb1816128204d23bf48 : SUCCESS in 25m 42s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/02a41f468bd84735a9495afc5b5f9387 : SUCCESS in 23m 27s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af6b032c9fbe4ce88833a2d81393ea39 : SUCCESS in 19m 10s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/198483e9957e486996c1abcfd4babfdf : SUCCESS in 1h 42m 03s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b83ba80f66784827b0f881ce94f2ce67 : SUCCESS in 1h 14m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/18f8385ea4a2463aad29a4f60c60c1e9 : SUCCESS in 24m 15s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/36c66cdc6c7a4a4f8376c9eb43d90645 : SUCCESS in 18m 54s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c72d733b8f0b4a689fd80f5964a26c09 : SUCCESS in 16m 57s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/83e554f829d44fcf81383d1523fd0170 : SUCCESS in 49m 37s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1bd562c4cda7484a962d5dad9185e7c7 : SUCCESS in 30m 45s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43e0a6344ff04abdb0a512cadd78e7e3 : SUCCESS in 21m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d6cfc42f0dc34c8a951dc02e0b25ab43 : SUCCESS in 19m 00s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/3159209c00a7413cb5050cfbe15aa1d3 : SUCCESS in 16m 16s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c539b27d7b9408c9de1a0221e6e5a8d : SUCCESS in 15m 11s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/a2cdc8572d7e49348334c135161f3030 : SUCCESS in 16m 17s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c8b40bf91cf64890976ead4a4d69acdd : SUCCESS in 19m 58s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea6b669971af417089d845ce44e26f05 : SUCCESS in 16m 15s","accounts_in_message":[],"_revision_number":1},{"id":"4421862b2a0d24af1fa52fd64df2933cef79b5b2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-08-08 09:31:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b061a64441d194923342b5b05a6650217faa378d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-08-08 09:39:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bc7f43eb7c627dc8393381136b61a1117d52a333","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 09:40:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"564512bd9dfae2ef1c80239927c003845675e419","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 11:52:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e376443c0f34287b331a3049171968a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8cb52bc8b72469f9e818eafaa3923a0 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d82ebbdf32d4e338d3f4ae12c829a08 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0872ab7d702c48df81b0badbcdf4c761 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a179c5152a804aaf9ec3eccecffe14b3 : SUCCESS in 9m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd29c16de30a4988bdade468051e51c5 : SUCCESS in 4m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1fdc7abe94ab40f08b3d1434e53381e2 : SUCCESS in 5m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b70e76c0d283432290b929ca3ce93d5a : SUCCESS in 14m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0f470a75709b4971a685e65a1b59a8e8 : SUCCESS in 16m 53s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6e4b48eebe28486e9fb6bd3fb6a56433 : SUCCESS in 53m 45s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/665513c511054fa4972b3fc509201623 : SUCCESS in 1h 16m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af2941a8521a4efd8d0af7400cf3cec4 : SUCCESS in 46m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4b1be832cc73497096f992d6d3dd4f32 : SUCCESS in 1h 01m 55s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/08523424818a45c5b7217993c817fb62 : SUCCESS in 25m 30s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/b232aafd5b8d41ab9b5a2cc721bd8fb5 : SUCCESS in 23m 43s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dc5c02bd92a42edb0737afe49fdd4fc : SUCCESS in 15m 18s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/91090ec242a94dcb8fbd5924e977da69 : SUCCESS in 1h 31m 28s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b914cc46ff0414591e4982f26d57712 : SUCCESS in 59m 19s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e4725af17ce04009aba6edc3a547074a : SUCCESS in 26m 29s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/69c7c46b1cd6467a9d5b4804af51e3f5 : SUCCESS in 19m 52s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e733b4ba662c4bf495d3ef0c995f273c : SUCCESS in 13m 49s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cd8a18770fd2431984434953099a9486 : SUCCESS in 29m 20s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/032c36f7f5d7443fba6c6ec33a4c835d : SUCCESS in 19m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/fe6fa1770f694b5f9b6d966c7286cb88 : SUCCESS in 19m 02s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/8b9065a7b525445cb58492f387810ebf : SUCCESS in 17m 31s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e34c305244a9480fb1b7492f5d485e74 : SUCCESS in 15m 40s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/3cf9aa543c184427b2f0c6765014a805 : SUCCESS in 20m 05s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/19c0a04aeb5f481a9f036b79c1780bf8 : SUCCESS in 19m 08s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ee6cbc9ebaf480ead1b06d4475f50b7 : SUCCESS in 12m 15s","accounts_in_message":[],"_revision_number":1},{"id":"0a3790d95a6a4dca9d5fbb38ce8ef6a63567f7bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 11:52:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a3b3154fca8e1c7e13260cc2dab9f7d72d05935d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 11:54:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c4b0457985a46c0b364e4b109a2d207\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faa9652da4444235a84dc2270b0c5e2a : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/252110bfc0cb476ebe49280843a89b04 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7c46f57d9fb3c5da112d2d8ae95c2105285a80c0","revisions":{"7c46f57d9fb3c5da112d2d8ae95c2105285a80c0":{"kind":"REWORK","_number":1,"created":"2022-08-05 21:49:08.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/06/852106/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/852106/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/852106/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/852106/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/852106/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/852106/1"}}},"commit":{"parents":[{"commit":"582f05020b382cab693d4eff58ecfdeeeb104397","subject":"Add missing declaration of kolla_enable_hacluster","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/582f05020b382cab693d4eff58ecfdeeeb104397"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-07-04 16:24:37.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-08-05 21:49:08.000000000","tz":0},"subject":"Update to latest Rocky Linux cloud image","message":"Update to latest Rocky Linux cloud image\n\nThe 8.5 image has moved to the Rocky Linux Vault.\n\nChange-Id: I86c92a1b0b8cf5e4dc534e04709e38bd90f7660e\n(cherry picked from commit 0030c5f1ed5aeac7d991187b1761e0f664c3669a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c46f57d9fb3c5da112d2d8ae95c2105285a80c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c46f57d9fb3c5da112d2d8ae95c2105285a80c0"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
