)]}'
{"id":"openstack%2Fkayobe~854236","triplet_id":"openstack%2Fkayobe~master~I9dce73a8a205c0c0ad02eba3a10e02b82f5b191d","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"29543":{"account":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"last_update":"2023-04-14 07:27:57.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-04-14 07:27:57.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-04-14 07:27:57.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-04-14 07:27:57.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-04-14 07:27:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9dce73a8a205c0c0ad02eba3a10e02b82f5b191d","subject":"Fix passwords.yml generation when parent directory doesn\u0027t exist","status":"MERGED","created":"2022-08-23 15:47:08.000000000","updated":"2023-04-14 07:28:57.000000000","submitted":"2023-04-14 07:27:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854236","meta_rev_id":"9137a2230b9d1e7effa1f87a32192ceebf961589","_number":854236,"virtual_id_number":854236,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-14 07:27:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-04-14 05:40:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},{"value":1,"date":"2023-04-13 16:41:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-04-14 05:40:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-04-13 16:41:27.000000000","_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-23 17:54:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-29 13:36:12.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-09-16 08:31:51.000000000","updated_by":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"reviewer":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"state":"REVIEWER"},{"updated":"2022-10-13 20:04:05.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-04-13 15:34:12.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-04-13 16:41:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-04-14 05:40:34.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"ef6030f6e56e58aed1cbffcb7df8dd1bb44e3c5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-08-23 15:47:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00f294be316d7782ab7680826fa22257bc140398","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 17:54:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4eb98a765054e11b2c6606a7d008ba4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97f6e3bb850b45f79eed33a5ab360af9 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52071a565d774ca59e8f238cc4c08431 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc576f99b044452a815a6a7085ae5684 : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7563c89249604c749dcb5f09f2ef0507 : SUCCESS in 6m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee70597007a74cc8932cfbeb94625c0d : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a8ff8288b334e2cae8d1e572d0fad74 : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18e29bc4fae74f82979cd160afd24237 : SUCCESS in 3m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7e05cb3771be4eed933e88ba753d0599 : SUCCESS in 6m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/639605f898c24ba0be1b4f5abd0c6493 : SUCCESS in 15m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0b22e7925331447582e502d94b62bc71 : SUCCESS in 20m 03s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0a11390e9e5443e2901468c8af48f060 : SUCCESS in 1h 07m 50s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/1e57568775f84ae39c16b71da97476b5 : SUCCESS in 1h 18m 42s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5edfae814dbc4421aed58a16b7a85c71 : SUCCESS in 53m 34s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/765ade5e3029462f9d2ef58186abe3ed : SUCCESS in 1h 20m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/716c31838c9342779daa8ca5def8ecf1 : SUCCESS in 22m 33s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/ad934ac05a824794ba45ef7319343dee : SUCCESS in 24m 29s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6877c68def7a4442a8dac6f2e91a1caf : SUCCESS in 16m 42s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e30c3a5d9b8844ed8036beddd2377a81 : SUCCESS in 2h 00m 25s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d98c5e27bc848539d3e080c056d6e04 : SUCCESS in 1h 26m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/aa09101c1dc34101a67dc3f6d36c2ed6 : SUCCESS in 27m 00s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/034981fe7b4c48b1a3c8d66d2e8388ac : SUCCESS in 23m 59s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a520d56c563447ccb17314f131c57e2c : SUCCESS in 19m 30s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/f672a00de450454b8463874bd0226d70 : FAILURE in 40m 28s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0b70e74a31354ad48cbeeb06d46714e0 : SUCCESS in 46m 03s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6927d4c43dda4f4cba5e1c5049df5d1d : SUCCESS in 28m 13s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5e3e3a57670e4edd92201725b50f25dc : SUCCESS in 16m 50s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/1a606f7353a24d20a73cda8eef4cf530 : SUCCESS in 19m 45s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e511b5194de64c0ea865d32ec33510ff : SUCCESS in 13m 13s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/31256ba385bc4df0862e211e0e44734b : SUCCESS in 22m 03s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/0d2c79998b4248bc8139725deb4d0ceb : SUCCESS in 17m 55s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84a947134a904ce790bb17348185dea4 : SUCCESS in 17m 37s","accounts_in_message":[],"_revision_number":1},{"id":"d3afca877f6a2f79ceeac65122f3bf63e307c539","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-08-29 13:36:12.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c4ac94fbf0a59d7c0a724daa01a178b3e1a5b10b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-09-13 08:57:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e6233e4269e73529ef87c14fc05739d9b789489e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-09-13 08:58:14.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6af86ecf034fc49489cd6fad60a10f2d7c76073a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 11:37:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4725776700e34d7883ddee121880ea20\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47e199c948bb4e07a46c576a40568c07 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64382a43ba214352839a8777a0579cd5 : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ad26bdc941e463ebf417232f50feb34 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1eb2fbf0b9194d8fb22d7f6fa19022c9 : SUCCESS in 7m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c649e09eebf74df784e88b76771473bd : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2961083f5b14064a271835f2981f384 : SUCCESS in 10m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d12ef92dee954de8b1a47a5d60031831 : SUCCESS in 3m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3197c14b86e84c36b32d7c52dd1ae9a9 : SUCCESS in 7m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cfe8993649ad45b58dd04970bd0e905b : SUCCESS in 16m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4275575321564252815c75aca63f730f : SUCCESS in 20m 12s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f367ae9120864d33b7b26c34dc2e1ef8 : FAILURE in 25m 29s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/f71cf9e5b8bc4c04a1b7777b5eff0c4f : FAILURE in 44m 52s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7dc1c1ec4f84263ad9e36d538871581 : SUCCESS in 58m 25s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/713d505d67fe457992e1b3cdb44325fd : FAILURE in 25m 15s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/290010493ac54083b42545f3d6347df8 : SUCCESS in 24m 17s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/1cb8b92c1439464fa4d985f12a2bb7bb : SUCCESS in 21m 52s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99d80429543340ecb63ebe7f1a0449b0 : SUCCESS in 19m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/97aec3a419254514962e538c5adf85ac : FAILURE in 1h 16m 16s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/318b25038a9a49349ee51c498bbecaec : SUCCESS in 2h 33m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8c97731b6c7045dda110c2bc34322ae9 : SUCCESS in 20m 33s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/51e812fb5a164b2fa497bf8bd52b2565 : SUCCESS in 24m 40s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49794ca499d945eca2ecb49e63bcfdec : SUCCESS in 15m 16s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/71532beb0aff42bda5306b7bb1c363b3 : FAILURE in 1h 26m 35s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cc4a4062e80e4ac993bc0599195093f6 : SUCCESS in 37m 01s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40afa85e33a74fd2999584a4f70cad5c : SUCCESS in 27m 42s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/95dcf4f80be64e1592f62cd0bd566c17 : SUCCESS in 21m 10s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/edb951d472994e91bb86db7e81c35b32 : SUCCESS in 19m 38s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33f211a9b32b4b6086048df83513f6e9 : SUCCESS in 14m 13s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/0ece45bfbfe4437e8a5c0414868db667 : SUCCESS in 18m 38s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6e7436fb917f453cb1f1aecf7eff5a2c : SUCCESS in 21m 58s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d7496327aa949f589a53129fd7e3e45 : SUCCESS in 16m 33s","accounts_in_message":[],"_revision_number":2},{"id":"8c2fc0fe0bb232d253c2094f7b1d84c0d09e0b1b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-09-15 08:31:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"628260244e9bfc01564b6ac70e87b2cd35ba48d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 09:43:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b64b0ef93c944a39f14ec996347f44c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/deb98474431a470b84d6e1c70c02c4a8 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/504748d725f749da86d9c7f135a58ab3 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bf56fc771ba499f8ba54f9f52b0aed9 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/053039aafabf42419a9fdfc3613232d0 : SUCCESS in 6m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8bb7a7ce21d4461ea66a55cb69a7ed6a : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/345d33df218f463c9eadfd20276d2195 : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/daee22a052754fdb8d8f628a18b34ad2 : SUCCESS in 3m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/255525e92f5a4d41a477643ef1554cde : SUCCESS in 6m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c5386cfd78c841f089d6232214b9e00c : SUCCESS in 11m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3038175a78eb49f0995eeaa81e199246 : SUCCESS in 15m 19s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00162ac7ea0840e3990f7db842813bee : SUCCESS in 52m 23s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2ce3cd7625040fdbc4ca6228f7b697e : SUCCESS in 19m 19s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad6b84cd515048ff9e63de8d641a0421 : SUCCESS in 1h 09m 21s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a48e7253656d4d47a5e22523bb5747c2 : SUCCESS in 14m 12s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/158de40f3abc4b26a6dd8801afa18a20 : SUCCESS in 28m 26s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/110d40db89e54b5fb97faa04ad1ca769 : SUCCESS in 12m 04s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/176b39c692b243a59500bebd467475f3 : SUCCESS in 12m 31s","accounts_in_message":[],"_revision_number":2},{"id":"06aff02aa512dd3d28ea8f7d68edc1203766c228","author":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"date":"2022-09-16 08:31:51.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"471dd73a833f526a73c51d7c4e7f21886f2dda6c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-09-20 08:47:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9c23b94350ef97b1def562b1f520bb95c2357f15","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-10-13 19:28:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"804dc4059894b32f2d4bbd688721bf89cea9b8b9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-10-13 20:04:05.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4444c0eb4c5882a31e31bec8a463e4b097620eca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-13 16:41:27.000000000","message":"Patch Set 3: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"5f00832226c36d3e972ac5ff20769505a6ee330c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-13 16:41:39.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5ce0e150b0a5d894f21effe2f50d36a5a62ee9ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 18:42:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7de3023e6ad43918f590f8bae6b6370\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1fe913ac0d13455cb2609f31b7e57995 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0344e172d2374aa3a43c4ebf354d48e7 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a0ec710a8ea4692833b6de0b530d4b9 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c64d15d43e364f5a919ed9b1086535fb : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5efb096b23fb4210b492c47a55c7b661 : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80a2235e91cc4b69a5ad8f1a31d8e6f4 : SUCCESS in 3m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/72b16c7cd0dc4100a72f363f774ac40e : SUCCESS in 4m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f8aafdea0a8b4b8087944343c5b62717 : SUCCESS in 8m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/628fada1773e4861b36dfd3ec4051aea : SUCCESS in 17m 01s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/390dcc06a664492d8a5bedce9dba6d62 : SUCCESS in 1h 04m 56s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/4e113cc2189247d6a6f2a6d91e25063e : SUCCESS in 51m 13s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c60ebdbf1734a73bcc4afa8e524a041 : SUCCESS in 54m 17s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e2977f6a701d44b2b1a481b5437947c2 : SUCCESS in 1h 05m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/938f8ce769af415896cfaa4f8ea03e63 : SUCCESS in 16m 29s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/4c0bde3e60044f928b094800b5b435ae : SUCCESS in 17m 11s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c70d8053618241f287c5023fddc23da1 : SUCCESS in 14m 05s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e778440b2749406aa0ff1f917fccc938 : SUCCESS in 1h 53m 36s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6202fcf80354f469cbf16ac3d97f84c : SUCCESS in 1h 03m 59s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/174dd69e117a433d86a1d14941f173fe : SUCCESS in 16m 38s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9227abfbbbe4e93aee70010dd375108 : SUCCESS in 11m 46s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/2cf74ac859af4c3ab7e6466fd0d75f9b : SUCCESS in 33m 25s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2251c0de231f41d39cddb237f1f43879 : SUCCESS in 25m 57s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ef36cc890694400873cb56003dc543e : SUCCESS in 23m 12s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2f8b74ac474b4f2e964ba7a4e787be11 : SUCCESS in 12m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5dab82a0b7454a7caad688f55efee063 : SUCCESS in 11m 34s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6b82e0e0b9de4f8bbc103673eff702b3 : SUCCESS in 16m 28s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c17f33aef2494a23be5c5145fa47ef07 : SUCCESS in 14m 09s","accounts_in_message":[],"_revision_number":3},{"id":"aba31c6c225552810c23e357cb94ce25404d18e1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-14 05:40:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"88e530f9a555e592f719fbde84b3c50c9c84b4ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 05:40:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"85e0950c4b090f7094e158afb659160b3477b801","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 07:27:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00df9c473e3f46c9a3e5c8bb560b2c64\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50ad275879dd480687ce19f723cc78b1 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ee55ac9a9ee4418ae6c056af3e37a37 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b510a62c51a464c9c9d92cd0a24eb2c : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e8ad718961d4fb58c0b31b914762923 : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/819159b50807449db7c9d069d250b1c0 : SUCCESS in 3m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1d033feb97e14bc89b695c198f47c3bc : SUCCESS in 4m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/72952018aa8c45b2953fc2efd9959967 : SUCCESS in 12m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c29ba542a7c94dbdbac4ed58aa752b6e : SUCCESS in 14m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/39074e8618074df393d791229b0ef6ea : SUCCESS in 56m 37s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66bed5a96a35410586b17b564fcc82bd : SUCCESS in 57m 49s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/60923b2186434d94860bf496cf6b518f : SUCCESS in 1h 06m 49s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0103d7ebd05640388153a494940ed5d9 : SUCCESS in 17m 10s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb8461b5973d49e8a7a3c36d541d7324 : SUCCESS in 16m 23s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/38dbfaabb9c9464ea8318cb5dc8d3aee : SUCCESS in 1h 32m 39s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff0bf47fed444b8db8880230aca9b29a : SUCCESS in 1h 06m 37s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6d9edfabc0ae44d781fd4514a0b586e1 : SUCCESS in 16m 52s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e93f12314914e898a35f5459e98a42d : SUCCESS in 16m 26s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5213f1170b2c427fad61c9d54e824dd7 : SUCCESS in 27m 42s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5865e83de0ef459a880e48fb9482aed1 : SUCCESS in 19m 37s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/315787c001b04ce3b245db6d64e3b047 : SUCCESS in 12m 33s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68b54c40c96b4d8aa3553c6cc7727d97 : SUCCESS in 8m 49s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ec215dd101dd4164b9019f5dc5324467 : SUCCESS in 17m 49s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31ba1e4583db43e2a2c3d083f538d098 : SUCCESS in 12m 16s","accounts_in_message":[],"_revision_number":3},{"id":"3a27dc3b26385c2c38956fdedc2fd869419837a4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 07:27:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9137a2230b9d1e7effa1f87a32192ceebf961589","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 07:28:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf95e5a17626490fb2807117c34c9383\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3913383476cd4f8fb08ec762c1ad5d9d : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41a35f226b54441c8520aedbb53839dd : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5287083116da2512b887529e67af4c4919e1e296","revisions":{"9a7cddd272b721bbfd72dea471d06eef3f86f0c6":{"kind":"REWORK","_number":1,"created":"2022-08-23 15:47:08.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/36/854236/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/36/854236/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/854236/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/854236/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/854236/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/36/854236/1"}}},"commit":{"parents":[{"commit":"a2f9801034872bb782c1dd62c7cdf6e019b5371b","subject":"Merge \"Add support for Ubuntu Jammy Jellyfish (22.04) LTS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a2f9801034872bb782c1dd62c7cdf6e019b5371b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-08-23 15:46:29.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-08-23 15:46:44.000000000","tz":60},"subject":"Fix passwords.yml generation when parent directory doesn\u0027t exist","message":"Fix passwords.yml generation when parent directory doesn\u0027t exist\n\nFixes an issue where generation of passwords.yml for Kolla Ansible\ncould fail if the directory containing the file does not exist. This is\ntypical in a multiple environment setup, when creating a new environment.\n\nTrivialFix\n\nChange-Id: I9dce73a8a205c0c0ad02eba3a10e02b82f5b191d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a7cddd272b721bbfd72dea471d06eef3f86f0c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a7cddd272b721bbfd72dea471d06eef3f86f0c6"}]},"branch":"refs/heads/master"},"8289a3f3a926d21d556b3a1eb73c5618d9787a48":{"kind":"REWORK","_number":2,"created":"2022-09-13 08:57:53.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/36/854236/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/36/854236/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/854236/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/854236/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/854236/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/36/854236/2"}}},"commit":{"parents":[{"commit":"a2f9801034872bb782c1dd62c7cdf6e019b5371b","subject":"Merge \"Add support for Ubuntu Jammy Jellyfish (22.04) LTS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a2f9801034872bb782c1dd62c7cdf6e019b5371b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-08-23 15:46:29.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-09-13 08:57:46.000000000","tz":60},"subject":"Fix passwords.yml generation when parent directory doesn\u0027t exist","message":"Fix passwords.yml generation when parent directory doesn\u0027t exist\n\nFixes an issue where generation of passwords.yml for Kolla Ansible\ncould fail if the directory containing the file does not exist. This is\ntypical in a multiple environment setup, when creating a new\nenvironment.\n\nStory: 2010293\nTask: 46275\n\nChange-Id: I9dce73a8a205c0c0ad02eba3a10e02b82f5b191d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8289a3f3a926d21d556b3a1eb73c5618d9787a48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8289a3f3a926d21d556b3a1eb73c5618d9787a48"}]},"branch":"refs/heads/master"},"5287083116da2512b887529e67af4c4919e1e296":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-04-13 16:41:27.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/36/854236/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/36/854236/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/854236/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/854236/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/854236/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/36/854236/3"}}},"commit":{"parents":[{"commit":"a2f9801034872bb782c1dd62c7cdf6e019b5371b","subject":"Merge \"Add support for Ubuntu Jammy Jellyfish (22.04) LTS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a2f9801034872bb782c1dd62c7cdf6e019b5371b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-08-23 15:46:29.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-13 16:41:27.000000000","tz":0},"subject":"Fix passwords.yml generation when parent directory doesn\u0027t exist","message":"Fix passwords.yml generation when parent directory doesn\u0027t exist\n\nFixes an issue where generation of passwords.yml for Kolla Ansible\ncould fail if the directory containing the file does not exist. This is\ntypical in a multiple environment setup, when creating a new\nenvironment.\n\nStory: 2010293\nTask: 46275\nCloses-Bug: #2015093\nChange-Id: I9dce73a8a205c0c0ad02eba3a10e02b82f5b191d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5287083116da2512b887529e67af4c4919e1e296"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5287083116da2512b887529e67af4c4919e1e296"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
