)]}'
{"id":"openstack%2Fkayobe~854800","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I3a95382cf6e761c5c0e5c7a47d6477b3cdfa8c19","project":"openstack/kayobe","branch":"stable/xena","topic":"seed-containers-names-stable/xena","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-08-30 19:56:24.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I3a95382cf6e761c5c0e5c7a47d6477b3cdfa8c19","subject":"Remove unused parameter from seed container example","status":"MERGED","created":"2022-08-26 16:22:55.000000000","updated":"2022-08-31 00:01:56.000000000","submitted":"2022-08-31 00:00:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854800-seed-containers-names-stable/xena","cherry_pick_of_change":850337,"cherry_pick_of_patch_set":1,"meta_rev_id":"f08d9f012b6f635ef7855718c6f394fcef0019ca","_number":854800,"virtual_id_number":854800,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-31 00:00:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-08-30 14:42:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-08-30 14:42:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-26 18:20:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-30 14:42:29.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"2f103194d3d53c03db838549f4d762409370ead3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-08-26 16:22:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7677ea708f54dd2c50506df1965db1df8301cea2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 18:20:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f0dcca2be4247269693520d1797bbd9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83947e74a1b7405d9e7ee8012e42b96f : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c576424833f463e9eb7e5e5d02a6bb5 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f35e6b0b49b4dac98ef4dd5740abfb4 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9439e70b49c946188f47a5cde76109c9 : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66409476476a44d087f97e7e4a73a73e : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/486e8a7dfb8e4240908151932fd9a53d : SUCCESS in 8m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/95db3292383241d79b7ef6c2047802d7 : SUCCESS in 6m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/00598246e21049beabc91b655c17ad9f : SUCCESS in 12m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fc27fb20c4fb415f9e262ba2b1bb1e98 : SUCCESS in 20m 51s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/359a1011330f4bb5a88ab0fcc1c3ee62 : SUCCESS in 53m 49s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/14cd1317cc9a4205ba95e628fda3ea03 : SUCCESS in 1h 10m 59s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b54ed83527c44daf874ad3cc379d0c4a : SUCCESS in 46m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/2477119d9cee48ff8659dba97619b16b : SUCCESS in 1h 09m 33s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4e336e4177d945e6b9e267e271706d2a : SUCCESS in 20m 01s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/e8d06d6751af40c595e3fa021c64dbb6 : SUCCESS in 24m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1306f97aa750496f99827ed47364b582 : SUCCESS in 13m 43s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5e30662d514449b2a36f23ff799bbcf7 : SUCCESS in 1h 56m 19s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebdc3b2ed8bb405e8f957034f1117c2c : SUCCESS in 1h 01m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4818fae980a94f679763987294285bbd : SUCCESS in 21m 20s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/957f5ea19d294eb997b698cf9a15bb66 : SUCCESS in 23m 50s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15e6bcb4b7fb4c60af1d26895c76a500 : SUCCESS in 13m 42s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/ebd44043d3164ade90401922c85556ae : SUCCESS in 39m 27s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ddb3bdb98ebb4af68af14bcd7da3350f : SUCCESS in 30m 34s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/839d6a63a484414ebf3485245bebb6cd : SUCCESS in 18m 17s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2acbd7fef8664ec4847e09533a2b3b78 : SUCCESS in 18m 09s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/893bae8fceeb41fa8f71e63e89898d6f : SUCCESS in 17m 43s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07bef732b8224adab34c4f6c0f21d045 : SUCCESS in 11m 14s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/923fde8b22274f5b9a600085545ee855 : SUCCESS in 16m 03s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a98c9bcf188643b4b0f32ef7ab5c0165 : SUCCESS in 22m 23s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbf812c8fc384ddd86f4658c0472f207 : SUCCESS in 12m 24s","accounts_in_message":[],"_revision_number":1},{"id":"939fdb5cc77dc2d947baeace5cf55e1b2e85428f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-08-30 14:42:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4b009d320885330ab7508ac73742388c548e7f48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 14:43:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"58be1452f62d6eb316989633b068005eb9f26b97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 19:32:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b8c9eaa6ebe4da0972abbcd44d58302\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da87842249954cca9d0efa82e1e0db1e : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d5cd19c1058412bb70e6d1fa7e43195 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c8718033233449f9f502bf3b1ff7927 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f14df781d75b4b20b0762b8221a9ac02 : SUCCESS in 10m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/99be74f8c7a94e8c987ade716a77e8c3 : SUCCESS in 9m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bdf1bac9f7354912a14e5c449fe0bc85 : SUCCESS in 15m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/78a78b6fa8bf49e9b39621e4802b7947 : SUCCESS in 23m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a1bfe91228954d3e98c73c5f4cfc7548 : FAILURE in 1h 08m 18s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/0b7219807bdb43779a84a292dae242ee : SUCCESS in 1h 15m 16s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1798f73070b34a7e89e42ccaf56ace89 : SUCCESS in 1h 00m 54s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/640213a8b77b46d494ce52bcf277aaee : SUCCESS in 1h 14m 29s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3e68fcf135824ee8b73493bce93a5dc9 : SUCCESS in 28m 02s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/813dcfc6d5db4e549acc4764f2c03ea9 : SUCCESS in 27m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af4a673f01b945fa8e228b3b6756dfcb : SUCCESS in 18m 01s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/30962a1aade24ba390eb401a0a532176 : SUCCESS in 2h 33m 37s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ae73f9ee81942a0ae8801f8462dbd6e : SUCCESS in 1h 39m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/947b89c73ad04862a9f42da803c94e55 : SUCCESS in 24m 12s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/c8d10dea877a47e1abd84405047787ed : SUCCESS in 20m 39s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10e7c2c8e011483badd822380405e799 : SUCCESS in 18m 53s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/91ce5444d4f546d38f3cc022993a2b46 : SUCCESS in 30m 26s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96148df7a45c4922b53ff927b546cf57 : SUCCESS in 25m 53s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2f2f6722ca8c41059d3ebafa0279572a : SUCCESS in 23m 24s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6d0594e2c1e447d89ada165f3fade949 : SUCCESS in 22m 01s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e41074e7b984e92a25d3559ba2f445d : SUCCESS in 13m 28s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/29a2c541116348d0ae5ff29f0ad93f6d : SUCCESS in 22m 54s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c0bf11904d84406e96cb6f174a8a148b : SUCCESS in 17m 16s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cd8b2afaa644adba5d618aa4850767c : SUCCESS in 27m 59s","accounts_in_message":[],"_revision_number":1},{"id":"32b5c96c24b3aaa1bdb6e9adb886a0ef57046907","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-08-30 19:56:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"be6c6090269cce1e8d2852a5ab833b20c94a37cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 21:50:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6b4ef73c10948ba96003bd0e523aafd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc44abeb695147979681e9b86872ef32 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f916ad02ba64fe58d5061a5679d6208 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a87d716e453643f39612fc2cc23127c6 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6e030e4f26044b3ba8fa2b29ef8746d : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fad6ff7f11414256b18bea23580dae44 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/382331399de24d93abd77e2470cec48c : SUCCESS in 9m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bc7dbbd9807348508bff97149aad5a97 : SUCCESS in 7m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b0b295a2cfef44c8948febe7054d55e4 : SUCCESS in 12m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f107e5bbda854df7adc83044ad4ae170 : SUCCESS in 20m 36s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6a608c0d3d4b4cecabbbf3c1cbd19ba9 : SUCCESS in 1h 15m 51s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/e3245cab2920454cb7d042878c0b8674 : SUCCESS in 1h 31m 00s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbe82c1b3331464999c34edae826a780 : SUCCESS in 1h 08m 02s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/cd6049199a9e40a48532630be6a3c964 : SUCCESS in 1h 18m 15s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/264d0d8bda1447e890e68fa3d0030939 : SUCCESS in 29m 37s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/f6a38865a21b4e9b956babb60f37fb97 : SUCCESS in 24m 15s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7023036f68094c2591ba2dea6a27b1e8 : SUCCESS in 14m 50s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d1e906dfe05e4818978ea4036fece183 : SUCCESS in 1h 45m 54s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/477be22d0443479682b5505ad108e39c : SUCCESS in 1h 08m 55s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3af011246aba48dfb8a2ba4ae6a00011 : SUCCESS in 24m 43s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/19a2f658c6ff444d8567fa77a755202c : SUCCESS in 23m 49s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a942baca5a6e4c719ca393e7b574adb8 : SUCCESS in 19m 47s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/c60c2d3c2312454ba7e8c0a17a222442 : SUCCESS in 48m 10s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b6ad04a79c1245d88b9ffee892c4b608 : SUCCESS in 33m 32s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05151199bc2a45cdb3aef7def1e8b9cd : SUCCESS in 22m 11s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/093a436068f3418c848fd7cd115b10cc : SUCCESS in 16m 44s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/0f38e3a5896c497aa8118b2b7eda56ba : SUCCESS in 15m 14s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/437055d1b61043ddacfd4197c6cd694e : SUCCESS in 11m 43s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/2a1508ac0a13439bbfa216d3160b6d5f : SUCCESS in 17m 50s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/8dcc123b7ed84ee98235947897e26451 : SUCCESS in 19m 37s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ec2b2a5ec7d42bfa59f6fb00485f4dc : SUCCESS in 17m 05s","accounts_in_message":[],"_revision_number":1},{"id":"60024c4f2f88ed5c296a27caeea6c388ef58309b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 21:51:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1710830b10b8c920607f507f3ba774510170598b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 00:00:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/676660f562c14fa592197dd63f5e8c70\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bf65d7d0f2149fba54e56200fe63183 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad544e539d124ed7a2907b59fb987bbc : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63d12a7383a84f09861f482c52d1b443 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d243fe0d3a7488ea75e18c3b38b04c4 : SUCCESS in 9m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a52f4e59f17d498cb1c5bcc3bc536f22 : SUCCESS in 6m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/50f0731a5e2f43f3bce013227734d9da : SUCCESS in 15m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9ae5d91f216b454ca52c4583650ace48 : SUCCESS in 35m 04s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a036a1634e1f4e65817f3713cdaf17af : SUCCESS in 1h 18m 58s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/44a00701f2d14ebf8ed6b229e949fb90 : SUCCESS in 1h 15m 19s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c49cceaaee244f7810ba371542f51e3 : SUCCESS in 57m 48s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/95aa0b05ff3547ec90409ebd0e853a74 : SUCCESS in 1h 12m 23s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e97006d6ed5540f6af17467ab54a4391 : SUCCESS in 27m 07s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/1fae3b7c39df43f79445bea9e86bf148 : SUCCESS in 23m 40s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e4551b0cc1d4c37afa6cf26c88fb66c : SUCCESS in 16m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/91cb3fb4ba6346948e949b03f4b50a90 : SUCCESS in 2h 04m 22s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34cd2efa57314e7b8ad4eedd4569872d : SUCCESS in 1h 14m 45s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5d5e57a3607e475e812a08ac9a6f8da7 : SUCCESS in 22m 58s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/7add7bca5fef48c8a1906bdf5a01f9e6 : SUCCESS in 20m 36s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b837255053f247b78f7691d409c91a6a : SUCCESS in 16m 50s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6095197a783b4bb2bfd0f89daf7d2676 : SUCCESS in 28m 48s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bef5b47dc3f144c5b495412edf2d8f61 : SUCCESS in 25m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/19a1ee8144ef4903b4605d50f1ab0831 : SUCCESS in 18m 11s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/00a2ee08a5574311872099ef5fdab6f8 : SUCCESS in 15m 28s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4587d52df5d64391b8619556bb2ca22e : SUCCESS in 13m 06s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e216ffccf26d47469f5c3239768082f4 : SUCCESS in 20m 25s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c831011f9e3748e4a7923cdf42fa65de : SUCCESS in 22m 04s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5312e90ecedd4da9b9d881cc26fceb81 : SUCCESS in 14m 17s","accounts_in_message":[],"_revision_number":1},{"id":"f60cc3a952ad539ce1d054160e9bc795655a38dd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 00:00:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f08d9f012b6f635ef7855718c6f394fcef0019ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 00:01:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/670d5cf4e350458db0a7a85de0366438\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1737708bd1b94ebfa484ca267cfcd664 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c709806b24fe940cd12cbc693004122e0ff617cf","revisions":{"c709806b24fe940cd12cbc693004122e0ff617cf":{"kind":"REWORK","_number":1,"created":"2022-08-26 16:22:55.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/00/854800/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/854800/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/854800/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/854800/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/854800/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/854800/1"}}},"commit":{"parents":[{"commit":"85efbbe7792b143796653a1d3d9b844de530d752","subject":"Merge \"Update to latest Rocky Linux cloud image\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/85efbbe7792b143796653a1d3d9b844de530d752"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-07-19 12:02:56.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-08-26 16:22:55.000000000","tz":0},"subject":"Remove unused parameter from seed container example","message":"Remove unused parameter from seed container example\n\nThe keys in the seed_containers dictionary are used as container names.\n\nChange-Id: I3a95382cf6e761c5c0e5c7a47d6477b3cdfa8c19\n(cherry picked from commit a9d00a2073cf2f29ecb7abff6ca1cc6a59947348)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c709806b24fe940cd12cbc693004122e0ff617cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c709806b24fe940cd12cbc693004122e0ff617cf"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
