)]}'
{"id":"openstack%2Fkayobe~854801","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I3a95382cf6e761c5c0e5c7a47d6477b3cdfa8c19","project":"openstack/kayobe","branch":"stable/wallaby","topic":"seed-containers-names-stable/wallaby","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-08-30 20:01:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3a95382cf6e761c5c0e5c7a47d6477b3cdfa8c19","subject":"Remove unused parameter from seed container example","status":"MERGED","created":"2022-08-26 16:23:03.000000000","updated":"2022-08-30 20:02:19.000000000","submitted":"2022-08-30 20:01:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854801-seed-containers-names-stable/wallaby","cherry_pick_of_change":850337,"cherry_pick_of_patch_set":1,"meta_rev_id":"7e44054f0f0a4078ddee52a2e0d74ecbe4a0f212","_number":854801,"virtual_id_number":854801,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-30 20:01:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-08-30 14:42:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-08-30 14:42:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-26 18:27:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-30 14:42:41.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"ee3a8bec070482ea657698dd6bbfb90cb8b63cc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-08-26 16:23:03.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"711adfc7291bedca05aa4b1d8d76f30821af7b68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 18:27:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19cdcbd7e89c40f7ae66a92584e3a8a4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21229373f7364df994f007ebdda682b0 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76d822c33e404280844c73170e3f7b67 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06198d5f1f594a2f8a4c29f7634c588f : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe72f1794029425ebd6df0a7f1356649 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23b98f8827d649bca089b72f943bbfe0 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b0846cfceab48078a208ef3e2a08fe6 : SUCCESS in 7m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0e609030e0994e9d9178ae059d22b0b0 : SUCCESS in 6m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b931b8710f394ebbaa4f9b3b02dd3daa : SUCCESS in 10m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e547d746fd73454da06691ceb236757c : SUCCESS in 20m 50s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/96a1d399c8f84fbda2460ce7d5e867d4 : SUCCESS in 1h 57m 14s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9b75cfea8574a71b088344cc5ffd950 : SUCCESS in 56m 41s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7e6fd2dd4fdc41dda78d9aae17edb2c3 : SUCCESS in 1h 13m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0745e5588a55484cadb3b9f7c2a8a1de : SUCCESS in 18m 59s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c09f1330d184f63a066e3127a7f87c7 : SUCCESS in 17m 43s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3d01f0a973564aa0a5c0d8374005a540 : SUCCESS in 1h 41m 24s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7112f3de7c424e92a6c68de17e991c61 : SUCCESS in 21m 43s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/822ef7c0afa34c6c9416b56825af7e84 : SUCCESS in 14m 52s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/a1191a1f506c433a938a18cb9e6a03a9 : SUCCESS in 54m 41s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ba6dcc5dd3ae4e7985942ab46b056c23 : SUCCESS in 24m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/01acee4a99c547219a2fa5e176225d6c : SUCCESS in 18m 07s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da7f2cafc94746a28a808e0fd8482222 : SUCCESS in 10m 51s","accounts_in_message":[],"_revision_number":1},{"id":"59d96bbbe7fd03b0c4aa1df19d9b5b08247f7c5a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-08-30 14:42:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"57fee8c02eebcdc6838c5c915aec2ed626e0f93f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 14:43:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0d532db076bca8cc56e4be0f6a5f4b500853e599","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 20:01:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25ed713d465a4c6d840b1862baa1b894\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27c84d9ae501481f8e257fba50bc196b : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e07ccea05c5a4481af4147718bd5a8fb : SUCCESS in 8m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c1300fee0be44edb9bc57a447477f86 : SUCCESS in 8m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7d0a0e475bc452c9faef30a6c82bde6 : SUCCESS in 10m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/334756084e6247caa7921aaebdb93aa5 : SUCCESS in 7m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/21e23359b03f428eb268f0f027b85075 : SUCCESS in 12m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/afd820738a854d1d8d30c9f454b6660c : SUCCESS in 21m 48s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2e8a69f79ddc4725908f18c32b3482fd : SUCCESS in 1h 01m 48s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89297e67f18d4383adb99a86ab0a36cf : SUCCESS in 47m 17s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f5c5e51b34d74beea6f2ecdae05b8325 : SUCCESS in 59m 24s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/92963d2b9ce64ecaa737fad188261ecc : SUCCESS in 17m 21s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1b807816a5f442cb2df19f0a4f06760 : SUCCESS in 13m 15s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/266193b8ad8f4cf8803838ada57645c8 : SUCCESS in 1h 51m 41s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3fab642ac77c4cc2aef5d589a255ac07 : SUCCESS in 23m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67f40e11360b48b483643aac7ecd099a : SUCCESS in 13m 40s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9b37e21441404e14b2f3a829faf98879 : SUCCESS in 22m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a543be74e4ba423db3986fd42f54a9ce : SUCCESS in 15m 34s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c734b3918c8a479d8fd33139f1ef69d4 : SUCCESS in 11m 14s","accounts_in_message":[],"_revision_number":1},{"id":"bc98c39f3808d95de909c509deaeb1b730c111f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 20:01:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7e44054f0f0a4078ddee52a2e0d74ecbe4a0f212","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 20:02:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe773b662a9246e0a4403aa16d25aab7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cd0c11659a443a7b4d4930cfeef4866 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9a7fdc429763bd95e7b3029f9dc4d77e63b3bede","revisions":{"9a7fdc429763bd95e7b3029f9dc4d77e63b3bede":{"kind":"REWORK","_number":1,"created":"2022-08-26 16:23:03.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/854801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/854801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/854801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/854801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/854801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/854801/1"}}},"commit":{"parents":[{"commit":"dcdbceac8d5a1f9aa06bc44d68a6804b9f49790b","subject":"ovn: add network group to neutron-ovn-metadata-agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dcdbceac8d5a1f9aa06bc44d68a6804b9f49790b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-07-19 12:02:56.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-08-26 16:23:03.000000000","tz":0},"subject":"Remove unused parameter from seed container example","message":"Remove unused parameter from seed container example\n\nThe keys in the seed_containers dictionary are used as container names.\n\nChange-Id: I3a95382cf6e761c5c0e5c7a47d6477b3cdfa8c19\n(cherry picked from commit a9d00a2073cf2f29ecb7abff6ca1cc6a59947348)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a7fdc429763bd95e7b3029f9dc4d77e63b3bede"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a7fdc429763bd95e7b3029f9dc4d77e63b3bede"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
