)]}'
{"id":"openstack%2Fkayobe~855318","triplet_id":"openstack%2Fkayobe~master~I40212b40a4d391272f4063f42e8a7043b3d58fd7","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-09-08 13:29:12.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-09-08 09:35:35.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I40212b40a4d391272f4063f42e8a7043b3d58fd7","subject":"nclu-switch: Fix order in template","status":"MERGED","created":"2022-08-31 08:10:58.000000000","updated":"2022-09-08 13:30:37.000000000","submitted":"2022-09-08 13:29:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855318","meta_rev_id":"7e10b3b6683cdf6027bd049a60e7ca105fbf22a0","_number":855318,"virtual_id_number":855318,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-08 13:29:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-09-08 09:34:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-08 09:34:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-08 09:35:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-31 10:16:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-31 10:59:04.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-08-31 11:05:19.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-09-08 09:34:17.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"779a4d813ec94154ca9e2cecf35be91b1e34b28b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-08-31 08:10:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b79e1c6aadcb6aba38fa65e3e30f022653d26d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 10:16:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d80231e2e9b140569982b9b50b96b52f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14ab35c5f366442885ed9bfc4c1ee8e2 : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cfede507def4bd09798ba86244f2014 : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c76181f4927747febccdd367aba5b73c : SUCCESS in 9m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c284ee8757b453f95bb335c5e415d16 : SUCCESS in 8m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/386e61cb06cb47a39ac0907623b561c6 : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5d7b86e37ce4baabd1e326509f409b3 : SUCCESS in 11m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b6793ae89dc343db8c47c7621590761c : SUCCESS in 8m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/22a28f6c5285477c99c1cc4cd886ed89 : SUCCESS in 14m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1915941bb3bc44349e4d57890a26e03e : SUCCESS in 22m 59s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c8182a7bc1484e7e84c3150269efb24f : SUCCESS in 1h 08m 39s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/04506b6a37f840cc9aba4d0e5bbcd093 : SUCCESS in 1h 11m 01s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d30cd0d659a149189c76c99f2a5076e8 : SUCCESS in 56m 49s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ed84081b53474e85a3739b76fc9ad1db : SUCCESS in 1h 13m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c914ff65db6c440cae6c368847f69e0b : SUCCESS in 27m 23s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/aaf0262f6b36492cbaa6c9e64ac71401 : SUCCESS in 22m 03s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc95fb20026244c0b3d497a9a342925f : SUCCESS in 19m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4c9a62452e824438863226bb3bffeffe : SUCCESS in 1h 53m 58s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2678cf41a8a14299b6e95c8f54fc4805 : SUCCESS in 1h 18m 29s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2971ea2dde1d43feaa8402aa6be5f4f6 : SUCCESS in 23m 51s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/a8aedfaa3e45441f93ac5f79ef1a0ffd : SUCCESS in 20m 39s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3849a04435f44405930e73d4675929b1 : SUCCESS in 18m 04s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/e0158b4532ce40ea97b702806533e007 : FAILURE in 30m 51s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bdb646de107243419f9ad565a63b26dc : SUCCESS in 40m 28s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9641eaa8ab734f33ae26e74b0488d2cb : SUCCESS in 29m 57s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ccf0cc60308b4f6abf0551155595f4c6 : SUCCESS in 17m 16s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/89e76f5bbd8b46ae9ae009df09f6dd7e : SUCCESS in 15m 30s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b92ab904e75044b0a0fefea9bb2fbc5f : SUCCESS in 12m 33s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/ab0c6c62e95944139191393fe711b7e9 : SUCCESS in 19m 37s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c7a75f3b243145d9b835c26d129e6a3e : SUCCESS in 19m 28s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fcba1de62b8441bda1dbe75adf206fe0 : SUCCESS in 14m 43s","accounts_in_message":[],"_revision_number":1},{"id":"47061827e3dee8accb431c0be9ff6da406e15af9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-08-31 10:59:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"88fd75887608a35c1dd5673194990927270be62f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-08-31 11:05:19.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b1d240034a97c8d20712b2aa8703ee1e8d349713","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-08 05:14:34.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e1c34db409952c1aff28cebdc5d409309a33c0f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-08 05:17:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"98f61fcbc47b2dbf2b9e8ee8ad68fa38e8f8bd68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 07:52:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b63483285a9a4371b55dc28089257719\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3717666657724f0fa15f1d3ef81bc144 : SUCCESS in 10m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a632d112ee1438c8a3828617af42036 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9be36d27dc0d4067a4bd2f94b4878fd5 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c1e593db83147e38254a6e896fcacd8 : SUCCESS in 6m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14774230714a4cd5bf12f45d125c3263 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/341b2b5a5616419aa067923cbd75d5cc : SUCCESS in 12m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00cc05d042194f9294ac2d6e06299bef : SUCCESS in 3m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3022f6bae2fa4e24b05fefdae448c4b5 : SUCCESS in 6m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1b4eaf51909349b6bc8d86946e233010 : SUCCESS in 17m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c5c074fd59964b4a9e047dad1e4364fd : SUCCESS in 17m 35s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/108fe57fd86b4dffa38594959f6bc978 : SUCCESS in 1h 18m 29s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/6a009755beba49db9888bba1b68bf4d5 : SUCCESS in 1h 07m 53s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/abd3b6f029e043a99322a2c5b33dd651 : SUCCESS in 54m 17s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c12ede43230845e4b0e910f27618a952 : SUCCESS in 1h 13m 11s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/820475b9426a4bbbae19797ad762eb5c : SUCCESS in 27m 10s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/d39e42e4602c4318b933a0ca0441bb9c : SUCCESS in 22m 42s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/540048a43ba74945ac5dae1042ce620e : SUCCESS in 22m 26s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4221773b69794098972c0eac801c833f : SUCCESS in 2h 27m 22s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0097319de02421692da436961cc18b4 : SUCCESS in 1h 14m 08s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3cdc27c2031f46ae86022a450dfe6d9d : SUCCESS in 18m 50s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/2444251f83c447619863f92fc899bc57 : SUCCESS in 23m 28s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fc37873fcdf4583acdaa29ba515b0d5 : SUCCESS in 16m 40s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/7994ca9b4a1e4cf5892823b62de4fd7a : FAILURE in 34m 28s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b52ffd44f2c1485db9199b17ffa75b5f : SUCCESS in 43m 21s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d0597d9a4754490a132b76d837a27f2 : SUCCESS in 30m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b3c2ee65dc004bfd8da98f3e994cbef1 : SUCCESS in 16m 29s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/31a0e54b2429439cbb6a9f2d475adc57 : SUCCESS in 15m 44s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39622f9fefa748b79f2de3367fef926f : SUCCESS in 12m 11s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/6a3f5458b2a44da88d9ab9db3ac0086a : SUCCESS in 17m 33s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/303d704d5b62449d8c934e04a6b98c73 : SUCCESS in 18m 10s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7f90345442d4040bafdd9eef80969e3 : SUCCESS in 13m 04s","accounts_in_message":[],"_revision_number":3},{"id":"69f48656066e71b385905c240722bd056355a440","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-09-08 09:33:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"39f193ffc7e0d88fc647cd74b5689b6e8240a267","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-09-08 09:34:13.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ffc431237613412bc540b3dc2fd3bedcd3c12165","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-09-08 09:34:17.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"864e7168b8a201160a0be798b67fab1e8e72726d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-09-08 09:35:35.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"fe28cb61ce7885ffd20a7f66e9b4b4bd610b1948","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 11:31:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e376ebc85e4644a2a5b9344811c1802b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d397a0d7b354c8b983b92a24e553b72 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69cf6ef2ec664168b74c3ed42823b20b : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6b601636bfa4cb4b2b9a16ad6cce299 : SUCCESS in 9m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11afa2a71b8a44c0829ee48a805b439d : SUCCESS in 8m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/faa8bf6a16a74d94a21df900944adc9a : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91f9bf88106b4548b28b9a7791675d2c : SUCCESS in 11m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9192345abe6e492ca95dfca7fbf184f8 : SUCCESS in 4m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b967a99da40f4af39ae238f188f979b8 : SUCCESS in 6m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8cf9eda439d048ae9167d8c7f7c27345 : SUCCESS in 15m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/916d38ee468d44938e32a6acc8238833 : SUCCESS in 18m 45s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4e6d012355e349f1bc0f41f974e97cce : SUCCESS in 59m 58s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/ee027c491d4543208b460fb83c0c0e89 : SUCCESS in 1h 10m 53s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4b1333d51874b6ababa4434dc62056e : SUCCESS in 54m 28s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ea96a788f6ef43f7aa04632bef74038e : SUCCESS in 1h 00m 35s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/33840ace51ec4e0bb1d970657083690f : SUCCESS in 21m 09s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/c9d147d875be4a9591b47722bf698ed8 : SUCCESS in 26m 24s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3af2f9e85c1d462785b263c806b4f013 : SUCCESS in 15m 42s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ecab064d04114bed8f58535560cc2885 : SUCCESS in 1h 46m 24s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc6dbb79cc2349e6a4999c3b52ca8565 : SUCCESS in 1h 19m 43s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3bb9df8ac7464eb3bffda3f5ae33d1f4 : SUCCESS in 21m 33s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/0153ffb561b84e6399e016128d45212a : SUCCESS in 22m 07s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/911bafea6e9d463c81a6d9f096e88a88 : SUCCESS in 20m 15s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/cc4a582c6560405fb269edd78124322b : FAILURE in 32m 47s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3b0cdfc51755407d80229e55fcef8335 : SUCCESS in 38m 48s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/273849b930084a5ebc1e2c912e68bc19 : SUCCESS in 27m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/42d39d0b9ba441768a88162d7a3fcfe5 : SUCCESS in 16m 21s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c456754fe5b046468ea8b6b0d423dc92 : SUCCESS in 16m 33s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c53fa054811a47dd9540773927a8357c : SUCCESS in 14m 23s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/9c16e0040d5f4e8e8c2a5b1b81f2b56b : SUCCESS in 20m 12s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/472ee71594fb44f184a5f7e080e5e27a : SUCCESS in 23m 26s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc5596ca196c4823b735c1aa7da988d8 : SUCCESS in 16m 45s","accounts_in_message":[],"_revision_number":4},{"id":"497398c328d55a572a1edec539df7417a7a0bbc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 11:32:15.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"508bfac50aa591fc26229d1d3734b58a27d34c0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 13:29:11.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c87dd352a9b6466b890662bf7ea10041\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ac77445ae9c47cbab0528bc02a1cc1a : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfa1023a63e24ff6a0f46312187ec965 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3ed0f979c3641d9b2966682a7b0f753 : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65262d86cc694a85b9d591fe575a497f : SUCCESS in 9m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64f7e0fbc3284efbba08f0d32ac89c8e : SUCCESS in 4m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/675125f064c2462b8db2928040fc5227 : SUCCESS in 6m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3923ca89395c448496a6107bd9f17269 : SUCCESS in 14m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6f82860916104f2295c26a4bb0315598 : SUCCESS in 17m 52s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/415b7aa08457446b847c966f0a44f6e9 : SUCCESS in 1h 01m 34s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/188acd49b459459aa789395f9143790e : SUCCESS in 1h 10m 20s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/985b09bb08204a10915359507fb0ed8c : SUCCESS in 53m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/04848be60b04436a9e1296914930a2ff : SUCCESS in 1h 11m 43s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8b22f40bb53c47e29fe53b6e37c6d68e : SUCCESS in 23m 39s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/c79a239b0a8948bf82ac838b5c7a0e86 : SUCCESS in 26m 59s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0fa0b9fb064749da85ad116bdadcd076 : SUCCESS in 16m 27s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bc3b47d962824e87b6078756dddf4f79 : SUCCESS in 1h 56m 01s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eaa6eae2f112476ba7c2b9dc85bab27b : SUCCESS in 1h 06m 06s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d7a265bc18c04cb58c749d1e5c6cfb45 : SUCCESS in 24m 19s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/7c204fd88a0341b2b3a2a90999d1bbfe : SUCCESS in 24m 06s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e74ddced9d794df8b52bd33d437ee9dc : SUCCESS in 15m 11s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1c98c99cef794a2aa23ae618177beb89 : SUCCESS in 39m 30s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd0ace099b084ec29b7bcb804e1ec8f9 : SUCCESS in 34m 11s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/342ade067a7d4cbd8c72ab73945e0da0 : SUCCESS in 18m 53s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/fe769a024a7c4571b23dc11ba51d6a6b : SUCCESS in 18m 45s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6bfa82a13d714adaa162a9aa16c5b4d4 : SUCCESS in 14m 16s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/b7325f5c998449b0b058d22a64cab46e : SUCCESS in 20m 15s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/fcfa7f3d5d3d4f719f02b06eaf51b03e : SUCCESS in 21m 53s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e93a8895d39d4aa5819d6a8a330f64ae : SUCCESS in 12m 43s","accounts_in_message":[],"_revision_number":4},{"id":"2ad7179f16f453b7e69bee047542332e464903ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 13:29:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"7e10b3b6683cdf6027bd049a60e7ca105fbf22a0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 13:30:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c10d4ea16fcc47e5bf1115f0306ee618\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20ed02c6462b4f66b27e56bdae82adf1 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c959a06af9aa4d609bf0d92fd83a8930 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cfb5c004eafd7df7e0c5e94b7697811e174301ff","revisions":{"33ad4a90ec54a9b8b10ee74b9c4279562e1e1301":{"kind":"REWORK","_number":1,"created":"2022-08-31 08:10:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/855318/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/855318/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/855318/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/855318/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/855318/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/855318/1"}}},"commit":{"parents":[{"commit":"6e2f5a1884745dbb596ce6f127e7eb875ab63cdc","subject":"Merge \"Fix IP allocation limit in overcloud inventory discover\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e2f5a1884745dbb596ce6f127e7eb875ab63cdc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-08-31 08:10:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-08-31 08:10:55.000000000","tz":120},"subject":"nclu-switch: Fix order in template","message":"nclu-switch: Fix order in template\n\nCurrently description is applied first to the interface, but if it\u0027s a virtual\none - it won\u0027t exist before running the configuration commands.\n\nChange-Id: I40212b40a4d391272f4063f42e8a7043b3d58fd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/33ad4a90ec54a9b8b10ee74b9c4279562e1e1301"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/33ad4a90ec54a9b8b10ee74b9c4279562e1e1301"}]},"branch":"refs/heads/master"},"4a24d00fb4eac52713e755542efe9efc956f2082":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-09-08 05:14:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/855318/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/855318/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/855318/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/855318/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/855318/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/855318/2"}}},"commit":{"parents":[{"commit":"6e2f5a1884745dbb596ce6f127e7eb875ab63cdc","subject":"Merge \"Fix IP allocation limit in overcloud inventory discover\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e2f5a1884745dbb596ce6f127e7eb875ab63cdc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-08-31 08:10:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-08 05:14:33.000000000","tz":0},"subject":"nclu-switch: Fix order in template","message":"nclu-switch: Fix order in template\n\nCurrently description is applied first to the interface, but if\nit\u0027s a virtual one - it won\u0027t exist before running the\nconfiguration commands.\n\nStory: 2010279\nTask: 46232\n\nChange-Id: I40212b40a4d391272f4063f42e8a7043b3d58fd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a24d00fb4eac52713e755542efe9efc956f2082"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a24d00fb4eac52713e755542efe9efc956f2082"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2946c5a7d5b131a25b5a0ae0d61e187102a61bc2":{"kind":"REWORK","_number":3,"created":"2022-09-08 05:17:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/855318/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/855318/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/855318/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/855318/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/855318/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/855318/3"}}},"commit":{"parents":[{"commit":"6e2f5a1884745dbb596ce6f127e7eb875ab63cdc","subject":"Merge \"Fix IP allocation limit in overcloud inventory discover\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e2f5a1884745dbb596ce6f127e7eb875ab63cdc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-08-31 08:10:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-08 05:17:41.000000000","tz":120},"subject":"nclu-switch: Fix order in template","message":"nclu-switch: Fix order in template\n\nCurrently description is applied first to the interface, but if\nit\u0027s a virtual one - it won\u0027t exist before running the\nconfiguration commands.\n\nStory: 2010279\nTask: 46232\n\nChange-Id: I40212b40a4d391272f4063f42e8a7043b3d58fd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2946c5a7d5b131a25b5a0ae0d61e187102a61bc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2946c5a7d5b131a25b5a0ae0d61e187102a61bc2"}]},"branch":"refs/heads/master"},"cfb5c004eafd7df7e0c5e94b7697811e174301ff":{"kind":"REWORK","_number":4,"created":"2022-09-08 09:33:43.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/18/855318/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/855318/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/855318/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/855318/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/855318/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/855318/4"}}},"commit":{"parents":[{"commit":"6e2f5a1884745dbb596ce6f127e7eb875ab63cdc","subject":"Merge \"Fix IP allocation limit in overcloud inventory discover\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e2f5a1884745dbb596ce6f127e7eb875ab63cdc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-08-31 08:10:02.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-09-08 09:33:33.000000000","tz":120},"subject":"nclu-switch: Fix order in template","message":"nclu-switch: Fix order in template\n\nCurrently description is applied first to the interface, but if it\u0027s a\nvirtual one - it won\u0027t exist before running the configuration commands.\n\nStory: 2010279\nTask: 46232\n\nChange-Id: I40212b40a4d391272f4063f42e8a7043b3d58fd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cfb5c004eafd7df7e0c5e94b7697811e174301ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cfb5c004eafd7df7e0c5e94b7697811e174301ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
