)]}'
{"id":"openstack%2Fkayobe~855909","triplet_id":"openstack%2Fkayobe~master~Ib50dd61685f13c60ace67213ddd1e714a80dece3","project":"openstack/kayobe","branch":"master","topic":"upper-constraints-split","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-09-15 09:01:15.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-09-15 08:59:10.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"34940":{"account":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"last_update":"2022-09-15 09:14:53.000000000","reason":"\u003cGERRIT_ACCOUNT_34940\u003e replied on the change","reason_account":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-09-15 10:18:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib50dd61685f13c60ace67213ddd1e714a80dece3","subject":"Add new DIB upper constraints variables","status":"MERGED","created":"2022-09-05 14:30:27.000000000","updated":"2022-09-15 10:19:30.000000000","submitted":"2022-09-15 10:18:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":52,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855909-upper-constraints-split","meta_rev_id":"9eaa0aa619c6e8ce47058181c89dab16ffc309b2","_number":855909,"virtual_id_number":855909,"owner":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-15 10:18:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-09-15 09:01:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-09-15 08:59:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-09-15 09:01:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-05 16:15:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-06 07:23:59.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-09-06 10:57:38.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2022-09-07 07:35:57.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2022-09-07 08:29:18.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-09-13 09:15:31.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2022-09-13 09:15:36.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-09-13 13:11:05.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REMOVED"}],"messages":[{"id":"f64aacdd701513204cebefc0d2e138e8e8c6af9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-05 14:30:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b53f0f9544d9370c17edffab146de60487cb4e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 16:15:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca196d4c050242dc8f6fad3b9a608347\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47037b51046d4e9785dbd0571c15d942 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c863ac1945e49f18bf2d33fc5ffe401 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d44d586d3454465b16fc76adf0fb08f : SUCCESS in 9m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4230d2390e8e404b94de164d9e1eb989 : SUCCESS in 6m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4de61098fbb94878a7c49c18bedc1940 : SUCCESS in 9m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e57ec0e34af0460a90965a67e99e52c9 : SUCCESS in 9m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0ce5f8b36e64c57a7a85fe36ce66a93 : SUCCESS in 6m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8bc8140379d8472d807abb9ec27209a8 : SUCCESS in 5m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3da45689ecfc4eeb8033f44abe8a0b2e : SUCCESS in 16m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e4c747b22dc4484a9b22f94dd825d5eb : SUCCESS in 15m 19s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/41a1404e2d86495092952dd6fa718878 : SUCCESS in 1h 05m 43s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/2d1ec9b1973141e3a9f8881628f6189e : SUCCESS in 1h 15m 08s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d280380222674fc1917db9b9d7682551 : SUCCESS in 50m 26s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6917333c0c4e40cb838a85b545dcdaad : SUCCESS in 1h 06m 55s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/23c0dcc9678246d0814a7fc3142cdaa9 : SUCCESS in 21m 30s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/76d521a04ec44f419586c5e51c1d2c42 : SUCCESS in 22m 07s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/839b02c45a93488e9264972c9634a83c : SUCCESS in 19m 43s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7ba1360fca894ea6b42c571ee32f7e1a : SUCCESS in 1h 32m 59s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30404c1459b74d6aa0b618b40371d51a : SUCCESS in 1h 15m 05s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/20f4e038f23045ff9cf6d4f64246bac5 : SUCCESS in 18m 37s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/b2164376490e4639a46326a855e8c2fc : SUCCESS in 19m 38s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4162d21403d4b3291595ac79b27f4ea : SUCCESS in 20m 36s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/36b582d415c14e36981b0278375d0f37 : FAILURE in 25m 22s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b9210677a71b442280501366362df26f : SUCCESS in 45m 25s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed154f369a5140bab3e30898dd8913f8 : SUCCESS in 29m 28s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4e43668c98d74e439ef16678662b2b21 : SUCCESS in 20m 13s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c447d39623c1458b86ec3e81d9b0ea0c : SUCCESS in 15m 45s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a526903cefb5457a959d26b73ee67ae3 : SUCCESS in 10m 55s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/48293af511b3421e82d61521ed90261b : SUCCESS in 22m 52s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e59c006d29bb42f0b60d244a1f428b58 : SUCCESS in 17m 26s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9248f580eee04ee3997dabf0aa3ffc18 : SUCCESS in 12m 47s","accounts_in_message":[],"_revision_number":1},{"id":"a56318c16cda6fdc171f3256e0a1473ff984a86d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-06 07:23:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7ff8fee4ec5f1f2d8ae8be44321a2617472cc8b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-06 10:53:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0bea7b54d3518e2c6e24bb8087f34eba483dbb28","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-06 10:57:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8e65f2e865a403d57a7317348b6d52cd1035667d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-06 11:03:16.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"1a7237c9452e1c9fe8b1ece0d3dc98ea594c3e9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 13:03:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/20f54d9e95d244689cb66050d62d26ce\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49a78481709441d8b30dcb85e2c4ceb2 : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a05cc19df164e10bc1773af6d15529c : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbf783c9aa2046e9b64fd96e11c68e50 : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a829ce594ca4ddf9d5532eb73c65ea5 : SUCCESS in 6m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40d9ece54e0642c1ac1b276b8e0e3633 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8e32f389e444179be049a8c10f79137 : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4754c3f452504f8287b230d6c75b67be : SUCCESS in 4m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/df8d09118f22490c944193c7527cf82d : SUCCESS in 6m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4efb381053064fad933d9ff673bf225c : SUCCESS in 14m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/de0716c284374c68996130d744491627 : SUCCESS in 19m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/43a116219d9a4bd49d2218f3077b8c92 : SUCCESS in 1h 10m 41s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/32aaba21530048728b0180ca69d97349 : SUCCESS in 1h 09m 50s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e231943a7f6e45608544b1651ea0582a : FAILURE in 47m 21s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d2f9b0cd93b347f2a104126bb45f1d73 : SUCCESS in 1h 11m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d3c0de78a615459dbf23566605f27c21 : SUCCESS in 24m 54s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/d268bf4414fc4b58b40c03ada4e896fc : SUCCESS in 24m 02s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/809e8a75f5424616bff126aa792b8239 : SUCCESS in 19m 25s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d34977effe1840e1af963106aeb95f8b : SUCCESS in 1h 56m 55s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ad59f19ff754ec38133430db2fcf74e : SUCCESS in 1h 22m 14s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/68b18f938d3648edba9a3ad7e5a1b8d6 : SUCCESS in 22m 18s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/d100c1eacc61435d85b5f0919d8d596b : SUCCESS in 21m 37s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c68b4cc00e8b4519a233d067d0c28234 : SUCCESS in 21m 59s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/6e0856c1fac54e5cb9ccfab1f0c44723 : FAILURE in 32m 47s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/415aeaf8b3a94a608103e237c262fe77 : SUCCESS in 43m 15s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98ee9ef1eaaf4e8b81c0cea7336a7b69 : SUCCESS in 34m 28s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2442f9c88503448684ae312c88a8ec67 : SUCCESS in 21m 16s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/021ec8bcbe394db3b8876387597d2d3a : SUCCESS in 18m 15s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ca729e9f6e94c708c151427791d5c02 : SUCCESS in 14m 17s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/7733732b436b4e71933ac96b266f12fb : SUCCESS in 19m 41s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/49709c41452f4618a94f81ca71c02e1f : SUCCESS in 19m 28s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a39f1a1cd2349a4bada89d72aada11f : SUCCESS in 15m 15s","accounts_in_message":[],"_revision_number":3},{"id":"8735f7d522e15ec4fd067473a29e5dc94b554a5a","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-06 13:46:44.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b880887bea74811003cc9f7b3aebe0c7d2342430","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-06 14:20:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3663af70804d22f68eff62595244509c9c6cd6bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 15:49:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d41e738a93f349819fbe69699db64878\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eaa7688054164175b54b9bae57d63859 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7472c2d6a15346439bef105059fc033f : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3853b0052ca84e06ad3fd4e1683912a9 : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc394bae6c064b65b4c5b75ff491a370 : SUCCESS in 6m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9bb4f1193d2c42ab8cdee3a02bbd5700 : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fa6dd5a1afd400394eb48e1b6b70cca : SUCCESS in 9m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bfea0ad7b91487b92c1c7d031056a19 : SUCCESS in 4m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/688a8ef5c5a34e58b8a9d5fd047d2ac5 : SUCCESS in 6m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c9c0b114f4eb44af80dee95105c1aa78 : SUCCESS in 14m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f34eebed512e499f919b076518cdfbaf : SUCCESS in 18m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/cb1e163b810a4c6cb1d8cf0c9062c548 : SUCCESS in 1h 16m 10s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/b2572f5e6f974de392b99cbbbe33cd6c : SUCCESS in 1h 14m 57s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9ace69f46844dfd9373bcae5cbe963e : SUCCESS in 50m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/891a07ff64314ce194098a74fd12135f : SUCCESS in 1h 11m 09s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e330cb168b534a5dbade2b7c2e45d262 : SUCCESS in 23m 17s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/9bd195882525440fbb178dbe8b6fb8ba : SUCCESS in 23m 21s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/069acc3da5e94b01a5fcb523d158f42e : SUCCESS in 17m 37s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3772c81764e545cd998bd0a343b7fee2 : SUCCESS in 2h 00m 40s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7f14470901c4c59833e035551c97d57 : SUCCESS in 1h 07m 54s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/658ad2bd24c540cd880eabb6d8e05eee : SUCCESS in 27m 12s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/7df3f4ced42743a0b47b087d2e7f9694 : SUCCESS in 19m 38s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/283fbb7a34b3404288ab5c585c11270c : SUCCESS in 22m 37s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/a5d5c11bc6244db7b6df38acfad8af08 : FAILURE in 33m 00s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/114122be4979446187bf117ecc33ae72 : SUCCESS in 46m 47s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ee7545b3281472c9409ec60ab10bd2d : SUCCESS in 26m 21s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e900aefc9a294abc99bd8ac6a65034e4 : SUCCESS in 17m 10s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/d159e992e0ac49099d8fd6a85ef0aeb1 : SUCCESS in 16m 50s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/685934065e8b4c7da0feac0a7e2d0f7f : SUCCESS in 15m 06s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/7b9f33c1105741d6967e65d8072d4739 : SUCCESS in 23m 03s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e6bc15694ebd403d86edfa49fc06d989 : SUCCESS in 19m 12s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a79559334648455e9600e1c4ad0484f1 : SUCCESS in 14m 30s","accounts_in_message":[],"_revision_number":3},{"id":"3ef412fb0b60742b1592e3e26cdb420b4b8fcc65","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-07 07:34:14.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e21538aa13a21956eb6f13808f1bf94790ae5287","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2022-09-07 07:35:57.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5b6c5b958ce4d32573b1a5b71df4edb935453d34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-07 07:37:12.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"5fe29b8fa9350def91b80ad13511290a8749b53f","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-07 07:37:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4a4ed3342c288005ee16486364bcb48062b994be","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-07 07:41:17.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"92afd124ec236bec4d08c366baa5eee4dd4dffba","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-09-07 08:29:18.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"c2d0d547965e2362b49bffc4d3d5b96d26187daf","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-07 08:48:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dc43222a45ba3d5bfc61665bfcb5dbf4d232b208","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-07 08:52:36.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"ddd84a9f89647514f4e2c5d0e74bd780b2e4d1d4","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-07 09:09:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b2147f34755e9d30e68267a51647113160fd7095","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 09:43:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ef2b885bb5a4074a534af4b73579d31\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51804e736c764fdc9548ece7e4f0948e : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8cef9a14492446eaf523c3ef0e05984 : SUCCESS in 8m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f22b1b21775494d8121b9b2fe991c19 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54586ed7a6da4b73a6f473e6d598e98b : SUCCESS in 8m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/487f572105414e498011a644abc3f340 : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db8def30490548518f22075d6d75116e : SUCCESS in 11m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92506a882b6a49028f09b3a7fa7340d6 : SUCCESS in 6m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0006a0128d7f45fa9db9f8f9a77a052b : SUCCESS in 6m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/68fea2cb85d14f88a9ca40d16f0e2fd9 : SUCCESS in 16m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fa6617736e3d479ca2b0d1dff0e1eaa6 : SUCCESS in 15m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d4d5dff799f1405c92ebe3101181ead7 : SUCCESS in 1h 10m 08s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/4ba1bbb1486a4162a495b044af27ef60 : SUCCESS in 1h 15m 37s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b85f40e90ef4f7eb75bfae5ee0f2b1a : SUCCESS in 1h 00m 29s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/fe2796f981b243e8a94e4265711816b3 : SUCCESS in 1h 12m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/522649a847c34a94b1035cdf703d68dd : SUCCESS in 27m 48s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/57dde2076ba644009203e8e61b2f4294 : SUCCESS in 22m 36s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3522a7cf02294f4c869acb2ae5def209 : SUCCESS in 15m 59s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ab68449289f14b36b2e829855e8cfd46 : SUCCESS in 1h 50m 25s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5978b87b6db54e249f73cbfcb2885c05 : SUCCESS in 1h 13m 49s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/27d834a783494047882610628a9ed678 : SUCCESS in 25m 49s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/a5741bfabf014952ad761a0432d3b0a1 : SUCCESS in 20m 54s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e81846e53193435e98f114a103bc3c94 : SUCCESS in 17m 29s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/708703d9102d4aeea04edf25365f4270 : FAILURE in 31m 14s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9d501d780d24433987a6b21b726ffc4c : SUCCESS in 39m 43s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f5748c65d7043b194a49e7aec0abe47 : SUCCESS in 34m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/44498b1462954eecbe2f6f54840fc187 : SUCCESS in 14m 55s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a852e6ddba3145a388c17e606a1a7859 : SUCCESS in 17m 10s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ab2a4a903ee4d47bdd9a2fd93a09520 : SUCCESS in 11m 18s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/29fc0654c4c64e268724afcbc19736b0 : SUCCESS in 18m 03s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/5ea2564fbde448bc9769bd53b8c783f4 : SUCCESS in 19m 09s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3ff289b6c65439a9f1545986dde6dd3 : SUCCESS in 13m 04s","accounts_in_message":[],"_revision_number":4},{"id":"769fc53c798ebaba50a902bbcbd02874523ada7a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-09-07 13:30:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bb8869eff570c2ce50a86ca24d7097d45e773209","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-09 09:51:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3cf1f4c8806941f7999791e7833ac544033cda12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 09:51:54.000000000","message":"Patch Set 5: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kayobe for 855909,5","accounts_in_message":[],"_revision_number":5},{"id":"03952f5ccbf3b8171970788a3892f3b891099b6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-09 10:16:54.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f798e159952e7b2e49abf43b598a9b922271d924","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-09 11:28:54.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"78e12d97ce749f0bd9d3d1beae8c139c6abb0b50","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-09-09 12:36:33.000000000","message":"Patch Set 6: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":6},{"id":"566e3a47f4f1bf21f00cdffb6a6f827d163d3e47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 12:44:07.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bbb9e22119846208ff79d9aa9f0e1b5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/354e99a6df514deb824c028edc5a8889 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bb8740fb53041bb99da1369009d786a : SUCCESS in 8m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca77060be4ec4a65899b8704337c0ae6 : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f259ddb2fa44a29961353f5745a57c0 : SUCCESS in 9m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1aff31e42a3d467e83f820604f52cd26 : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7a25bc5695f4f1fbf54169b4584cfbd : SUCCESS in 11m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4657c0b3acd4ab4b0d4d292233be369 : SUCCESS in 4m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e149e5e5c33f4b5e99bf6d83dbfc3f97 : SUCCESS in 8m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/89c1eee7870c41dcae38eb3c52ffbbcb : SUCCESS in 14m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7cd6ce20c16f46d48743a7057c742641 : SUCCESS in 20m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/cac54bcfcb564556b16a967e7a40b386 : SUCCESS in 1h 05m 05s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/5ea840cdd86b4e6d8d1bbf6b715881aa : SUCCESS in 1h 15m 38s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ce4cc0a85c246bf99d12209b4665acb : SUCCESS in 1h 02m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5e9d0ed2ad7f4ab5b5087c048bede072 : SUCCESS in 1h 05m 28s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/35d52617167248ae801cc1199e10ff63 : SUCCESS in 29m 12s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/2b8ff08e9bc24e8a91b218950e096e8d : SUCCESS in 24m 11s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c19b8d60a4f841ed820a09fe733ca56d : SUCCESS in 15m 52s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/dcf8858f870249b3bc6c65e298cb2910 : SUCCESS in 2h 23m 54s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d98f9a6ddd5345d4a2db402c46ad2fa9 : SUCCESS in 1h 08m 08s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/05c28d69c1594f2abd1bb0d880746561 : SUCCESS in 22m 56s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/dbeac6dc0aa84905b42d738ebead9467 : SUCCESS in 23m 05s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92583e5c115a42e18b931c690e83448d : SUCCESS in 19m 43s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/707d14bb32564b92a29d958871e9fa9b : FAILURE in 28m 05s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d45ab9b3aeff4afcb93dbf6a2e1b2b8f : SUCCESS in 1h 09m 27s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11fe003a6bea4263ba69f8bb25c6b3df : SUCCESS in 34m 49s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2984d5ec4255479993441905f2439991 : SUCCESS in 14m 07s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6c9d6bc9263a43649c9bf8a64b9f8793 : SUCCESS in 17m 18s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8a1458888db497cba7294e8efee265e : SUCCESS in 16m 42s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/b2940e73d81240b88354d6fadaaae756 : SUCCESS in 21m 29s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/10f3a23e7e324e649c1af024adc492ba : SUCCESS in 19m 44s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/635e630981264bb88f5f3ebf799b704b : SUCCESS in 14m 27s","accounts_in_message":[],"_revision_number":6},{"id":"0c087eee950c31900f7845112a7c33b8801d4b7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-09 14:48:22.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8f95eea13786ad388eab400949c6c86ad0365767","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-09 14:48:29.000000000","message":"Patch Set 6:\n\n(9 comments)","accounts_in_message":[],"_revision_number":6},{"id":"0bed4abf99371185e60fe6ae4d68a6304f6b0e69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 16:30:40.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9ecc59f5d404637be6b210bddb14bb8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28c7ad44575f4328b61743021024ab2c : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb1d053981e849f1bb9b4b320031ca74 : SUCCESS in 8m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c17b3c01aa3c4bb0a06ee1e99f7a03c0 : SUCCESS in 10m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ed108ea744e447297bde11bcbf8645e : SUCCESS in 9m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5857826ac0cf460683f58d60bb43e728 : SUCCESS in 9m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/705b99ec34b64ca2a3999e57fbc68a24 : SUCCESS in 13m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/728909175a0d46edba0bbb110cebd706 : SUCCESS in 7m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/29d62ddc6e4042728f9ab966921c3ed0 : SUCCESS in 9m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7310cbff950b4ee5b6f1704930e10dac : SUCCESS in 14m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d9c158232881491893908e3e571685ba : SUCCESS in 18m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/57ada5198ecf41068bbdb78c3bea92c3 : SUCCESS in 1h 11m 20s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/098c21c7eeff4d209b36d528c5545678 : SUCCESS in 1h 10m 02s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/866811ccdc544fc3a68558e4346d6708 : SUCCESS in 51m 24s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/cc6cd90432374bee8ac930593be6ccfc : SUCCESS in 1h 10m 46s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ad9a61f9c08644a894cd03b4867a98bd : SUCCESS in 22m 16s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/c73e9cd7adb74c679eb657c0cc4adf63 : SUCCESS in 23m 41s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8810365cccc43acb78d114d55c71c3f : SUCCESS in 15m 22s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/052ab4fe7abb4651bb9e22c8400cc685 : SUCCESS in 1h 29m 31s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/201d984526db4d2e9b09fe92b2859ee6 : SUCCESS in 1h 21m 25s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3cc3ae3e4b4248659a02f8d5c921db9f : SUCCESS in 19m 41s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/15863ff8077a4af0a1978cad46d4581f : SUCCESS in 22m 22s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bad6ec52f0ec4328a1caa9715f44000f : SUCCESS in 17m 02s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/c2a4369a8da24d4884dd9a4d2bc5b964 : FAILURE in 34m 09s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/894cbda0d6da48c4b807335e794fec62 : SUCCESS in 42m 19s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8668b1ac67840f3bddc4eac058c66c5 : SUCCESS in 33m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0f4eb7cb05c54aaa80a12d4549a070ef : SUCCESS in 18m 25s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/d3a61adc339c46ca9f4e4fde4d7e2e51 : SUCCESS in 18m 14s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eff7a532fda8458c829567d3052d6802 : SUCCESS in 13m 14s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/dde4cd1a24b046d78f655dc7cdd1a65c : SUCCESS in 22m 13s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e3f829cd811a491692b1c88625483847 : SUCCESS in 21m 17s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/687bd1e810f5471183afa7e71509e70c : SUCCESS in 12m 12s","accounts_in_message":[],"_revision_number":7},{"id":"40221b4dbb8470b386bb5d2c8716862bef302d5b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-09-13 09:15:31.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"f09bf3953da1887a61c67d0643a617b04d1d7317","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-09-13 09:15:36.000000000","message":"Patch Set 7: Code-Review-1","accounts_in_message":[],"_revision_number":7},{"id":"5f9cef44c4409b942b4e5dd90cea8282ce290341","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-13 09:18:12.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a5b2d9992664692ac7a0d68b8807e37b72c96122","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-13 09:19:21.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3a62343d404aa16801f08bc76115433c7d108103","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-13 09:24:11.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c7b5aa7149faa40d2245cf95fd624da28b4c3e37","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-13 09:25:50.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"88aee8f4ff4774e339f5b79fd223bf99dce5c227","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 10:49:37.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/312c2d4dcfa04a54981845c603f70ed4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/144be474528e47f5be7d229cc006dec5 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98689677b5604c44a530c8b06739501a : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cd57710cee94406b8fbce98d229766d : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a59fb157befa4cabb4ebe6bae8ec946d : SUCCESS in 7m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00fbf77022674e0b8df2f686bb7f1268 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf58a7e2e6a04b0ba81039846bcc9e3b : SUCCESS in 9m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e181d28801041a08d501a0dfb88c2fb : SUCCESS in 3m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/485a6af05ab646e099fb1fdb19c6c012 : SUCCESS in 7m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8af57ec9641a419d8a712114b86f71fa : SUCCESS in 14m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7a8ea90085424d5a81b3bb8519487c9c : SUCCESS in 17m 12s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/57cfa959e0e14078873906c3c487f69b : FAILURE in 23m 38s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/89b65d2dceaa416c9ca4a8fbb8eb0264 : FAILURE in 25m 46s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e4d401b9175423482460d58c6b45b64 : SUCCESS in 45m 19s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9dcb8b8b91b340fe811b02c95751c816 : FAILURE in 28m 02s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c9c0c536b02b4ce7aee80becc7a5e55d : SUCCESS in 19m 38s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/334fcfdb8dc549b2a76270ed31a4ce09 : SUCCESS in 21m 28s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33ff125a5361449bb10fe8c5c086dafd : SUCCESS in 16m 40s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/48270c72dc9744d3bbbbf0169466ce79 : FAILURE in 1h 20m 42s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d831a8030c849e18ccb7e11bb849aa6 : SUCCESS in 57m 37s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/de7d3401d3d445d386b67efefd43b511 : SUCCESS in 25m 30s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/2794f219b6f941e1b7956068b664aaf5 : SUCCESS in 23m 17s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a5a8b86a639477085a0b21edf3a5273 : SUCCESS in 19m 04s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/1b21e46f50ca4d9a84670e1817607892 : FAILURE in 35m 59s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b5c0501d54d3489aab16cd6d28ab1848 : SUCCESS in 43m 35s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d95b67672781441d8f5011f43a3e85aa : SUCCESS in 33m 54s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/dec2f9b2291a421a98701693aea21ee0 : SUCCESS in 18m 21s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/b669deacfd7f4ff184022c678149d4ad : SUCCESS in 16m 49s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a2a39974bbd43be9eb5ba4b7c7e3e20 : SUCCESS in 14m 29s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/551d5ef320cf453cb22a5f271823a51d : SUCCESS in 23m 38s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/402539aa660f4ae9892c73598ad2fdae : SUCCESS in 17m 36s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/682f8cacd2604c7baab76bb2ca0637c7 : SUCCESS in 16m 41s","accounts_in_message":[],"_revision_number":8},{"id":"faa112a2735e6b95c28ac45489d0c57615f9adab","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-09-13 12:46:04.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"191b39d23237589a9f0c342c54a5fa3aa62874f2","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-13 12:51:15.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"426fc32b655d76db3f6b4f694c6f13472bfdb871","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-09-13 12:55:00.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"35553f29c99d57d6ba503b952553cd5a71af0006","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-13 12:57:38.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8adc7310e570844959d637c942e857127ece8dbe","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-13 13:11:05.000000000","message":"Removed cc \u003cGERRIT_ACCOUNT_24072\u003e.","accounts_in_message":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"_revision_number":8},{"id":"1fc44f2ecd513932f819c2a639e1aeb2b539ede4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 14:26:35.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b295a9f71ef441e8dc7ffdd959dda79\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/610f2ae229374c2781fa1b689da55785 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae191a7d888c486fb498ee821b518a24 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ceb508e880e48a19856f11418ab645f : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48f42c83034746358b7265ef0be30285 : SUCCESS in 6m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9947adf4138e4a54a982416d23ab9987 : SUCCESS in 9m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cc0db45607c48b7bda58191d327c6e4 : SUCCESS in 11m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c6294823a984bd3a51e68bcded827c7 : SUCCESS in 6m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a721c6970c8d4cb1afff50dbdc168646 : SUCCESS in 8m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ac99d6db84ca467db23616d6b0eaae2b : SUCCESS in 15m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e8ab03de435b4ee8bc411f270de43397 : SUCCESS in 19m 28s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f27a5489cc554b4092fedfcf6cb0d347 : SUCCESS in 59m 32s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4da63200a3444fdfb75672990d55356a : SUCCESS in 18m 24s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5af2c9a79247470a80441da9d2156d68 : SUCCESS in 1h 32m 18s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa5bd7d6d8c749dbaffd1c5c3101c6a4 : SUCCESS in 21m 29s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7b712123ea14bc79316e9aa220e0821 : SUCCESS in 29m 17s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa345697b3f142e7b267180e3a363989 : SUCCESS in 15m 33s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01796d767475496e9106fec58314f9b4 : SUCCESS in 16m 25s","accounts_in_message":[],"_revision_number":8},{"id":"295a2b09a4be1bd3f511300ef3d9d9257cbd1170","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-09-14 08:28:34.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b56e615f627d3489529d89f438fbc67f1e8e0662","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-14 08:47:29.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1f14c0849065a00c9cc38ff2ce2b8f15c7c77a48","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-15 08:55:10.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8c1f77bc407e5ea1616656af7af94ce315f98cec","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-09-15 08:59:10.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"8151e3ca0304b8b127c0168afa7739d74c18be68","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-09-15 09:00:09.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b482760a6d9323f2ec1a962332e6543ba24cbbaf","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-15 09:01:15.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"3e1470599181c6ce5cc8e66f401c64dc9bffba98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 09:01:46.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"8c8b3713fa161179e981764b30a8b938c683acb4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-09-15 09:07:56.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f83544145da5fbfdbc6ef1ccbf91db49b00b0fac","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-15 09:14:53.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b76145a3e2c0b5d5c529a0678b083acc2a089dd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 10:18:11.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f68c0270072047a3b850d93757b8fd44\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48182bb21dc441a4a2dcbb1ff5fcef27 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c9203b982414c6d8855af6deb309feb : SUCCESS in 7m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c57a7758aa894a34b8b6a660baf973e7 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33677d8e678844b7a203f72d6846c233 : SUCCESS in 11m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/510ff4139ac94d03bf8b76239a35b069 : SUCCESS in 4m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/81594c14f6ed426eb775e1b1ad983579 : SUCCESS in 7m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4e23613ba56a4bab86efaed654eb3521 : SUCCESS in 14m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/25db9f26b1d74079a645f02546dd826f : SUCCESS in 17m 27s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab597a8b09b04c24b68def35fe709102 : SUCCESS in 56m 44s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6c06bd1c4204cfab779994c3032fb7a : SUCCESS in 17m 35s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c04908fafb7c43b280b86a4088655e88 : SUCCESS in 1h 15m 30s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb6db088baf6494186f1237a8ca9d7c6 : SUCCESS in 19m 12s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95b09ff7dee9460d9bcd0f2cd4d0fff0 : SUCCESS in 33m 55s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2bae7387cd74fcd8cb6bf32359722cd : SUCCESS in 14m 39s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0c407b17cd24388b8ccd5063faf45e9 : SUCCESS in 16m 27s","accounts_in_message":[],"_revision_number":8},{"id":"29101f99bf6f5dd3290b8649e30ee477f4db6ee3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 10:18:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"9eaa0aa619c6e8ce47058181c89dab16ffc309b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 10:19:30.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d49c16000ba0413eb8d533b1945a624c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c46ab7360974dd3b454f10ad9e5f314 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a08a6d7f75d4ea29393f6b15fc19eb8 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"e10f176ce1356fc270ade868a97b01dc62902939","revisions":{"ca43f2795112e13f8c4d8566f6535c172d77ad06":{"kind":"REWORK","_number":1,"created":"2022-09-05 14:30:27.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/09/855909/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/855909/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/855909/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/855909/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/855909/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/855909/1"}}},"commit":{"parents":[{"commit":"6e2f5a1884745dbb596ce6f127e7eb875ab63cdc","subject":"Merge \"Fix IP allocation limit in overcloud inventory discover\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e2f5a1884745dbb596ce6f127e7eb875ab63cdc"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-05 14:29:38.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-05 14:29:38.000000000","tz":120},"subject":"added new variable for DIB OS images","message":"added new variable for DIB OS images\n\nos_images_dib_upper_constraints_file\n\nChange-Id: Ib50dd61685f13c60ace67213ddd1e714a80dece3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca43f2795112e13f8c4d8566f6535c172d77ad06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca43f2795112e13f8c4d8566f6535c172d77ad06"}]},"branch":"refs/heads/master"},"2bd8f82c5f257ae49a7bd497cbf175edf64a0f58":{"kind":"REWORK","_number":2,"created":"2022-09-06 10:53:26.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/09/855909/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/855909/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/855909/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/855909/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/855909/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/855909/2"}}},"commit":{"parents":[{"commit":"6e2f5a1884745dbb596ce6f127e7eb875ab63cdc","subject":"Merge \"Fix IP allocation limit in overcloud inventory discover\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e2f5a1884745dbb596ce6f127e7eb875ab63cdc"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-05 14:29:38.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-06 10:53:02.000000000","tz":120},"subject":"added new variable for DIB OS images","message":"added new variable for DIB OS images\n\nos_images_dib_upper_constraints_file\n\nChange-Id: Ib50dd61685f13c60ace67213ddd1e714a80dece3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2bd8f82c5f257ae49a7bd497cbf175edf64a0f58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2bd8f82c5f257ae49a7bd497cbf175edf64a0f58"}]},"branch":"refs/heads/master"},"a6964940481ed46c254df78f509d0809a17d6a22":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-09-06 11:03:16.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/09/855909/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/855909/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/855909/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/855909/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/855909/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/855909/3"}}},"commit":{"parents":[{"commit":"6e2f5a1884745dbb596ce6f127e7eb875ab63cdc","subject":"Merge \"Fix IP allocation limit in overcloud inventory discover\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e2f5a1884745dbb596ce6f127e7eb875ab63cdc"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-05 14:29:38.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-06 11:03:16.000000000","tz":0},"subject":"added new variable for DIB OS images","message":"added new variable for DIB OS images\n\nos_images_dib_upper_constraints_file variable\nwas added to facilitate control over the DIB\nupper constraints without changing them for\nother components\n\nChange-Id: Ib50dd61685f13c60ace67213ddd1e714a80dece3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a6964940481ed46c254df78f509d0809a17d6a22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a6964940481ed46c254df78f509d0809a17d6a22"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e26999e7c3c373af7865cbf5073a314e0f008ab0":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-09-07 07:37:12.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/09/855909/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/855909/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/855909/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/855909/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/855909/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/855909/4"}}},"commit":{"parents":[{"commit":"6e2f5a1884745dbb596ce6f127e7eb875ab63cdc","subject":"Merge \"Fix IP allocation limit in overcloud inventory discover\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e2f5a1884745dbb596ce6f127e7eb875ab63cdc"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-05 14:29:38.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-07 07:37:12.000000000","tz":0},"subject":"overcloud-dib: add os_images_dib_upper_constraints_file","message":"overcloud-dib: add os_images_dib_upper_constraints_file\n\nos_images_dib_upper_constraints_file variable\nwas added to facilitate control over the DIB\nupper constraints without changing them for\nother components\n\nChange-Id: Ib50dd61685f13c60ace67213ddd1e714a80dece3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e26999e7c3c373af7865cbf5073a314e0f008ab0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e26999e7c3c373af7865cbf5073a314e0f008ab0"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2f9176a1052e90700d6263a33c4c5334090d3c5d":{"kind":"REWORK","_number":5,"created":"2022-09-09 09:51:31.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/09/855909/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/855909/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/855909/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/855909/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/855909/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/855909/5"}}},"commit":{"parents":[{"commit":"6e2f5a1884745dbb596ce6f127e7eb875ab63cdc","subject":"Merge \"Fix IP allocation limit in overcloud inventory discover\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e2f5a1884745dbb596ce6f127e7eb875ab63cdc"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-05 14:29:38.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-09 09:41:41.000000000","tz":120},"subject":"added new dib upper constraints variables","message":"added new dib upper constraints variables\n\novercloud-dib: added overcloud_dib_dib_upper_constraints_file\nipa: added ipa_build_dib_upper_constraints_file\nadded os_images_dib_upper_constraints_file to build jobs\n\nNew variables were added to facilitate control over the DIB\nupper constraints without changing them for other components.\n\nChange-Id: Ib50dd61685f13c60ace67213ddd1e714a80dece3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f9176a1052e90700d6263a33c4c5334090d3c5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f9176a1052e90700d6263a33c4c5334090d3c5d"}]},"branch":"refs/heads/master"},"29b640c0f70c37c2fb617bf12b6e0dbc084f05fb":{"kind":"REWORK","_number":6,"created":"2022-09-09 10:16:54.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/09/855909/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/855909/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/855909/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/855909/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/855909/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/855909/6"}}},"commit":{"parents":[{"commit":"4a3f88694e04067bd9b266143ddae12260417d11","subject":"Merge \"Move group_vars to an inventory directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a3f88694e04067bd9b266143ddae12260417d11"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-05 14:29:38.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-09 10:16:52.000000000","tz":120},"subject":"added new dib upper constraints variables","message":"added new dib upper constraints variables\n\novercloud-dib: added overcloud_dib_dib_upper_constraints_file\nipa: added ipa_build_dib_upper_constraints_file\nadded os_images_dib_upper_constraints_file to build jobs\n\nNew variables were added to facilitate control over the DIB\nupper constraints without changing them for other components.\n\nChange-Id: Ib50dd61685f13c60ace67213ddd1e714a80dece3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/29b640c0f70c37c2fb617bf12b6e0dbc084f05fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/29b640c0f70c37c2fb617bf12b6e0dbc084f05fb"}]},"branch":"refs/heads/master"},"6d8d0c259d261fc0f83989c07114194bf703bd66":{"kind":"REWORK","_number":7,"created":"2022-09-09 14:48:22.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/09/855909/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/855909/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/855909/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/855909/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/855909/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/855909/7"}}},"commit":{"parents":[{"commit":"4a3f88694e04067bd9b266143ddae12260417d11","subject":"Merge \"Move group_vars to an inventory directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a3f88694e04067bd9b266143ddae12260417d11"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-05 14:29:38.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-09 14:47:04.000000000","tz":120},"subject":"Add new DIB upper constraints variables","message":"Add new DIB upper constraints variables\n\novercloud-dib: added overcloud_dib_dib_upper_constraints_file\nipa: added ipa_build_dib_upper_constraints_file\nadded os_images_dib_upper_constraints_file to image build\nplaybooks\n\nNew variables were added to facilitate control over the DIB\nupper constraints without changing them for other components.\nThey are empty by default in order to facilitate Rocky9\nimage builds.\n\nChange-Id: Ib50dd61685f13c60ace67213ddd1e714a80dece3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6d8d0c259d261fc0f83989c07114194bf703bd66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6d8d0c259d261fc0f83989c07114194bf703bd66"}]},"branch":"refs/heads/master"},"e10f176ce1356fc270ade868a97b01dc62902939":{"kind":"REWORK","_number":8,"created":"2022-09-13 09:24:11.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/09/855909/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/855909/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/855909/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/855909/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/855909/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/855909/8"}}},"commit":{"parents":[{"commit":"4a3f88694e04067bd9b266143ddae12260417d11","subject":"Merge \"Move group_vars to an inventory directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a3f88694e04067bd9b266143ddae12260417d11"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-05 14:29:38.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-13 09:23:58.000000000","tz":120},"subject":"Add new DIB upper constraints variables","message":"Add new DIB upper constraints variables\n\novercloud-dib: added overcloud_dib_dib_upper_constraints_file\nipa: added ipa_build_dib_upper_constraints_file\nadded os_images_dib_upper_constraints_file to image build\nplaybooks\n\nNew variables were added to facilitate control over the DIB\nupper constraints without changing them for other components.\nThey are empty by default in order to facilitate Rocky9\nimage builds.\n\nChange-Id: Ib50dd61685f13c60ace67213ddd1e714a80dece3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e10f176ce1356fc270ade868a97b01dc62902939"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e10f176ce1356fc270ade868a97b01dc62902939"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
