)]}'
{"id":"openstack%2Fkayobe~856454","triplet_id":"openstack%2Fkayobe~master~I8129b2a6793ea36bb296ef8337a5461f73d6e557","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2022-11-07 19:15:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8129b2a6793ea36bb296ef8337a5461f73d6e557","subject":"Adds missing sasl options to compute.yml","status":"MERGED","created":"2022-09-08 09:51:52.000000000","updated":"2022-11-07 19:16:37.000000000","submitted":"2022-11-07 19:15:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856454","meta_rev_id":"480d6303f9465af0ce93e6e92e21059bd3315150","_number":856454,"virtual_id_number":856454,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-07 19:15:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":1,"date":"2022-11-03 07:17:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":2,"date":"2022-09-13 09:07:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-07 18:00:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-07 18:00:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-08 11:44:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-13 09:07:05.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-11-03 07:17:12.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"},{"updated":"2022-11-07 18:00:56.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"829bc9c651f5f3de71b310ee0d3878251b8737b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-09-08 09:51:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ba1459af93e9576680d661d667d128ba37e1b56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 11:44:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ca4e111c0324649b09aa69d7da5fcb3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d2dc2e2f65c4db392b221c84ad08c0d : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbc8fffed18c4be0ba39b6ad94bbd1ab : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a712bb46fc3b48438315c4ad5bc2c2bc : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9374f6363f8419bab03086f273eec7d : SUCCESS in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8433c495501c455c8bfc823825290865 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2c2ea28c5ea48c1ac54ebc0b4d50c30 : SUCCESS in 10m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/156a43bb7de4424d869c28213933d59c : SUCCESS in 7m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d04feb10b3ff4181b177d111220f7a79 : SUCCESS in 15m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/91236ac729c14cfa9dc7f05c6f24d047 : SUCCESS in 19m 28s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/df9adc9417aa43e2803aa0f57d59855a : SUCCESS in 1h 08m 53s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/c8eaca224c50492d8175ce644a1b84e6 : SUCCESS in 1h 06m 18s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b25e685bb23a4b03957ecd72bd1da3ee : SUCCESS in 54m 58s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/72d1e586c8af46ebbe4817610f8546c3 : SUCCESS in 1h 07m 02s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/685876f3d43641289f6b50c314ced795 : SUCCESS in 23m 13s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/b89172d976d04b218a0e4c5965c1ad4a : SUCCESS in 23m 17s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dfc16a89bb66476e98eee941ca920c52 : SUCCESS in 17m 23s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6d0c651c3a964b968aa58a5d46e40480 : SUCCESS in 1h 48m 25s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b3e53b4ee9b4e039df889cbf55c5467 : SUCCESS in 1h 11m 36s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e9e1ef3a586341baa53cd09c1152c8c7 : SUCCESS in 23m 51s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/2a5768102bcb4a3981a8bfaacf9181b7 : SUCCESS in 20m 35s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee682a07033f4b028302e0a12714bd47 : SUCCESS in 18m 55s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/c46c23186d794489a32efe1beeaa05bd : FAILURE in 31m 56s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7f60beda574245fbb95b31f535d57077 : SUCCESS in 39m 11s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e69b527c74d34847aa3f439bf6498f6e : SUCCESS in 29m 29s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/defbdb1e964a4af68c201e3a216c4737 : SUCCESS in 20m 43s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/b1362740350247c2ba96bf7b56623418 : SUCCESS in 16m 42s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac25c8bebeae4d97a2a99f7009b4b631 : SUCCESS in 15m 27s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e5c247587d6d462f946db162342a5679 : SUCCESS in 23m 01s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/24f3e729b1424d47b4f718b0379ab1ad : SUCCESS in 21m 07s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/752ebb1a6a0d424ba89e2d338d477146 : SUCCESS in 18m 48s","accounts_in_message":[],"_revision_number":1},{"id":"5e0e31aede3fab76c24e35f79116bcca02247ccd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-09-13 09:07:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e90a800c7d2ef34d16ea1668c74d14bbdabd84f3","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2022-11-03 07:17:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bd499a6836e05e69a0fc4c800fa4c30d5cf357b9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-11-07 18:00:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fe76f0f2b2232d2b47162e06a721cfd926620756","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 18:01:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e5c8a4bb9bbdbd7d5ceab514438039f96eb97baa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 19:15:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1ba8a0b81d04adebb9c20743a2072ca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1314a3a055e14deeba0fe9e065b7ea58 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6091b4c35f0457083551b3fc8cd6ee0 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39b27841c6564f61a729f136384c8dce : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e92f3433f2542a8a3bbca6677a6922f : SUCCESS in 7m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/93289f487fe041e4be78d733b718656a : SUCCESS in 5m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8acba00b7e0848019b3535de87c760af : SUCCESS in 12m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/562e23095c7a48a8ab42b57d66152daa : SUCCESS in 13m 18s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38111f0e04664bce971dec4476c39b3a : SUCCESS in 55m 52s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c46f7e52560544609c55a57041b62d14 : SUCCESS in 14m 48s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0461b55f4dad4098a7fcfa28becacefe : SUCCESS in 1h 13m 16s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec09fd5da2ef44998238ce594b5dff25 : SUCCESS in 14m 08s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/009dbb6af49a49eb9851d1c64f1ad38d : SUCCESS in 29m 15s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6ec99ade2de438fa0cc1079aef2d1aa : SUCCESS in 12m 01s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26fb9c07abf74ddbb905b34acad57e15 : SUCCESS in 12m 28s","accounts_in_message":[],"_revision_number":1},{"id":"a5096f38875e55c9cd97ff68d3df0e394421ab11","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 19:15:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"480d6303f9465af0ce93e6e92e21059bd3315150","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 19:16:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0342146f5141412f9c1f2d19082105ff\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f9a87b15ca442219292189c1d0fbd1c : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a4e9d5124ce3179451e977a529e6f6744f20dae5","revisions":{"a4e9d5124ce3179451e977a529e6f6744f20dae5":{"kind":"REWORK","_number":1,"created":"2022-09-08 09:51:52.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/54/856454/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/54/856454/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/54/856454/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/54/856454/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/54/856454/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/54/856454/1"}}},"commit":{"parents":[{"commit":"6e2f5a1884745dbb596ce6f127e7eb875ab63cdc","subject":"Merge \"Fix IP allocation limit in overcloud inventory discover\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e2f5a1884745dbb596ce6f127e7eb875ab63cdc"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-09-08 09:49:56.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-09-08 09:49:56.000000000","tz":60},"subject":"Adds missing sasl options to compute.yml","message":"Adds missing sasl options to compute.yml\n\nTrivialFix\n\nChange-Id: I8129b2a6793ea36bb296ef8337a5461f73d6e557\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a4e9d5124ce3179451e977a529e6f6744f20dae5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a4e9d5124ce3179451e977a529e6f6744f20dae5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
