)]}'
{"id":"openstack%2Fkayobe~857594","triplet_id":"openstack%2Fkayobe~master~Ibc4c9c93f52414c23cd7d8b4f1b32c85cb233681","project":"openstack/kayobe","branch":"master","topic":"bifrost-credentials-fix","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-09-29 16:03:22.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-10-04 10:02:57.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"34940":{"account":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"last_update":"2022-10-04 11:24:59.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-10-04 11:24:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibc4c9c93f52414c23cd7d8b4f1b32c85cb233681","subject":"Update write_bifrost_clouds_yaml logic","status":"MERGED","created":"2022-09-14 09:03:13.000000000","updated":"2022-10-04 11:26:09.000000000","submitted":"2022-10-04 11:24:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":62,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"857594-bifrost-credentials-fix","meta_rev_id":"a780fce23c8c55a27f997b405d43514e1bb2d6ca","_number":857594,"virtual_id_number":857594,"owner":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-04 11:24:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-10-04 10:03:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-10-04 10:02:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-10-04 10:03:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-14 09:50:22.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-09-14 11:30:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-29 13:01:43.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-09-29 13:04:21.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-09-30 08:24:37.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"3b9e33b0e362ec171a9ccc5cb3275ed0e573bf73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-14 09:03:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c109ebc0665fa896b505a44d796bb4dd205d7874","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-14 09:07:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3253e4ce2c8f335583b075ebad5e8d68458f0a15","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-09-14 09:50:22.000000000","message":"Patch Set 2: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8d8191a8e550806cd3fa4d43f46995bd66db859c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-14 10:04:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1c7c13c0ea668e7b5b954fff3f33f4fa2e8a7c43","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-14 10:05:03.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4133b316d3c4d9712a92501bca34aab5f339173d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 11:30:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0960c3879cc64ed3b23d693c74d72875\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7164915b6a2a4add94ba58d22c17009c : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcd51250447f4035a01403c04637de40 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3da680f4ec74403d93d29debe1239426 : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8374db16a1584a9490845cf6d98c150c : SUCCESS in 7m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c14135cc43cc476485d74975c9598cce : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a1500958be04fc6be4272e4bc8749dc : SUCCESS in 9m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33a2212b2661485ca0a2c87f54667fb3 : SUCCESS in 4m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5bd8da69b87647adb16e704cc2c0faf2 : SUCCESS in 6m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f2e2310f06eb466f817703bd39d25cda : SUCCESS in 16m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6918cab694804fb08425387b8b760dad : SUCCESS in 15m 22s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb88b7fe3e734b6d890757bc6d3e888d : SUCCESS in 1h 00m 43s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b109c90a736f427ab146c6045acbc44f : SUCCESS in 19m 49s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a61c5852db441b0afd8d03bfd1c6590 : SUCCESS in 1h 24m 11s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87bbc3da89064a2db4a7ed249a73c00f : SUCCESS in 20m 47s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f557b56193b47edb7476f5ef0687fd1 : SUCCESS in 27m 55s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/861e494508024b37bd38746706e952ab : SUCCESS in 15m 00s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d9ad303a5e44b858547cf74c085c71a : SUCCESS in 13m 14s","accounts_in_message":[],"_revision_number":3},{"id":"720fa327b98ba351a5a7069f8fbb81023645cb3e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-09-14 13:08:16.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"fa9cc024e9d4776cf64f9ca8b5c660ca320403a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-14 13:17:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b02149d76d2f4563329f987fd5b1acf7fe387448","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-14 13:17:14.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"77d3b6d2196fccbc50cacc02eadbeb31a5d2ebdf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-09-14 13:19:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"20b93484a7cb917026eb1adaf0ccf4b71273a156","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-14 13:23:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b4c69126c84648497b231a234ce073761cebbe5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-14 13:28:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2f783c9c0eb67e71d7a36b4d523b116c4def98b1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-09-14 13:47:44.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"8e9ad27bf72542c454231bd888ef1c143d76ba49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-14 14:00:06.000000000","message":"Uploaded patch set 6: New patch set was added with same tree, parent, and commit message as Patch Set 5.","accounts_in_message":[],"_revision_number":6},{"id":"521e63bafdf56549ff7f971c010aa03f46bae097","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-14 14:01:08.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"36714e63219d04e820d8e122e5bd48f0086f27ae","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-14 14:02:33.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"c574914fa7dd369bf6dd4bc724fa6d5158af907f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-09-14 14:13:36.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b5a14b8af2da354dee1b6a008f76a1df3a951d81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-14 14:18:41.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4f945894356b3e3096c843bd3fe7e086290c16be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 15:19:57.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de5ea6b63e064a7aa1fb1854e5f81476\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/596b4704b6ab43bc94bc938de05c6e99 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/063a80d12e2146f8becf00be5987157e : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00984acd71954eeeb512b836833b8932 : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3a5031e59a64790958b7f0750ca05b3 : SUCCESS in 7m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14028a1cd6db4e89b8819a371fd3d7f3 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43ce038055214a599585d460e99336aa : SUCCESS in 8m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/10eb2d74bad04c06b1d391ed7eb303a0 : SUCCESS in 6m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/efb82d367bcf4abd85d34844872142d6 : SUCCESS in 11m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8c70c7fb82f44a0099da125a880a3ef0 : SUCCESS in 18m 04s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a910487e38714cc6a428fcbee6dc2643 : SUCCESS in 46m 44s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c8c1baa693b4aa3b037ea659bfd6039 : SUCCESS in 18m 38s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0922c032178e49108dc1fa71a7d2891d : SUCCESS in 59m 22s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b57ff4070c147708da7d7bd30a4e41b : SUCCESS in 18m 12s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5eb73a0ec294d2a80f8d746d11c6e2b : SUCCESS in 26m 30s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/571ba4392e9143338a84439039dae7f8 : SUCCESS in 15m 00s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2494e8059d5a4e0480cecc7743c81397 : SUCCESS in 15m 36s","accounts_in_message":[],"_revision_number":8},{"id":"cf4885ed5d933cf3b734141f0337bd76be992e8b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-09-14 15:43:05.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7f6abf5e2e4126afaf9dd9f88ef7c47eec736c81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-15 07:45:56.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"8fd42fa04f9a564cd9b8a0a59183e994c484c680","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-15 07:50:40.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"b6dc9bd7052adec3cda44c3092716058ca8d41ec","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-15 07:50:51.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"e6f0ac5c63f0c0432bf57345cbee93b54a6c2cd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-09-15 08:26:49.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"70c1f9df5885a5e0f86556bd324cad302377bf4c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-09-15 08:26:55.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"4076b0aded8808da3f1ba6572d1ed415f6632405","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-15 09:10:23.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"3899562ce164e5ff8021038fac98b6c5e552b294","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 09:47:01.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a714a6a78284dc699641fae2d43fc30\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2dad08bb26c749109a0463507b35737b : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b27c91aaa9344b9ae3820b378ff2978 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42995cf8f93b4bc580ac05010d32dcf5 : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3cd6ab75d1c4b79baf7f47c2aa6e59f : SUCCESS in 7m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/19a447395ac24e2883e24503157dce5e : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99572925c99f4a1280dd9a2777e7e1d0 : SUCCESS in 10m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/29231de94fa64380b184019f1c1c1248 : SUCCESS in 6m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3eece984583e43799860d78d71f9f6da : SUCCESS in 14m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c853419489bf4208a2980c3d59f7ccaf : SUCCESS in 18m 29s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56e02546b09444318f15cb4aecf7d9e7 : SUCCESS in 52m 22s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f8760a3cd8e404497bafef7221050a0 : SUCCESS in 19m 50s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b1faab02b2c4b65be33ae9389a2747e : SUCCESS in 1h 18m 14s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ea2c53b4003402c922dd5fdae034332 : SUCCESS in 19m 03s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/458c0153b28142048fec21636626cd64 : SUCCESS in 25m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b985f7482534b24bf847d528df5c1a5 : SUCCESS in 11m 19s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d80fbf010bdf44c1a6aad42017491c7d : SUCCESS in 13m 00s","accounts_in_message":[],"_revision_number":11},{"id":"edf950ebd3d05d1c5f1f7e65b14fc0d24be16612","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-09-15 11:39:52.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"d7542036ff821094622624009358d0a704e643f2","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-15 11:52:58.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"48be1544aaa6356b506b6ef10bf9b0ef84f0c6d0","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-19 07:42:53.000000000","message":"Patch Set 11: -Code-Review","accounts_in_message":[],"_revision_number":11},{"id":"b646c0140ad5e96c6a72c317249b67535f67e819","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-19 07:44:33.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"221e6064b372810b0f6ce93589cbdb261545ee2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-19 07:53:14.000000000","message":"Patch Set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"401700b5bd137260b66bc55881bbd42950ae464c","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-19 07:58:16.000000000","message":"Patch Set 12: Code-Review-1","accounts_in_message":[],"_revision_number":12},{"id":"272f844cb8d6bc0797ef5983706fa1fe373a9b76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 08:54:22.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/42e77ffea33642f2aa949cf89247d47a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1557d7f044742aaaf4083449ca1b070 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e108d32e0014780bb729f949f45d652 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3678dc21f7ad4ed19a6b2b8d84158c9d : SUCCESS in 6m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ac8b9ec553e4e33ad4b10034bfe1e83 : SUCCESS in 7m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20fefe3f76d248e5a265115c58271bba : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/580a964edea24829ac8d2c7098232fa6 : SUCCESS in 7m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dbf59134dcfd4af8ae85b44ab576dc5d : SUCCESS in 8m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cb110dc6705d468796b9233dea5a341c : SUCCESS in 15m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/87bd85d7de1f40a0945526e08670fb39 : SUCCESS in 19m 25s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71e3ac104cde43a68a6e4b453c131ab2 : SUCCESS in 58m 38s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1000e1faec874ac790f1651fa376c0d9 : SUCCESS in 16m 02s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cff731da1be14bfa90d4414208b2979c : FAILURE in 59m 47s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acdcb3ba62bc47f3b3b7a78748ef52c3 : SUCCESS in 20m 40s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ef05917c1204a74b47da1e3e3b21e96 : SUCCESS in 25m 14s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b3e961423a748a6b61d2c70bffc5412 : SUCCESS in 16m 15s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a22e9da55dfe47cebe3f192df3f19420 : SUCCESS in 13m 53s","accounts_in_message":[],"_revision_number":12},{"id":"5db18f3d445719fd0452982695fbb575b6209139","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-19 12:00:49.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"018ebe13b530aa62378e48967a428ca76367a173","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 14:56:29.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7032433b9ed4772b8369b2d8d7cfbf7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c29619748f3e40cdb179360d3c31f1aa : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9ac7128f703491fa835e52359efd811 : SUCCESS in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fc9aa5dc86c44edbc59aad05407f4a2 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0788ec0db61413b92858d2e8f94f87f : SUCCESS in 7m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e241548179404421920bc0e6b10c1218 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8933f1435734d40bb59eb1afe4d0bad : SUCCESS in 10m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d54368cab76e4444b7d2fc826943acd4 : SUCCESS in 5m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bab6f250bc5d48b9990d3b9c2047b36e : SUCCESS in 16m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/603567237a0146efa049a3c3c2eb4dea : SUCCESS in 16m 41s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d7d8f48fe544d2bbff46827d8a24525 : SUCCESS in 55m 47s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e7bf871df9242cf9817c2f2589124d4 : SUCCESS in 17m 59s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d86c9fe7fbf9482d85cac912e1ddaef6 : FAILURE in 1h 14m 43s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bed2f1c21b144296a5b9d21077852dc1 : SUCCESS in 17m 34s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62d71806ccbf49a88bfc39e1dff1c208 : SUCCESS in 31m 30s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ae4e5bc1e47430080f9a350b1fba797 : SUCCESS in 14m 18s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9bf4b668af24437fa0d0d6fefa5abb18 : SUCCESS in 17m 56s","accounts_in_message":[],"_revision_number":12},{"id":"9d5c2a75ad5cd9d058a3bdcf74f545eceb33d709","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-29 07:20:56.000000000","message":"Patch Set 12: -Code-Review","accounts_in_message":[],"_revision_number":12},{"id":"5729ec3d3b4ea20445024c87651147b2c9c675d3","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-29 07:22:55.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"35ea157e3f28e59306cce80fbb640a76fe1546d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 07:23:17.000000000","message":"Patch Set 12:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla-ansible for 857782,1","accounts_in_message":[],"_revision_number":12},{"id":"1be6274848f5b8e559dd6f50011580f2bece212c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-29 07:42:13.000000000","message":"Patch Set 13: Commit message was updated.","accounts_in_message":[],"_revision_number":13},{"id":"a994a382670c48230cd51519f7433443a8cfd0f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-29 07:42:57.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"9141f1af92995c99cce174bece025c8313b26f48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 08:50:28.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ad3c229731f4d179a419aec749c679d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05c294daf34b484792bdf8bbc53bc39b : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ba81bb359c543268661c41f8dd6a63e : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20af510772134787823172129652d191 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/340d9684d258454aa6f677875dc8b5f6 : SUCCESS in 6m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed981ae094cd4bf392e6a5fcfebb5fc4 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce0e37314c124010a150db2a98a00c82 : SUCCESS in 9m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/92e6c5eaecbe499abc37c22ac20d21ba : SUCCESS in 6m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3fc0a11f5d754bc096b26788af2c262f : SUCCESS in 14m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/032c747cfcbb4704ba8e83d477e97eee : SUCCESS in 18m 09s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1d166a5da1147fea60cfdad7113c6c7 : SUCCESS in 52m 07s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85c74a5ab82245ba9b09b9840f82cd84 : SUCCESS in 17m 59s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f82f4701a82243c09bda3514795e4b64 : FAILURE in 1h 04m 34s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80098878aad448e4847022d7f415a7cc : SUCCESS in 17m 38s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3aa384c7a764534b8e9d2076a000d74 : SUCCESS in 28m 27s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e445e9c53914d4a8953d56cc53183dd : SUCCESS in 15m 10s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6bbc1de8141d497c9f36eaab95d97316 : SUCCESS in 13m 38s","accounts_in_message":[],"_revision_number":14},{"id":"774d7eb4186d9f6351c8463efb81bec19369616a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-29 11:39:28.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"3cc0116cc9926d5275dafaf05bc93fd269bdb468","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 12:57:12.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8dc9b939d5c47cca4316e6425de3ac7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7fd85cd1b804914b8809d946a3a8601 : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/504b9ee57fe14e589c1f9a3a75d09bc3 : SUCCESS in 7m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb751f9b722e4a9ba8d4a8bb21898358 : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4db876907ea642b4b61f14655628e4a0 : SUCCESS in 8m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ea9923cd8db41d4aa29147b0d854cf5 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a967e8892044898b4c29aa3229b9cbc : SUCCESS in 11m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/393981b683f843adb19bc65a4d5ae2ee : SUCCESS in 6m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cd702503a6014f99ab9618c13e079221 : SUCCESS in 15m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0bc7f430edb947c48777bd9949c0c68d : SUCCESS in 15m 49s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec203517f169442eb6d14357d39879af : SUCCESS in 56m 02s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/587d7a1042f54ff48c522726f6ba5c7f : SUCCESS in 14m 50s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4d3f131c255465fa2d24c5eb1a0bf31 : SUCCESS in 1h 15m 41s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6cf1ec4f3bb46378a2a542b727441f6 : SUCCESS in 14m 10s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6a22f66f4994686b6cd38d3c5097e1d : SUCCESS in 25m 19s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/420b99eb03f140a1b73ca4f2083d5d25 : SUCCESS in 13m 50s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/101cc629ea3148668672bdfefc053eed : SUCCESS in 12m 36s","accounts_in_message":[],"_revision_number":15},{"id":"283a9af951058ec3a1cbd11022afd6d228511e8c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-29 13:01:43.000000000","message":"Patch Set 15: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"50ac8bcb23a7bd00a0561ca465cb9f02b48d47dc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-29 13:03:23.000000000","message":"Patch Set 15: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"c5d4d8eb0549ced05e473172846d5c09bc52801c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-09-29 13:04:21.000000000","message":"Patch Set 15:\n\n(4 comments)","accounts_in_message":[],"_revision_number":15},{"id":"cdde34b6a43ce5e5610ff672845c2968cd750521","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-29 13:31:32.000000000","message":"Patch Set 16: Commit message was updated.","accounts_in_message":[],"_revision_number":16},{"id":"d2462e0c47bd2f300d4b4ffc1731449daf775893","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-29 13:36:16.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"f2c5aace7bb25b78191df46ce96b9ddf79690523","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-29 13:36:28.000000000","message":"Patch Set 17:\n\n(4 comments)","accounts_in_message":[],"_revision_number":17},{"id":"e12acc1909098f07e064993d2e1f110260830043","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 15:02:58.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bbe14c85f9e406991d728cba94dc3c5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a778724c61a5477ea8b636047c3b9a29 : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e175681c847b4d85b53510e6828183ef : SUCCESS in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4f138c0b2094bf99736b848aa5c4582 : SUCCESS in 7m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebf91e49642241788e9909f0f3981a41 : SUCCESS in 7m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c2d742737224a629747c15707450829 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09c46fcd081c4edea578da970c0ace42 : SUCCESS in 10m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6101f0b392f749079908079f4fc9104f : SUCCESS in 6m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/25cffaffcac143cfbf4a4e563973a82e : SUCCESS in 26m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c79c805ce92e440386e2fc2383301b62 : SUCCESS in 17m 55s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e0e5870ab514007a0e2f689a89c5171 : SUCCESS in 56m 04s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c00e208683e646b2af95f7233ce6bee9 : SUCCESS in 17m 59s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49ec204eaf714bda8433456af7548a89 : SUCCESS in 1h 18m 55s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52dbd898515a426080ece1a7581d4d3b : SUCCESS in 18m 36s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42b00a4443c44696bad307cd9eacdbaf : SUCCESS in 29m 31s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/746040bc60bd49b2acc6d0f3f8040818 : SUCCESS in 13m 11s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb6bb98e4f7f498c9537225de623c50a : SUCCESS in 14m 50s","accounts_in_message":[],"_revision_number":17},{"id":"8e4983d86333dedde0b31b23c870d0b6fb2daa61","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-29 16:03:22.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"ee03dd53baca9e105cd3ac64aaf13a78d8521e10","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-09-30 08:24:37.000000000","message":"Patch Set 17: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":17},{"id":"3ec242aa1d00a01d3b8a3be8c39d6c6f3621548a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-30 12:57:51.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"90ddc93cd589f8d9c123f38f6999fb6ea3efd604","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-09-30 12:58:14.000000000","message":"Patch Set 18:\n\n(8 comments)","accounts_in_message":[],"_revision_number":18},{"id":"d83ed7d32ee545f7db4668badaad1bbed9759e49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 13:56:12.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d6306e9f4784885a7a799b53ea69417\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8de69058bb974b458c7a7510bb3d1f12 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/980a986aa69c45a09b9c00c67e4c4a7a : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f4db60d65a645bab848b24db23975a3 : SUCCESS in 6m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc55f767bfd747b2bf1dda605a999372 : SUCCESS in 6m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e7b294417584e308dc454d14cf1e778 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c29d33b06d74b388fa0db7ff7fe0ffc : SUCCESS in 10m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e2bc8257ee454eb5b017db57e994b036 : SUCCESS in 6m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/39372acc4e944d23867d2860f9043a34 : SUCCESS in 14m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/927b920057e948c5a0b9f655d4e8421b : SUCCESS in 19m 56s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc5a013095614ee18ed7171385334e98 : SUCCESS in 46m 23s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7196028b24c14741ad7a164582df3033 : SUCCESS in 18m 46s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81fdb219a6df490bbdad5600ad75ed93 : SUCCESS in 56m 42s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3cf3c286be194dba99c18dd2bf7d6898 : SUCCESS in 18m 48s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0d9382013cf4c0188684bea1bc5955f : SUCCESS in 31m 50s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9640e8b637ab4933802f96008b084853 : SUCCESS in 13m 06s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41ffac6703ae4134a4ec6784ff708b8b : SUCCESS in 11m 58s","accounts_in_message":[],"_revision_number":18},{"id":"20b902075b448b9174501e079a34237fb495935b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-10-03 08:55:05.000000000","message":"Patch Set 18: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"b11511051c79bc07335de1960beb40a2c57c96b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-10-04 07:19:12.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"60915b3e98dce999cc9961e8ce879fec6e21b887","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 08:36:20.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d612ed14e0314f98973b333166a3587a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5e920ee83d74a31be6c714fa26f9170 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1f5f86c78774fc4a1680040f4d8bde1 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/720a298b1522482197ae3ccf9bf93004 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a69bce74378242738e0bd15fb2305531 : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df96d44286ce4e2385d8998cd9f09307 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b057566480004e2d904437e221407fad : SUCCESS in 7m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/21e6889f464b46b2965a39c04153bc86 : SUCCESS in 7m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/39432a332f6246f992e00a1bbd3883a7 : SUCCESS in 14m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ec96e7509c744a309fae2e39344bb08c : SUCCESS in 17m 22s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84a05b462f1e4c4cbfe67c77f23b41a9 : SUCCESS in 43m 52s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/637a6184785f40989db44c6a41c8f309 : SUCCESS in 18m 18s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e5ad4f693df425db4a6fc8621b0c4d5 : SUCCESS in 1h 12m 21s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30d58515f6ce46b8b2b28302c73b1219 : SUCCESS in 17m 34s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c5b133a1b3f48d5a4e1f6845dd33015 : SUCCESS in 22m 16s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81e809ca3c4f4a0faea804278a18bab1 : SUCCESS in 13m 37s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8dfe638d6f494b77b21cd778303a9a1b : SUCCESS in 11m 21s","accounts_in_message":[],"_revision_number":19},{"id":"594f0a4d5e2de78d9c7896a7a64fc407df97977e","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2022-10-04 09:02:44.000000000","message":"Patch Set 19:\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"e9ee494414fa31e47ef6fcb9640b2f4c8c4b7619","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-10-04 10:02:57.000000000","message":"Patch Set 19: Code-Review+2","accounts_in_message":[],"_revision_number":19},{"id":"e3f8092ffde4674ab9a0c5283578af2c7daec5a5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-04 10:03:37.000000000","message":"Patch Set 19: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":19},{"id":"2c56298ecdeceb5df8db7322758d3129fee819fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 10:04:03.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"b8493d6bc7dbe52cb993e0f62bf081c5a38562c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 11:24:59.000000000","message":"Patch Set 19: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73af043f4ec1450c8e800223727a0516\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab12881823614a54a9432128318abb3b : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73cdd2ea4ebe4931b21f37de7a119553 : SUCCESS in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5fb95ead81734e08bb46fe1916bdd5bf : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b44498e141794c549eac1386e6008a04 : SUCCESS in 10m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9a767c1c30d94437a34fed290e717e99 : SUCCESS in 6m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ce89b06e4a9e4797acb66b7090aa4d47 : SUCCESS in 14m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/822778c0611e42c191ecaf7a73258aaf : SUCCESS in 18m 09s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aeb636299f5942af920a6ac4db117ee5 : SUCCESS in 57m 29s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5702dd2b6e1c4c21835cd9fd66e31ca5 : SUCCESS in 18m 02s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5063dceecbf7405a9c6ab35a7c4b1f91 : SUCCESS in 1h 19m 58s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad3fd1ab556f40cb887a4694fc52501a : SUCCESS in 19m 07s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9281cdf157146c8a509342d3beaabdd : SUCCESS in 32m 55s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5709c46291a749c29e678cf33a740c86 : SUCCESS in 14m 11s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2fba709f3f924d87a5297c8cbb175906 : SUCCESS in 16m 08s","accounts_in_message":[],"_revision_number":19},{"id":"f57bb7013a60a5583f644f72f1b60b8905377be5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 11:24:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":19},{"id":"a780fce23c8c55a27f997b405d43514e1bb2d6ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 11:26:09.000000000","message":"Patch Set 19:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15877f6cc3c2424e9e3e47944bc2fb3d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d82888c4e55a4b73913eefc861b7b0a5 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"20e2ac9cbd666ea1e856c24d3ccd6f44e99d2df1","revisions":{"c201082855d5998ade4946a8d1880fa20b32d2ee":{"kind":"REWORK","_number":1,"created":"2022-09-14 09:03:13.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/94/857594/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/857594/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/857594/1"}}},"commit":{"parents":[{"commit":"9cbd1e2f52b84a418ba1b3ec5c1119713774c926","subject":"CI: Drop centos8s and rocky8 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9cbd1e2f52b84a418ba1b3ec5c1119713774c926"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 09:01:53.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 09:01:53.000000000","tz":120},"subject":"updated  write_bifrost_clouds_yaml logic","message":"updated  write_bifrost_clouds_yaml logic\n\nin order to facilitate running ironic checks\nbefore openstack is deployed\n\nChange-Id: Ibc4c9c93f52414c23cd7d8b4f1b32c85cb233681\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c201082855d5998ade4946a8d1880fa20b32d2ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c201082855d5998ade4946a8d1880fa20b32d2ee"}]},"branch":"refs/heads/master"},"86c0ea137a838d3efcb9c31a63871a6961760525":{"kind":"REWORK","_number":2,"created":"2022-09-14 09:07:32.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/94/857594/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/857594/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/857594/2"}}},"commit":{"parents":[{"commit":"9cbd1e2f52b84a418ba1b3ec5c1119713774c926","subject":"CI: Drop centos8s and rocky8 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9cbd1e2f52b84a418ba1b3ec5c1119713774c926"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 09:01:53.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 09:07:13.000000000","tz":120},"subject":"updated  write_bifrost_clouds_yaml logic","message":"updated  write_bifrost_clouds_yaml logic\n\nin order to facilitate running ironic checks\nbefore openstack is deployed\n\nChange-Id: Ibc4c9c93f52414c23cd7d8b4f1b32c85cb233681\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86c0ea137a838d3efcb9c31a63871a6961760525"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86c0ea137a838d3efcb9c31a63871a6961760525"}]},"branch":"refs/heads/master"},"5302eb16549c68e67e4dbc71d64c0af409271897":{"kind":"REWORK","_number":3,"created":"2022-09-14 10:04:40.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/94/857594/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/857594/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/857594/3"}}},"commit":{"parents":[{"commit":"9cbd1e2f52b84a418ba1b3ec5c1119713774c926","subject":"CI: Drop centos8s and rocky8 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9cbd1e2f52b84a418ba1b3ec5c1119713774c926"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 09:01:53.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 10:04:28.000000000","tz":120},"subject":"updated  write_bifrost_clouds_yaml logic","message":"updated  write_bifrost_clouds_yaml logic\n\nin order to facilitate running ironic checks\nbefore openstack is deployed\n\nChange-Id: Ibc4c9c93f52414c23cd7d8b4f1b32c85cb233681\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5302eb16549c68e67e4dbc71d64c0af409271897"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5302eb16549c68e67e4dbc71d64c0af409271897"}]},"branch":"refs/heads/master"},"07e66ec523d2d58b3594e559bf229134d93ab617":{"kind":"REWORK","_number":4,"created":"2022-09-14 13:17:02.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/94/857594/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/857594/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/857594/4"}}},"commit":{"parents":[{"commit":"9cbd1e2f52b84a418ba1b3ec5c1119713774c926","subject":"CI: Drop centos8s and rocky8 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9cbd1e2f52b84a418ba1b3ec5c1119713774c926"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 09:01:53.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 13:16:50.000000000","tz":120},"subject":"updated  write_bifrost_clouds_yaml logic","message":"updated  write_bifrost_clouds_yaml logic\n\nin order to facilitate running ironic checks\nbefore openstack is deployed\n\nChange-Id: Ibc4c9c93f52414c23cd7d8b4f1b32c85cb233681\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/07e66ec523d2d58b3594e559bf229134d93ab617"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/07e66ec523d2d58b3594e559bf229134d93ab617"}]},"branch":"refs/heads/master"},"017f94ba4949a5a727ae71d965163607d7df11ec":{"kind":"REWORK","_number":5,"created":"2022-09-14 13:28:31.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/94/857594/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/857594/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/857594/5"}}},"commit":{"parents":[{"commit":"9cbd1e2f52b84a418ba1b3ec5c1119713774c926","subject":"CI: Drop centos8s and rocky8 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9cbd1e2f52b84a418ba1b3ec5c1119713774c926"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 09:01:53.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 13:28:21.000000000","tz":120},"subject":"updated  write_bifrost_clouds_yaml logic","message":"updated  write_bifrost_clouds_yaml logic\n\nin order to facilitate running ironic checks\nbefore openstack is deployed\n\nChange-Id: Ibc4c9c93f52414c23cd7d8b4f1b32c85cb233681\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/017f94ba4949a5a727ae71d965163607d7df11ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/017f94ba4949a5a727ae71d965163607d7df11ec"}]},"branch":"refs/heads/master"},"c888e3752e301190fcbf86e52d112566d362424b":{"kind":"NO_CHANGE","_number":6,"created":"2022-09-14 14:00:06.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/94/857594/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/857594/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/857594/6"}}},"commit":{"parents":[{"commit":"9cbd1e2f52b84a418ba1b3ec5c1119713774c926","subject":"CI: Drop centos8s and rocky8 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9cbd1e2f52b84a418ba1b3ec5c1119713774c926"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 09:01:53.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 13:59:57.000000000","tz":120},"subject":"updated  write_bifrost_clouds_yaml logic","message":"updated  write_bifrost_clouds_yaml logic\n\nin order to facilitate running ironic checks\nbefore openstack is deployed\n\nChange-Id: Ibc4c9c93f52414c23cd7d8b4f1b32c85cb233681\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c888e3752e301190fcbf86e52d112566d362424b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c888e3752e301190fcbf86e52d112566d362424b"}]},"branch":"refs/heads/master"},"c03e9719878f3226dcba63d23d007264366185e9":{"kind":"REWORK","_number":7,"created":"2022-09-14 14:01:08.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/94/857594/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/857594/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/857594/7"}}},"commit":{"parents":[{"commit":"9cbd1e2f52b84a418ba1b3ec5c1119713774c926","subject":"CI: Drop centos8s and rocky8 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9cbd1e2f52b84a418ba1b3ec5c1119713774c926"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 09:01:53.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 14:00:59.000000000","tz":120},"subject":"updated  write_bifrost_clouds_yaml logic","message":"updated  write_bifrost_clouds_yaml logic\n\nin order to facilitate running ironic checks\nbefore openstack is deployed\n\nChange-Id: Ibc4c9c93f52414c23cd7d8b4f1b32c85cb233681\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c03e9719878f3226dcba63d23d007264366185e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c03e9719878f3226dcba63d23d007264366185e9"}]},"branch":"refs/heads/master"},"1b93d95b8b496271cab0d74b526f3700d3f7903f":{"kind":"REWORK","_number":8,"created":"2022-09-14 14:18:41.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/94/857594/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/857594/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/857594/8"}}},"commit":{"parents":[{"commit":"9cbd1e2f52b84a418ba1b3ec5c1119713774c926","subject":"CI: Drop centos8s and rocky8 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9cbd1e2f52b84a418ba1b3ec5c1119713774c926"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 09:01:53.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 14:18:32.000000000","tz":120},"subject":"updated  write_bifrost_clouds_yaml logic","message":"updated  write_bifrost_clouds_yaml logic\n\nin order to facilitate running ironic checks\nbefore openstack is deployed\n\nChange-Id: Ibc4c9c93f52414c23cd7d8b4f1b32c85cb233681\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b93d95b8b496271cab0d74b526f3700d3f7903f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b93d95b8b496271cab0d74b526f3700d3f7903f"}]},"branch":"refs/heads/master"},"926adeb0719497cb103c2f5d762785f62b7763fc":{"kind":"REWORK","_number":9,"created":"2022-09-15 07:45:56.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/94/857594/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/857594/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/857594/9"}}},"commit":{"parents":[{"commit":"9cbd1e2f52b84a418ba1b3ec5c1119713774c926","subject":"CI: Drop centos8s and rocky8 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9cbd1e2f52b84a418ba1b3ec5c1119713774c926"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 09:01:53.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-15 07:43:01.000000000","tz":120},"subject":"updated  write_bifrost_clouds_yaml logic","message":"updated  write_bifrost_clouds_yaml logic\n\nin order to facilitate running ironic checks\nbefore openstack is deployed\n\nChange-Id: Ibc4c9c93f52414c23cd7d8b4f1b32c85cb233681\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/926adeb0719497cb103c2f5d762785f62b7763fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/926adeb0719497cb103c2f5d762785f62b7763fc"}]},"branch":"refs/heads/master"},"67938b4827715fe450e741a22e58247f6e6c433c":{"kind":"REWORK","_number":10,"created":"2022-09-15 07:50:40.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/94/857594/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/857594/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/857594/10"}}},"commit":{"parents":[{"commit":"9cbd1e2f52b84a418ba1b3ec5c1119713774c926","subject":"CI: Drop centos8s and rocky8 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9cbd1e2f52b84a418ba1b3ec5c1119713774c926"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 09:01:53.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-15 07:50:32.000000000","tz":120},"subject":"updated  write_bifrost_clouds_yaml logic","message":"updated  write_bifrost_clouds_yaml logic\n\nin order to facilitate running ironic checks\nbefore openstack is deployed\n\nChange-Id: Ibc4c9c93f52414c23cd7d8b4f1b32c85cb233681\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67938b4827715fe450e741a22e58247f6e6c433c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67938b4827715fe450e741a22e58247f6e6c433c"}]},"branch":"refs/heads/master"},"fb0faef5cc59a070437dfb668e69b7d6fff97bd5":{"kind":"REWORK","_number":11,"created":"2022-09-15 08:26:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/94/857594/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/857594/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/857594/11"}}},"commit":{"parents":[{"commit":"9cbd1e2f52b84a418ba1b3ec5c1119713774c926","subject":"CI: Drop centos8s and rocky8 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9cbd1e2f52b84a418ba1b3ec5c1119713774c926"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 09:01:53.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-09-15 08:26:49.000000000","tz":0},"subject":"updated  write_bifrost_clouds_yaml logic","message":"updated  write_bifrost_clouds_yaml logic\n\nin order to facilitate running ironic checks\nbefore openstack is deployed\n\nChange-Id: Ibc4c9c93f52414c23cd7d8b4f1b32c85cb233681\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb0faef5cc59a070437dfb668e69b7d6fff97bd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb0faef5cc59a070437dfb668e69b7d6fff97bd5"}]},"branch":"refs/heads/master"},"c597316c845177a29c149896f2b6e98b38d99a79":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2022-09-19 07:53:14.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/94/857594/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/857594/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/857594/12"}}},"commit":{"parents":[{"commit":"9cbd1e2f52b84a418ba1b3ec5c1119713774c926","subject":"CI: Drop centos8s and rocky8 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9cbd1e2f52b84a418ba1b3ec5c1119713774c926"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 09:01:53.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-19 07:53:14.000000000","tz":0},"subject":"updated  write_bifrost_clouds_yaml logic","message":"updated  write_bifrost_clouds_yaml logic\n\nin order to facilitate running ironic checks\nbefore openstack is deployed\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/857782\n\nChange-Id: Ibc4c9c93f52414c23cd7d8b4f1b32c85cb233681\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c597316c845177a29c149896f2b6e98b38d99a79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c597316c845177a29c149896f2b6e98b38d99a79"}]},"branch":"refs/heads/master","description":"Edit commit message"},"4e85cda81b5b6cd08a156465e26a69e3dce8e7e3":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2022-09-29 07:42:13.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/94/857594/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/857594/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/857594/13"}}},"commit":{"parents":[{"commit":"9cbd1e2f52b84a418ba1b3ec5c1119713774c926","subject":"CI: Drop centos8s and rocky8 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9cbd1e2f52b84a418ba1b3ec5c1119713774c926"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 09:01:53.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-29 07:42:13.000000000","tz":0},"subject":"updated  write_bifrost_clouds_yaml logic","message":"updated  write_bifrost_clouds_yaml logic\n\nin order to facilitate running ironic checks\nbefore openstack is deployed\n\nChange-Id: Ibc4c9c93f52414c23cd7d8b4f1b32c85cb233681\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e85cda81b5b6cd08a156465e26a69e3dce8e7e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e85cda81b5b6cd08a156465e26a69e3dce8e7e3"}]},"branch":"refs/heads/master","description":"Edit commit message"},"bf045bbb24a6e7684448f015328155ac6c66096d":{"kind":"REWORK","_number":14,"created":"2022-09-29 07:42:57.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/94/857594/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/857594/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/857594/14"}}},"commit":{"parents":[{"commit":"608951fcce38428b155030badcff15a5a6a0445c","subject":"Merge \"Add new DIB upper constraints variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/608951fcce38428b155030badcff15a5a6a0445c"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 09:01:53.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-29 07:41:11.000000000","tz":120},"subject":"updated  write_bifrost_clouds_yaml logic","message":"updated  write_bifrost_clouds_yaml logic\n\nin order to facilitate running ironic checks\nbefore openstack is deployed\n\nChange-Id: Ibc4c9c93f52414c23cd7d8b4f1b32c85cb233681\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf045bbb24a6e7684448f015328155ac6c66096d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf045bbb24a6e7684448f015328155ac6c66096d"}]},"branch":"refs/heads/master"},"c32eb09bc9381884063e6df6b44acd28c9e125c8":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2022-09-29 11:39:28.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/94/857594/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/857594/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/857594/15"}}},"commit":{"parents":[{"commit":"05a2c1acbd2dd4dd187d947f0fc6904dd71e93e2","subject":"CI: Set previous_release to Yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05a2c1acbd2dd4dd187d947f0fc6904dd71e93e2"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 09:01:53.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-29 11:39:28.000000000","tz":0},"subject":"updated  write_bifrost_clouds_yaml logic","message":"updated  write_bifrost_clouds_yaml logic\n\nin order to facilitate running ironic checks\nbefore openstack is deployed\n\nChange-Id: Ibc4c9c93f52414c23cd7d8b4f1b32c85cb233681\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c32eb09bc9381884063e6df6b44acd28c9e125c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c32eb09bc9381884063e6df6b44acd28c9e125c8"}]},"branch":"refs/heads/master","description":"Rebase"},"dc3601ee42cc57385bbaf175e81c3b3043acb403":{"kind":"NO_CODE_CHANGE","_number":16,"created":"2022-09-29 13:31:32.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/94/857594/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/857594/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/857594/16"}}},"commit":{"parents":[{"commit":"05a2c1acbd2dd4dd187d947f0fc6904dd71e93e2","subject":"CI: Set previous_release to Yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05a2c1acbd2dd4dd187d947f0fc6904dd71e93e2"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 09:01:53.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-29 13:31:32.000000000","tz":0},"subject":"Update write_bifrost_clouds_yaml logic","message":"Update write_bifrost_clouds_yaml logic\n\nThis is changed in order to facilitate running Ironic checks before\nOpenStack is deployed.\n\nChange-Id: Ibc4c9c93f52414c23cd7d8b4f1b32c85cb233681\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dc3601ee42cc57385bbaf175e81c3b3043acb403"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dc3601ee42cc57385bbaf175e81c3b3043acb403"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f5e00098d0a61c05a89056936f77378e8e906aac":{"kind":"REWORK","_number":17,"created":"2022-09-29 13:36:16.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/94/857594/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/857594/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/857594/17"}}},"commit":{"parents":[{"commit":"05a2c1acbd2dd4dd187d947f0fc6904dd71e93e2","subject":"CI: Set previous_release to Yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05a2c1acbd2dd4dd187d947f0fc6904dd71e93e2"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 09:01:53.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-29 13:33:18.000000000","tz":0},"subject":"Update write_bifrost_clouds_yaml logic","message":"Update write_bifrost_clouds_yaml logic\n\nThis is changed in order to facilitate running Ironic checks before\nOpenStack is deployed.\n\nChange-Id: Ibc4c9c93f52414c23cd7d8b4f1b32c85cb233681\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5e00098d0a61c05a89056936f77378e8e906aac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5e00098d0a61c05a89056936f77378e8e906aac"}]},"branch":"refs/heads/master"},"0a46359431bf096c267f2889cb997cb5eefa29f3":{"kind":"REWORK","_number":18,"created":"2022-09-30 12:57:51.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/94/857594/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/857594/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/857594/18"}}},"commit":{"parents":[{"commit":"05a2c1acbd2dd4dd187d947f0fc6904dd71e93e2","subject":"CI: Set previous_release to Yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05a2c1acbd2dd4dd187d947f0fc6904dd71e93e2"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 09:01:53.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-30 12:57:49.000000000","tz":120},"subject":"Update write_bifrost_clouds_yaml logic","message":"Update write_bifrost_clouds_yaml logic\n\nThis is changed in order to facilitate running Ironic checks before\nOpenStack is deployed.\n\nChange-Id: Ibc4c9c93f52414c23cd7d8b4f1b32c85cb233681\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a46359431bf096c267f2889cb997cb5eefa29f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a46359431bf096c267f2889cb997cb5eefa29f3"}]},"branch":"refs/heads/master"},"20e2ac9cbd666ea1e856c24d3ccd6f44e99d2df1":{"kind":"REWORK","_number":19,"created":"2022-10-04 07:19:12.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/94/857594/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/857594/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/857594/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/857594/19"}}},"commit":{"parents":[{"commit":"05a2c1acbd2dd4dd187d947f0fc6904dd71e93e2","subject":"CI: Set previous_release to Yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05a2c1acbd2dd4dd187d947f0fc6904dd71e93e2"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-09-14 09:01:53.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2022-10-04 07:18:38.000000000","tz":120},"subject":"Update write_bifrost_clouds_yaml logic","message":"Update write_bifrost_clouds_yaml logic\n\nThis is changed in order to facilitate running Ironic checks before\nOpenStack is deployed.\n\nChange-Id: Ibc4c9c93f52414c23cd7d8b4f1b32c85cb233681\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/20e2ac9cbd666ea1e856c24d3ccd6f44e99d2df1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/20e2ac9cbd666ea1e856c24d3ccd6f44e99d2df1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
