)]}'
{"id":"openstack%2Fkayobe~858159","triplet_id":"openstack%2Fkayobe~master~I4f2176aa056fec62e31d07140e3d05779480a93d","project":"openstack/kayobe","branch":"master","topic":"skip-hooks","attention_set":{},"removed_from_attention_set":{"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2022-10-12 09:13:46.000000000","reason":"Change was submitted"},"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2022-10-05 13:27:45.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-10-12 09:13:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4f2176aa056fec62e31d07140e3d05779480a93d","subject":"added option to skip hooks","status":"MERGED","created":"2022-09-16 15:13:06.000000000","updated":"2022-10-12 09:14:42.000000000","submitted":"2022-10-12 09:13:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858159-skip-hooks","meta_rev_id":"cec246707e8e1ed34748d5fc6aae671171a903ac","_number":858159,"virtual_id_number":858159,"owner":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-12 09:13:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2022-10-05 13:27:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2022-10-12 06:54:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2022-10-12 06:54:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-16 16:30:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-20 10:00:32.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-10-05 12:52:56.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2022-10-05 13:27:45.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2022-10-12 06:54:13.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"30601130db4f3f565c8f01a08d8e324db156eb7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2022-09-16 15:13:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b597c745dd780aa90e0dd1911d265b63c17dda62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 16:30:28.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cdc7d9df0ba54f07b4489cc10e287efd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a4c24f1383d4380940dd00882ea8b71 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6090a47d4604794abd12cabf747c667 : FAILURE in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a7f5881c5c64f8998cc0e0d52fea2b4 : SUCCESS in 6m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f156222f6e54bd6affd14d8ad9444c0 : SUCCESS in 6m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/787631f0c1004dce8c45e270ce03e298 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a04a96a4c34423dbd3be9f8fde4252f : SUCCESS in 10m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7debb83c385f47c5994af736a54eb548 : SUCCESS in 6m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6a8f8980106745b8ad0d96fc06e32f2d : SUCCESS in 14m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/76ed993bbd0043ee98435ae452f8d307 : SUCCESS in 18m 16s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eac7d8b061924e03a956cc7ef18efbf0 : SUCCESS in 55m 23s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b68fd6c289b4c29990ce03d11d6c4e6 : SUCCESS in 18m 53s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfa7400c5225434eb2a52afaca51a030 : FAILURE in 1h 14m 19s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34c50f853c20430e97ff18a62caf08eb : SUCCESS in 19m 03s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4fbea2fa0341452381c44ed8c556298a : SUCCESS in 34m 17s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4756dfa456e2498385bc685bbc73e1c3 : SUCCESS in 16m 15s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72be191f01154f56a80d7e0f485c6a8d : SUCCESS in 15m 23s","accounts_in_message":[],"_revision_number":1},{"id":"9ebd581ec87d8fd4227a74c49bb2eb7d4c98e571","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-09-20 10:00:32.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f64bafc67473deab92f52defedcbb458afcedc96","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-09-20 10:00:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"324a5fec55db25ee3c562bcd1d1a16580ebb621d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-09-20 10:02:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"55d7eb192998da7209ef96b2ee256431904c2bf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2022-09-22 14:25:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8ee85a63cb015798f48da07b60fec64f77041a02","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2022-09-22 15:17:49.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"31f1ddbbcaaa0e0c8b376a7be58ec656ff48fa41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 15:45:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4d8c1c55e804a05ad4318dade0fc794\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc581df213c74c3c90eb6f74972e867e : SUCCESS in 8m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b803978526642e8923a404d9410cf58 : FAILURE in 8m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88bad42aba984c278e840cb99ee683e3 : SUCCESS in 9m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b52943466a447fe8dc9a92c1b564697 : SUCCESS in 9m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9153792cb0904d44b967aa5b8e7902f4 : SUCCESS in 9m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1890b162c9ab402bae00286a637ae599 : SUCCESS in 12m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4227f64c94d946e78fdb5a21262a2d82 : SUCCESS in 6m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8ed8624b38484499808f920cdb737b2e : SUCCESS in 8m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/76f3b92848934c669c09cfd1399df6dd : SUCCESS in 16m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6c3bbac41a3b492398d437c9c86513f8 : SUCCESS in 18m 55s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38a591c5da4842f3afcf020fb54ba14b : SUCCESS in 1h 03m 53s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/abd262bbe8b04e27a9ef3ea5404eb0aa : SUCCESS in 19m 31s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22d6b054eef0409faf4402d33f724259 : FAILURE in 1h 18m 47s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6696097b1ecf48bd9f4af8fe89d902b9 : SUCCESS in 19m 22s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d4250aaa772411a9a195ed1e1348525 : SUCCESS in 34m 07s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d181b316dd494257a02678a68736183e : SUCCESS in 15m 43s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67f16dbdfa7a4d888361047020243601 : SUCCESS in 15m 55s","accounts_in_message":[],"_revision_number":2},{"id":"786a64898489e63655709f37ed1c986f734fd149","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2022-09-30 08:49:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b44a259585f9bc66df612d44de490afcea260e83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 10:05:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53ae3e26f1894e7ebf35622b033d1238\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95b0ab8cb9964b92bd1682128c48c949 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/932de03e9fd5481cbb84b7c736cfa635 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97e11de7622c46f1b56c3a7e39570ab3 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6af7376d1e254fb1b9d98ee29b773538 : SUCCESS in 6m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6dd052278ff446a081b980c62e8a3916 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00ab5f73f1164f7ea8947f6ee1db79ca : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b38bb65a76f84901a6a1a3317c566be1 : SUCCESS in 4m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5bd665ca65914e30951bedf92d5eeeb0 : SUCCESS in 6m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2272504fc52b438c8aa0dbf98a753215 : SUCCESS in 15m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2db9c6c2762a4546a4d2359d9efd9d6b : SUCCESS in 17m 57s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d78cf7513ea74b4b82c8cb523211917c : SUCCESS in 44m 33s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa073cd4b9c44e95bf7597fceb482eff : SUCCESS in 17m 00s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec27f7fd6608406496c88a513ad4fb44 : SUCCESS in 1h 14m 33s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/149cf26aa1bb48268962a4aad44aed56 : SUCCESS in 17m 38s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab111d1ea56840ca970ac875a70fa1e0 : SUCCESS in 26m 14s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d75846c729104bc298aa1844de53cf96 : SUCCESS in 13m 36s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4367dcf292f44927939d09d054d577b8 : SUCCESS in 13m 50s","accounts_in_message":[],"_revision_number":3},{"id":"107d520de6bcdd4f49307758c84ffaae851834aa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-10-03 14:35:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"eaeb874a9943403a78536eca197451ffc7b09828","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2022-10-04 11:42:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"39b3ded7a8857c01b3381140faa50ded4a55705b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 12:59:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c55b281f4fc642b99328d716f30225df\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccd2e8d5c10d4cfead0d8779bc569958 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fafc5e31a4e44a7e9bff0e39745a9d1d : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7db02d63c944952b0b726363a5acb8f : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e94d989bb37c40868ae7ef3ad55ba912 : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0e3d606507649fc99b0b05ed890e6e8 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67afc15c8b3f4a6b8475df0487b99149 : SUCCESS in 8m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2688ebd96d9949b39fee05bd4a75a189 : SUCCESS in 4m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1cba1e26bb674e4aaa61c00d6d73216c : SUCCESS in 6m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/69dbba72601c46aea67a539bdb5d46d7 : SUCCESS in 12m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ec32c14497704241860955b61d23f9f2 : SUCCESS in 18m 25s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b7debb30af74274974577e5b142becc : SUCCESS in 53m 51s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6846fadc3eb748619507ac0f7c6a88c6 : SUCCESS in 17m 32s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ed7761e6d924c66bf9bbd06914b6ea7 : SUCCESS in 1h 15m 43s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38b624e97246446690462b13cafd9a92 : SUCCESS in 14m 41s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f4b266c2b1c4a6ea51537f2094dc9a4 : SUCCESS in 28m 27s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a95bc337884f4750b2abeea3a6f14a90 : SUCCESS in 11m 23s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ec28813865c49b28b22861c02a5e2a9 : SUCCESS in 13m 57s","accounts_in_message":[],"_revision_number":4},{"id":"04b3dc55155ce4616612d00f229e5bfcc96a63be","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-10-04 13:17:00.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b3dcd4fdb5e4dc890db1709ca162a1a7f2b9fecd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2022-10-04 13:40:41.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"aace77a303568f1b47445dac5c28a55f5ab8b335","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 14:52:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b5ca21a6754457b98e9ec759eec6088\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/960a47129efc4272ac84b0d5e44b4080 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8a75b5a2f9b4f84a2d77e01aca90f3d : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/584416c80f064a5b95608eb80aaac818 : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3fcc02074454133b16beb1c850c0a8e : SUCCESS in 6m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b4a66aa81ffc4ba2b8f08bf71bb3ad5f : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d95a8fb4cadc4cfbb99a71d98129b719 : SUCCESS in 10m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2546c85642e64d5c815513d0da94ce00 : SUCCESS in 4m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b425fc7d580b4042882b2aa01b877de7 : SUCCESS in 7m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/015e661218ac48c39a0182bdfb093c3f : SUCCESS in 14m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fa5be1af9b61413aa0e2e5db91dd6578 : SUCCESS in 16m 35s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0f5efa3c8874a45be45b89002baff27 : SUCCESS in 57m 48s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e04cfe48542742b8b00d6dcf00e3f38c : SUCCESS in 17m 18s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3889dca6c8c147ca891aa5bd33f49ee6 : SUCCESS in 1h 10m 12s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36444d6d799446548ff94234eb008ea9 : SUCCESS in 17m 50s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d850beafffc48589449b8c78a79b410 : SUCCESS in 32m 15s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1efb201c2899423b801e77ffdcc4593b : SUCCESS in 14m 11s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8065faaf80264776a824207839d3c723 : SUCCESS in 15m 25s","accounts_in_message":[],"_revision_number":5},{"id":"d4c0e258f0b245b71595fc58623707c7eeae7038","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2022-10-05 09:01:27.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"61ae4ed56ea8104337041c15643f89ac824b2bc9","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-10-05 12:52:56.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7b9b83e4251df7f5a41b9a4fb863f5e96f15ca47","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2022-10-05 13:04:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bac00079be487534c4d532b94700e2306371547c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-10-05 13:27:45.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bdd5ce8a3a1ae291034f84d5a4a8b0242494be84","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-12 06:54:13.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"a6a3f5bd672f7ec389a35877fd74cc3ec76d785f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 06:54:44.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"466c2eefede9620ef76380ee005a0b2e054fca0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 09:13:46.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9f3d422975f44cab983b11d0aea3cdc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1027d29812348488053fb3f97528321 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bbe754243994e1c8e3ca1e8a7ea3aab : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9bd239f9b7b4abaae017de6bb51f502 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10d3ec7707ae4c02a416cb1b341dcf48 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/645d7ee19eec4b7c90beea0648a6da1d : SUCCESS in 3m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c880105072b64f68a3f45f05a27d7652 : SUCCESS in 4m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/eca3ec03dbca4c3c8bf4ceda2730299d : SUCCESS in 12m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b4251bbee77f4ed59c29ea3e926625cc : SUCCESS in 13m 45s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2df2afdd75bb4faebf49b5f3de599970 : SUCCESS in 40m 07s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/665ed64fa90c400fa3fdd63290150e31 : SUCCESS in 14m 27s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59df26c7c5504432a789944b558e2144 : SUCCESS in 1h 10m 23s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/616792a58d4f4d0ca2487c49ae60f3ab : SUCCESS in 15m 36s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c28e650c07864e89be8070620b74ebe3 : SUCCESS in 30m 16s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f4bf6f429b647ffa72d9c5735655028 : SUCCESS in 10m 30s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c329c4e406934534bfc7eda14c666d61 : SUCCESS in 13m 39s","accounts_in_message":[],"_revision_number":5},{"id":"0b42e8683625e65be7e4a0938d3fe7bba64cbd55","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 09:13:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"cec246707e8e1ed34748d5fc6aae671171a903ac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 09:14:42.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/789523a1acc847ec9eb87426b399445e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba2a30a055dd43bb89175056ede25886 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4f1e6908d1b4cb78be30d8bec7b24ff : SUCCESS in 36s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d7069283fd89e6164414d83afa688c3d2afae48f","revisions":{"d7bd4b0329d69b076528e234ae5a4d0e84fba0ae":{"kind":"REWORK","_number":1,"created":"2022-09-16 15:13:06.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/59/858159/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/858159/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/858159/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/858159/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/858159/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/858159/1"}}},"commit":{"parents":[{"commit":"608951fcce38428b155030badcff15a5a6a0445c","subject":"Merge \"Add new DIB upper constraints variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/608951fcce38428b155030badcff15a5a6a0445c"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2022-09-16 12:39:30.000000000","tz":60},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2022-09-16 15:12:24.000000000","tz":60},"subject":"added --skip-hooks","message":"added --skip-hooks\n\nChange-Id: I4f2176aa056fec62e31d07140e3d05779480a93d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7bd4b0329d69b076528e234ae5a4d0e84fba0ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7bd4b0329d69b076528e234ae5a4d0e84fba0ae"}]},"branch":"refs/heads/master"},"619c28b534767fc6aa7cb07efcfed2b739c7eb9b":{"kind":"REWORK","_number":2,"created":"2022-09-22 14:25:11.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/59/858159/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/858159/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/858159/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/858159/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/858159/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/858159/2"}}},"commit":{"parents":[{"commit":"608951fcce38428b155030badcff15a5a6a0445c","subject":"Merge \"Add new DIB upper constraints variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/608951fcce38428b155030badcff15a5a6a0445c"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2022-09-16 12:39:30.000000000","tz":60},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2022-09-22 14:20:48.000000000","tz":60},"subject":"added option to skip hooks","message":"added option to skip hooks\n\nadds the argument --skip-hooks/-sh which will stop the execution of\nhooked ansible playbooks. Either a pattern can be specified to match\nagainst or hook execution can be stopped altogether with \"all\"\n\nStory: 2009241\nTask: 43390\n\nChange-Id: I4f2176aa056fec62e31d07140e3d05779480a93d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/619c28b534767fc6aa7cb07efcfed2b739c7eb9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/619c28b534767fc6aa7cb07efcfed2b739c7eb9b"}]},"branch":"refs/heads/master"},"484559f55a578c07c882e2449fdf1a0b2e6eb93d":{"kind":"REWORK","_number":3,"created":"2022-09-30 08:49:18.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/59/858159/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/858159/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/858159/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/858159/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/858159/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/858159/3"}}},"commit":{"parents":[{"commit":"608951fcce38428b155030badcff15a5a6a0445c","subject":"Merge \"Add new DIB upper constraints variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/608951fcce38428b155030badcff15a5a6a0445c"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2022-09-16 12:39:30.000000000","tz":60},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2022-09-30 08:49:06.000000000","tz":60},"subject":"added option to skip hooks","message":"added option to skip hooks\n\nadds the argument --skip-hooks/-sh which will stop the execution of\nhooked ansible playbooks. Either a pattern can be specified to match\nagainst or hook execution can be stopped altogether with \"all\"\n\nStory: 2009241\nTask: 43390\n\nChange-Id: I4f2176aa056fec62e31d07140e3d05779480a93d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/484559f55a578c07c882e2449fdf1a0b2e6eb93d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/484559f55a578c07c882e2449fdf1a0b2e6eb93d"}]},"branch":"refs/heads/master"},"a1b47f111a245d90fa3f51ed0cde8e878e06f30d":{"kind":"REWORK","_number":4,"created":"2022-10-04 11:42:26.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/59/858159/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/858159/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/858159/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/858159/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/858159/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/858159/4"}}},"commit":{"parents":[{"commit":"608951fcce38428b155030badcff15a5a6a0445c","subject":"Merge \"Add new DIB upper constraints variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/608951fcce38428b155030badcff15a5a6a0445c"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2022-09-16 12:39:30.000000000","tz":60},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2022-10-04 11:41:41.000000000","tz":60},"subject":"added option to skip hooks","message":"added option to skip hooks\n\nadds the argument --skip-hooks/-sh which will stop the execution of\nhooked ansible playbooks. Either a pattern can be specified to match\nagainst or hook execution can be stopped altogether with \"all\"\n\nStory: 2009241\nTask: 43390\n\nChange-Id: I4f2176aa056fec62e31d07140e3d05779480a93d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a1b47f111a245d90fa3f51ed0cde8e878e06f30d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a1b47f111a245d90fa3f51ed0cde8e878e06f30d"}]},"branch":"refs/heads/master"},"d7069283fd89e6164414d83afa688c3d2afae48f":{"kind":"REWORK","_number":5,"created":"2022-10-04 13:40:41.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/59/858159/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/858159/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/858159/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/858159/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/858159/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/858159/5"}}},"commit":{"parents":[{"commit":"608951fcce38428b155030badcff15a5a6a0445c","subject":"Merge \"Add new DIB upper constraints variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/608951fcce38428b155030badcff15a5a6a0445c"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2022-09-16 12:39:30.000000000","tz":60},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2022-10-04 13:40:31.000000000","tz":60},"subject":"added option to skip hooks","message":"added option to skip hooks\n\nadds the argument --skip-hooks/-sh which will stop the execution of\nhooked ansible playbooks. Either a pattern can be specified to match\nagainst or hook execution can be stopped altogether with \"all\"\n\nStory: 2009241\nTask: 43390\n\nChange-Id: I4f2176aa056fec62e31d07140e3d05779480a93d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7069283fd89e6164414d83afa688c3d2afae48f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7069283fd89e6164414d83afa688c3d2afae48f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
