)]}'
{"id":"openstack%2Fkayobe~859784","triplet_id":"openstack%2Fkayobe~master~I295e8f5f1cc9b7af1cd45ac788db473510220170","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-10-04 11:35:54.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I295e8f5f1cc9b7af1cd45ac788db473510220170","subject":"CI: Set previous_release to Yoga","status":"MERGED","created":"2022-09-29 07:44:07.000000000","updated":"2022-10-04 11:35:54.000000000","submitted":"2022-09-29 16:06:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859784","meta_rev_id":"d9588d9f8ed93fee2e6ab9f2461deecb0dd9f9b4","_number":859784,"virtual_id_number":859784,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-29 16:06:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-09-29 12:41:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-29 12:24:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-09-29 13:21:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-29 09:34:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-29 12:24:55.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-10-04 11:31:15.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"}],"messages":[{"id":"b7becdbcce55646b1f9ac4345dcb84e5c4c2eb86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-29 07:44:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc3c578b43f76815808773df580654d8a70fab3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-29 07:56:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0e7229d3a77dedd3954e1c1065ee2cc7032a5b92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-29 08:10:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"90a4bb5bf4cb1f354e5fb673c1983809e5014b59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-29 08:21:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"67018c49520489c608372d1588484d6baf530a6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 09:34:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c05c5b4771184069b2b4e85bcaff7e4c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f220a370423f44b496349368dfa7b672 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52b74bf80e3e4242a79b0da62514d714 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a37523a264f4766a69072819bdf1e9e : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98bb920ca05345c9ac12f36097ca02fa : SUCCESS in 7m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b13ff3f67ea94a9abcd30dead7d069e9 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ff2c1435bf540b88e26b8f35de8e0c4 : SUCCESS in 9m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/92cbdba5642c4cb7b48761c666555bdd : SUCCESS in 6m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1f5a28f3add941ee964fb395c046c7bb : SUCCESS in 13m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/800f452690b34c1a88fa0ebb8480c699 : SUCCESS in 18m 35s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b77b6f874cc94da4b6a32edf1780b71a : SUCCESS in 46m 00s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c32aa6cce82a4d2f8ac991e2da0efc3c : SUCCESS in 18m 58s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47dbf0f6b9814a71b95bb4ad0025d034 : SUCCESS in 1h 07m 03s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4555d5ed712c47d3a8cee47124025d14 : SUCCESS in 18m 07s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e80f86947a8b4150aef84948c31eb3d5 : FAILURE in 21m 53s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b541b342d224838a391833f2512af56 : SUCCESS in 13m 48s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64dcf79a4d7f4db78eedf87d4a62311b : SUCCESS in 14m 31s","accounts_in_message":[],"_revision_number":4},{"id":"689d48c44b8e39dc14ade14578b1c26763580989","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-29 10:11:52.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"f5dced50a432b8a8dc449cf07b6f8dab8e2afbe0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 11:29:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1aca1b28595c46b7bcdb334ee06cbefa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adf15fee28674bfbbd24a649f13c4b2d : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/104aacaae59d483e820404dae80cc527 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d02c6edd74de41f8831eee07e62935c4 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6716a3a2a8d456e960c9674f3baabe5 : SUCCESS in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06bc8c33fed34b0f8dfc1274473c2513 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1b1432257df4562be64e40648b0b051 : SUCCESS in 8m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4803f0495388443497e8cedbe15c3e68 : SUCCESS in 5m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fb70b2c352cc436f81f0cea77b594a9a : SUCCESS in 15m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6a45d1e9190d43d29de3b2396af52aec : SUCCESS in 18m 22s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45590551fc2d40da96efc193096faabc : SUCCESS in 56m 41s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9acee3a4fcc241a2be3d5f72daf25500 : SUCCESS in 15m 44s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c79cc046b8334b0ca9cb5ca0a0fa09fb : SUCCESS in 1h 12m 30s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86b1ed7e2b154696a4bccd05aa8b2bdc : SUCCESS in 15m 32s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7407f771835b48c7aa137e785d14aac2 : SUCCESS in 28m 08s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5682f3b447d40da82eaee5d86a256d2 : SUCCESS in 11m 49s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17eba844a4524e47918ba07fbbc8e349 : SUCCESS in 15m 19s","accounts_in_message":[],"_revision_number":5},{"id":"5e57d249c1da6cd8028a0a84894948361a2d7e08","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-09-29 12:24:55.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"7d99aed6e70a8f00ff016158ba2388a49dd6aef4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-29 12:41:35.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"096936e39e44961ee56386de1eba9ea4d2575a2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 12:41:47.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/66d5f19e3e284859bcf24c2bcc0ef0de\n\n\nWarning:\n  Change 859828 in project openstack/kolla-ansible does not share a change queue with 859784 in project openstack/kayobe","accounts_in_message":[],"_revision_number":5},{"id":"8cf5ca637075c34443e7381df066e042acb9a1c2","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-29 12:42:39.000000000","message":"Removed Workflow+1 by \u003cGERRIT_ACCOUNT_22629\u003e\n","accounts_in_message":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"_revision_number":5},{"id":"2d2043e70f03fb8b729e14b6225c05381b93e961","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-29 13:21:26.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"3a930b11af946e29888880d1c410122feebb148f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-29 13:21:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"dd7d43709e68b0173bbe3c1e2bec878f6530df6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 14:51:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cefeb4abff944a909e385afdfffea230\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3f5ebc082b244678232e203f7ad6e21 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c3bc4595b944825a10b138af01a5bce : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8daf9eb35ed4763b93cb866854d9bb4 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce694b9ad28a44dcba335136e86a03c4 : SUCCESS in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4e088eea43654b508b3009cdf71ccbab : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69ed421cbe1042df893ff1de9ce9e3a9 : SUCCESS in 9m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/511dc895dbc2498389abf91edb731976 : SUCCESS in 6m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6430bf8554894036ac9b7837f809c9ae : SUCCESS in 15m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/41e3cf4519b7452dba85386a3636bb7e : SUCCESS in 15m 13s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ccad3462858848d4bd52243a14d1cee2 : SUCCESS in 1h 04m 23s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/326423f1b9154b0198e59e6039321256 : SUCCESS in 15m 11s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93eb502668f34e9e857fd0faf54e022f : SUCCESS in 1h 24m 29s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5306e05faef44eaa1e645917009a95e : SUCCESS in 16m 01s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa5d5135254647b8b3be1e60ea8cc79c : SUCCESS in 32m 14s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/811816c133794f628e17105978f29358 : SUCCESS in 12m 03s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39244094fb544cbab60f2ccee5708b91 : SUCCESS in 15m 20s","accounts_in_message":[],"_revision_number":5},{"id":"83347aadd6efab4b327bfe61dc28db52a1888195","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 14:52:15.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"065952ea64a3b2e8fb2d0bfcc28ccdc80fa49b13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 16:06:15.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d27ceb4b0057424c899fe6b46fa5d6fc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c18a30ae10ea40328af658ad427a228c : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d52e36ffdff4a6aa453336a04177609 : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77798a820d8142cdbc7c2af795e9b230 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c58eaf4e5736411e9de88f9e2cb4f3ca : SUCCESS in 8m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1712d12f21424326ae6d21d2e076286c : SUCCESS in 7m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a4968a62986b40f883f13ed06dcd4244 : SUCCESS in 14m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bbe87f8113a74d81b11d8b83d5d60bb0 : SUCCESS in 18m 08s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69d0296f9b324256ab810dffd33676ca : SUCCESS in 51m 55s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a00c9e14b304527b4bbf5792d514361 : SUCCESS in 15m 08s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4224763c650541d28a428b8a38199634 : SUCCESS in 1h 09m 28s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d45de0bbbdcf409cb487bc36283deabf : SUCCESS in 18m 32s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1137cdec097e451190cef560a3e353b2 : SUCCESS in 35m 45s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b02692c485a34f34bf9c08ab72b6bd75 : SUCCESS in 14m 50s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d19be1427bd4cf881efdb458ffdee7f : SUCCESS in 13m 11s","accounts_in_message":[],"_revision_number":5},{"id":"b6c32396593ec7495637d3e1b45b8a2c5ba3a18a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 16:06:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"08c273d82d2084b99e7cf403c05949b7fabe6255","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 16:10:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2419ec4cf0e74828b4d114566c79c8bc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccf907de78d64405bd33d7360f65603f : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":5},{"id":"d88224cc82d2e2c01dd4cbf91c9c140a3fab7be9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-10-04 11:31:15.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"d9588d9f8ed93fee2e6ab9f2461deecb0dd9f9b4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-04 11:35:54.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"05a2c1acbd2dd4dd187d947f0fc6904dd71e93e2","revisions":{"4d0b43ec589f93b2a34b0dbc7f3046d89c5b0a7c":{"kind":"REWORK","_number":1,"created":"2022-09-29 07:44:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/84/859784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/859784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/859784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/859784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/859784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/859784/1"}}},"commit":{"parents":[{"commit":"608951fcce38428b155030badcff15a5a6a0445c","subject":"Merge \"Add new DIB upper constraints variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/608951fcce38428b155030badcff15a5a6a0445c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-29 07:43:53.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-29 07:44:06.000000000","tz":120},"subject":"CI: Set previous_release to Yoga","message":"CI: Set previous_release to Yoga\n\nChange-Id: I295e8f5f1cc9b7af1cd45ac788db473510220170\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4d0b43ec589f93b2a34b0dbc7f3046d89c5b0a7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4d0b43ec589f93b2a34b0dbc7f3046d89c5b0a7c"}]},"branch":"refs/heads/master"},"8942afcd39fd10ffe83a5073be793320620b2b41":{"kind":"REWORK","_number":2,"created":"2022-09-29 07:56:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/84/859784/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/859784/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/859784/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/859784/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/859784/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/859784/2"}}},"commit":{"parents":[{"commit":"608951fcce38428b155030badcff15a5a6a0445c","subject":"Merge \"Add new DIB upper constraints variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/608951fcce38428b155030badcff15a5a6a0445c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-29 07:43:53.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-29 07:56:07.000000000","tz":120},"subject":"CI: Set previous_release to Yoga","message":"CI: Set previous_release to Yoga\n\nChange-Id: I295e8f5f1cc9b7af1cd45ac788db473510220170\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8942afcd39fd10ffe83a5073be793320620b2b41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8942afcd39fd10ffe83a5073be793320620b2b41"}]},"branch":"refs/heads/master"},"84a313a94b3425c27e0ce72222585455775f7bc0":{"kind":"REWORK","_number":3,"created":"2022-09-29 08:10:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/84/859784/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/859784/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/859784/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/859784/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/859784/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/859784/3"}}},"commit":{"parents":[{"commit":"608951fcce38428b155030badcff15a5a6a0445c","subject":"Merge \"Add new DIB upper constraints variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/608951fcce38428b155030badcff15a5a6a0445c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-29 07:43:53.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-29 08:10:50.000000000","tz":120},"subject":"CI: Set previous_release to Yoga","message":"CI: Set previous_release to Yoga\n\nChange-Id: I295e8f5f1cc9b7af1cd45ac788db473510220170\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/84a313a94b3425c27e0ce72222585455775f7bc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/84a313a94b3425c27e0ce72222585455775f7bc0"}]},"branch":"refs/heads/master"},"67ca8218c621758cb1381d2babfbc2ceaa526db0":{"kind":"REWORK","_number":4,"created":"2022-09-29 08:21:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/84/859784/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/859784/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/859784/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/859784/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/859784/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/859784/4"}}},"commit":{"parents":[{"commit":"608951fcce38428b155030badcff15a5a6a0445c","subject":"Merge \"Add new DIB upper constraints variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/608951fcce38428b155030badcff15a5a6a0445c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-29 07:43:53.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-29 08:20:44.000000000","tz":120},"subject":"CI: Set previous_release to Yoga","message":"CI: Set previous_release to Yoga\n\nChange-Id: I295e8f5f1cc9b7af1cd45ac788db473510220170\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67ca8218c621758cb1381d2babfbc2ceaa526db0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67ca8218c621758cb1381d2babfbc2ceaa526db0"}]},"branch":"refs/heads/master"},"05a2c1acbd2dd4dd187d947f0fc6904dd71e93e2":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-09-29 10:11:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/84/859784/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/859784/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/859784/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/859784/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/859784/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/859784/5"}}},"commit":{"parents":[{"commit":"608951fcce38428b155030badcff15a5a6a0445c","subject":"Merge \"Add new DIB upper constraints variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/608951fcce38428b155030badcff15a5a6a0445c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-29 07:43:53.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-09-29 10:11:52.000000000","tz":0},"subject":"CI: Set previous_release to Yoga","message":"CI: Set previous_release to Yoga\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/859828\n\nChange-Id: I295e8f5f1cc9b7af1cd45ac788db473510220170\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05a2c1acbd2dd4dd187d947f0fc6904dd71e93e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05a2c1acbd2dd4dd187d947f0fc6904dd71e93e2"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
