)]}'
{"id":"openstack%2Fkayobe~859982","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I4c3afb71600781dffe8072303d775cbfad16cbac","project":"openstack/kayobe","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-10-13 21:38:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4c3afb71600781dffe8072303d775cbfad16cbac","subject":"Update to latest CentOS Stream 8 cloud image","status":"MERGED","created":"2022-09-30 12:41:00.000000000","updated":"2022-10-13 21:39:03.000000000","submitted":"2022-10-13 21:38:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859982","meta_rev_id":"fa945651dffee4e41a09c078fbc80c3940bc9fa6","_number":859982,"virtual_id_number":859982,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-13 21:38:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-10-12 06:50:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-10-13 19:20:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-10-13 19:20:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-30 14:43:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-12 06:50:59.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-10-13 19:20:18.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"b5e0c1f61e727ac8be18dc8d00ff5f2323ff6c23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-09-30 12:41:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5199ee609c06deb781ec6f0406085f02319f8dd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 14:43:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b54c7041abb42d0b35ab7fa3f27c35c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0661a1a467f34472b8327ef49d3c478e : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de814b8ed13b4d2a8273f004b3ac7827 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5917752360ea474696543658994cdd22 : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2324630c4d91408b9d8dcb0b16f88385 : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dda8ef675daa4673851ed1630781baa0 : SUCCESS in 9m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/551ac7be42f747bb8b9b84fc77d15e18 : SUCCESS in 3m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a381abb8748d4e60b8aedbff8d28bad9 : SUCCESS in 5m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1ffb9858678a4189bfc7b94111fd13b4 : SUCCESS in 16m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b77f8bf5280f49e3a4028f3cb37f17fb : SUCCESS in 15m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ae12be5a1a4046538ff0f37b569ea1f3 : SUCCESS in 1h 31m 26s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/c10503194f7a4291819602db0034f538 : SUCCESS in 1h 13m 51s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4935127151204e11ae6f3a6a8c4d7614 : SUCCESS in 55m 56s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/663ef5b00e874ec5a4d98a39655796d3 : SUCCESS in 1h 07m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d095e0b8c3204e27af432d8cb2c65f28 : SUCCESS in 24m 23s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/0787adbab5234dbb809da0d90ca2c2fd : SUCCESS in 20m 22s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13d1774a43014f94a7acf2045039536d : SUCCESS in 16m 10s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9ce3da0d2cec4984906f509cdc98d533 : SUCCESS in 1h 57m 18s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d217fb553db4a43a0b8192f9a13bdaf : SUCCESS in 1h 41m 38s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c54a988a6060487e968e2d2f4b4995ef : SUCCESS in 22m 08s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/9c796929b16343ff9d384920ebb0e7d1 : SUCCESS in 32m 34s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3248539c54914ce3b14ee302689f7cbf : SUCCESS in 15m 42s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/873672b4ca9546c589174d6fea1abef8 : SUCCESS in 59m 49s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1cc3295545314209b9959f9feb66f18d : SUCCESS in 32m 17s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58e83c1232f1427ebde5cb74394028b2 : SUCCESS in 31m 56s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4d4172755e1b4004bdba8c9d50551ac5 : SUCCESS in 17m 31s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a50491ddac8c4738a07e615df5b324e5 : SUCCESS in 14m 47s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3d12c4d16264b7dbf7e14ba430dcc12 : SUCCESS in 12m 46s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/beb7489c34ba45caaeac3c3f133a8aec : SUCCESS in 19m 24s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/aa6c35a19a1548f486736d8fd7a1bd33 : SUCCESS in 16m 50s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/458c4c5382a44b77b379fe809c47639b : SUCCESS in 14m 10s","accounts_in_message":[],"_revision_number":1},{"id":"caa6aebfd50b66888d3265c0ab9290ab5f9a560e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-12 06:50:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2f14c7fbe54a8f0ef5de3c7418c249c8f2e095b9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-10-13 19:20:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fdbfb0f7c19400b576fe13c21654b9656929750c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 19:20:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f803b7c52ee6ad8f519413aa2f37168347d141e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 21:38:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0baa3500fe224d889a08788da57d527f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adaee501149941c19d2779656d2828e4 : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e95aa3e67ada4dc3992519d0df538e93 : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c65a1998b9e46c982e81c57da46d23b : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/596813fc5e3948efa7abbcad74eeed9d : SUCCESS in 9m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cc110dfb4d74bbca1b8620b36e4fe6a : SUCCESS in 3m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2685f4c6b56248f7b3e1dda1bc1436ec : SUCCESS in 6m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/79f0090aaf7f4756b0a52a8b2c780bc4 : SUCCESS in 15m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/98a4e3f100d44b9ea4cd5c9bb8251115 : SUCCESS in 17m 35s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f1116c4dd9234036a1edc70e917be90f : SUCCESS in 1h 07m 58s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/c5887aafbbb44db49d58b29c825335d5 : SUCCESS in 1h 24m 52s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9134a0680a2340f8acf40c1bfa2ae084 : SUCCESS in 56m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/504b4202f45644b3829b3e3b2a59e394 : SUCCESS in 1h 20m 50s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/effdf6ee29e04266b5f3bab28ec2ece9 : SUCCESS in 24m 24s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/3fbb053cf2b7451799ba4215ff08ed9d : SUCCESS in 20m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c53cca5d8f52469096fffa0a234338e8 : SUCCESS in 19m 24s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cb993934b4d74a7c985c47e8b2a9471a : SUCCESS in 1h 22m 35s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/977bcbf9703143a492b198c9235a2589 : SUCCESS in 1h 02m 30s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a99efdb761964ccda0d12338d1c64f21 : SUCCESS in 17m 02s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/abd90b2a68944f5f87e189bbd8f16619 : SUCCESS in 18m 48s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46945e89d211415d92d034c224ee4363 : SUCCESS in 14m 29s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/63cfd54f9d724a4da13dd97d1cf8a930 : SUCCESS in 27m 29s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da20ebea99e44165a8cd1ce246607b24 : SUCCESS in 29m 58s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/49ba475d905345429e8c59625253a5de : SUCCESS in 13m 42s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/bc46abfd3a814d50b6982d10721ad222 : SUCCESS in 14m 37s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52030312e1a7437fa358312efb5ab860 : SUCCESS in 12m 30s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/3766df51628e405c8c58771722aaa55e : SUCCESS in 16m 55s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/164d65631bf747d79217f9ca48e87daa : SUCCESS in 18m 43s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26f94f80bc0b4df981c0272bb1cc4499 : SUCCESS in 13m 48s","accounts_in_message":[],"_revision_number":1},{"id":"62fe2a7324742297f8267438c39f87aa6adfd444","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 21:38:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fa945651dffee4e41a09c078fbc80c3940bc9fa6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 21:39:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43c5b696e9d945b1a92f099490ef2a0a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f50e48b0684b4d1c9547e1a67cb939a5 : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b69fce00c92b4030b7f4562dadd6314a : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fd25eb9e04c86ce5faec70ef5cff15057bea3ca3","revisions":{"fd25eb9e04c86ce5faec70ef5cff15057bea3ca3":{"kind":"REWORK","_number":1,"created":"2022-09-30 12:41:00.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/82/859982/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/859982/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/859982/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/859982/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/859982/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/859982/1"}}},"commit":{"parents":[{"commit":"8477dbd7499dbb13087ef2ffe3d5021c6a7a22c9","subject":"[CI] Move queue setting to project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8477dbd7499dbb13087ef2ffe3d5021c6a7a22c9"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-09-15 11:13:51.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-09-30 12:40:19.000000000","tz":120},"subject":"Update to latest CentOS Stream 8 cloud image","message":"Update to latest CentOS Stream 8 cloud image\n\nChange-Id: I4c3afb71600781dffe8072303d775cbfad16cbac\n(cherry picked from commit 0db7684428d04dc8b5c24905ca3ff52ec6e63bd9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fd25eb9e04c86ce5faec70ef5cff15057bea3ca3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fd25eb9e04c86ce5faec70ef5cff15057bea3ca3"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
