)]}'
{"id":"openstack%2Fkayobe~859983","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I4c3afb71600781dffe8072303d775cbfad16cbac","project":"openstack/kayobe","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-10-14 14:42:44.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I4c3afb71600781dffe8072303d775cbfad16cbac","subject":"Update to latest CentOS Stream 8 cloud image","status":"MERGED","created":"2022-09-30 12:45:04.000000000","updated":"2022-10-14 18:28:07.000000000","submitted":"2022-10-14 18:27:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859983","meta_rev_id":"fcb7e647e72d380f51957f75803ff4dc9ac0aff4","_number":859983,"virtual_id_number":859983,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-14 18:27:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-10-13 19:24:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-10-13 19:24:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-13 19:24:35.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-10-14 16:53:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"451475cec19570a5c58a368fe9bbe58401c1f609","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-09-30 12:45:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8998997f0683d9d737aa25be8b08da7e88e49562","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-10-13 19:24:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"019129fa73594b82fd5bcc058e63d2daa7f6d2a0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-10-14 14:42:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"34ddf78e6b4b43fc4858df1a6764605a76a85422","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 16:53:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/338d3a46ab3444a8ab0f342d5d8ba29c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2074e1bf74744c6b73a6ca75d4f0438 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/241294b1976646c4bfacb31b9789c2eb : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62ad2d07e95449faae277a1696b71604 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1884974395247518119cf190671dfc4 : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/edb741e60892443e850205a22f046ab3 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e130d4ebd3be45cca37ccd077f494d8d : SUCCESS in 8m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7ebe76070e24fcb84c3a8693cc12684 : SUCCESS in 3m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/79d39747a43949e0b600ca4b7ad02caf : SUCCESS in 6m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/875a8ad17e6a4aeea0f44f03badf618f : SUCCESS in 12m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eddedea15b7b4aad94e5b8d53487789d : SUCCESS in 19m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/972c64aa776c4a5c8f9a5475487540a0 : SUCCESS in 59m 33s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/890ddcc2fd744326babf378d2c95074f : SUCCESS in 1h 10m 16s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5244b1779e5432f907c39b181607008 : SUCCESS in 49m 08s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/12e93c4e9a9546e38106aea1f6702f29 : SUCCESS in 1h 05m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6b3be219eefa40d790876fa4fa36f3ed : SUCCESS in 22m 38s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/e498ba79de984067b78962e000a24210 : SUCCESS in 22m 26s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/174809e1341740848321424180c7bf33 : SUCCESS in 17m 24s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c65eb76beb9e4048949978c266b5e8ff : SUCCESS in 1h 39m 21s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31fdd167ea344724a0f4c53c1757f920 : SUCCESS in 1h 03m 06s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e9bf4e7bc51049279d0f1baf82b64b9a : SUCCESS in 21m 26s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/0cfe5fa69a4b4b75bf68b28a77236f62 : SUCCESS in 20m 02s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/345d6f794fe547e9b8ff064ba020a2d3 : SUCCESS in 16m 23s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/5e4d3704763f4a1ab1ba9eff9e2ef45c : SUCCESS in 34m 53s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/702aa118fb644171a322ffa7f8c16d3e : SUCCESS in 25m 47s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7edd651fb7ca4b71816a3d699e02e845 : SUCCESS in 24m 14s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8c6a51bc70bd4308a90c1f220c5a71ec : SUCCESS in 14m 41s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2aeaa457738a401aa972e17cc2cbf7ba : SUCCESS in 17m 59s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3f016cc5a4341ddacc7a7ea7af0aeac : SUCCESS in 13m 25s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/ca3f7485824d4a30914e51f76800f59d : SUCCESS in 17m 04s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/9da4fb59a1044bc6ae4f254a58ac7dc6 : SUCCESS in 18m 01s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65c67f09627a44acae1fe1caa9358f08 : SUCCESS in 12m 27s","accounts_in_message":[],"_revision_number":1},{"id":"8a108bf35ee85ef223948826f536e2b3e94a27aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 16:54:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"72506145cc836b730cacc518082b0b1ab330758e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 18:27:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abc201ee472c4c73a5efbae73209a942\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a37ed83579f4685bbb423b5bc5272b2 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e8edfc6ad6445d99a4236f54bcfca4d : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5b894081d1d40b692d18b09639cb5e8 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6215a02a30d4484d8e4da79795ab6037 : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba793b96e19d4cff94a1eb8c6ebc0be6 : SUCCESS in 4m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4ab11c581a654a15a8c8c97753101cd5 : SUCCESS in 5m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/118b6390b95c4ff3aff6b9c95f9c5cb6 : SUCCESS in 11m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/45ee19cd306348099ce73e134d6cd2a2 : SUCCESS in 18m 07s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3cf63233e19545cc96b181ba8523a2dd : SUCCESS in 1h 05m 00s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/0385e26863f44b5280de4c8329b2d136 : SUCCESS in 1h 00m 44s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9acc5dc2f844723a0dd7006e279b345 : SUCCESS in 57m 43s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/77071c3858cf4244b14372f5504aab82 : SUCCESS in 1h 03m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2e566169421040daa57151e9f3d816fa : SUCCESS in 20m 37s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/2b1f7e461ae24d639516509559302a7f : SUCCESS in 22m 32s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5c47ec1f2194a43bf6108f1efe18ded : SUCCESS in 15m 34s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/98ba6f8a3f514598ad53fc5147fd27b5 : SUCCESS in 1h 23m 20s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9bd2462cfeb4cf8b0e352a5c160f125 : SUCCESS in 1h 08m 45s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/51e11439cfd3409ca5b74e66d82c012b : SUCCESS in 22m 21s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/6c831a27ca754d62a051e7d5e24d0bc0 : SUCCESS in 22m 04s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f65f87bb10a4455681b3a7644b36ac7c : SUCCESS in 15m 56s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5b3d12d5dbe24fbcaeaa3397d138ece9 : SUCCESS in 29m 23s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7db1e663776549eabd654df8ede1286e : SUCCESS in 18m 05s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/54cd058365e74e638a88a7f7acf99119 : SUCCESS in 17m 19s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/02bc367b4a1e426787d7caafd90c4f81 : SUCCESS in 17m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1be3c443f3a44d1fb7cac45d25e2d93f : SUCCESS in 11m 37s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/11cdb8e556d8443884f0821c298712ad : SUCCESS in 15m 43s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/82b015a25e944fe68296b44c30505a11 : SUCCESS in 20m 16s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/858a89a92d6b481fa62a7dd63433e415 : SUCCESS in 13m 56s","accounts_in_message":[],"_revision_number":1},{"id":"38dfe11f5abadb24aea7444275291d76a326c8c4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 18:27:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fcb7e647e72d380f51957f75803ff4dc9ac0aff4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 18:28:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5644e61918814c33831cd046cde2e5a0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c1ba86a48cf4dde80217b5aff54abf8 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4dba37c3aab54e22b6b63f9ae4ce723a : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a842bf3374891946b624ab1e8e27ed47447fc5ab","revisions":{"a842bf3374891946b624ab1e8e27ed47447fc5ab":{"kind":"REWORK","_number":1,"created":"2022-09-30 12:45:04.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/83/859983/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/83/859983/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/859983/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/859983/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/859983/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/83/859983/1"}}},"commit":{"parents":[{"commit":"612fc25dd0753c9c078a915888ae1f8fb282f2c7","subject":"nclu-switch: Fix order in template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/612fc25dd0753c9c078a915888ae1f8fb282f2c7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-09-15 11:13:51.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-09-30 12:44:42.000000000","tz":120},"subject":"Update to latest CentOS Stream 8 cloud image","message":"Update to latest CentOS Stream 8 cloud image\n\nChange-Id: I4c3afb71600781dffe8072303d775cbfad16cbac\n(cherry picked from commit 0db7684428d04dc8b5c24905ca3ff52ec6e63bd9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a842bf3374891946b624ab1e8e27ed47447fc5ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a842bf3374891946b624ab1e8e27ed47447fc5ab"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
