)]}'
{"id":"openstack%2Fkayobe~859985","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I4c3afb71600781dffe8072303d775cbfad16cbac","project":"openstack/kayobe","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-10-13 21:37:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4c3afb71600781dffe8072303d775cbfad16cbac","subject":"Update to latest CentOS Stream 8 cloud image","status":"MERGED","created":"2022-09-30 12:48:33.000000000","updated":"2022-10-13 21:38:58.000000000","submitted":"2022-10-13 21:37:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859985","meta_rev_id":"8cdf8a824d66cf50646a3e04ae87db1da2d6e0f6","_number":859985,"virtual_id_number":859985,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-13 21:37:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-10-12 06:51:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-10-13 19:20:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-10-13 19:20:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-30 14:48:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-12 06:51:04.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-10-13 19:20:07.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"3d1a1afbc8db21c83418fba1bb51ad531aeae72e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-09-30 12:48:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f26989d671dde399948084d8b17831769e8a8961","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 14:48:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76621577460841c18f3677c886382292\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aab1e154a1704096be6a6a179a9db908 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28ecc202091b4e1da7338a57db7c331c : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad374fa9367244a7a3a1bf2b5bbea949 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d3f2b3c5dce4efc871c21cdc679080a : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f1cbe8019314a1a8a9e5b80113c0c1a : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40db17aaea6340c6a2cb6a745e5de160 : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cd199df6a644ac5a9a20b3de6f445bb : SUCCESS in 3m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8aa581eb54bc4a3f8a321734434d5f1a : SUCCESS in 5m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b8b1654f2e8e46e7b57bc846a00cddab : SUCCESS in 10m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d40e24e84de3408e8f8a21c035b5fd52 : SUCCESS in 16m 16s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c7b87a2dfcea49b8a0b418085f4752cf : SUCCESS in 56m 19s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c68131eedeef41faa64eca41d0a0cabc : SUCCESS in 46m 35s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/47a9693e1ab24a70af931092c36616d9 : SUCCESS in 1h 18m 22s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/fdb246ad08574cc488404cb10470f669 : SUCCESS in 17m 08s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/723b2a00067240edbdefe7538e5160bf : SUCCESS in 14m 39s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/621b791038ef4fa7abe10a582c9f851b : SUCCESS in 1h 52m 14s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d381e2dc62f6438383d2c9f215dea231 : SUCCESS in 23m 49s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe58360021be476890c3caf8d48de8a3 : SUCCESS in 12m 38s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/6c51f567aed04565ace197845caf67be : SUCCESS in 46m 10s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0d8f7f49ff1f4c2aa8b935a34e24d2d3 : SUCCESS in 24m 13s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/091cac2b993141499914832f6e8ed222 : SUCCESS in 13m 23s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0390fd0b76f04270ad0fffcc0ef1aba2 : SUCCESS in 10m 02s","accounts_in_message":[],"_revision_number":1},{"id":"29b3bbaf29df933ed062073dc178c32de146542c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-12 06:51:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b38e8ef0a7d336433678b4dd24c05a1b92ab99df","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-10-13 19:20:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"703e7032d056ac98192502b9d85a2c4d085ce39f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 19:20:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9f7e3c7eb6b1fadd812d14d31bbfd9b688ed92e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 21:37:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7051acb8008438490f78c88550c07c1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0615b01151d4d60b447ebabfe58c7b9 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c2f2e828af84d1ea3e2761ce2602055 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb1921ef305f479e91b53e12807a2ee3 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ffdefcfe62b4229b058f48c155e02d5 : SUCCESS in 8m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0624040368304f3aa5ecfcf975935e29 : SUCCESS in 3m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2a5c3541af7a41f69899f0d9b6543049 : SUCCESS in 5m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bf92264278704e6bbbef4d318eb6bd70 : SUCCESS in 11m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d495c055e5e64acd80309b558a001cf8 : SUCCESS in 17m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/02fec2bce95241ed863f156f7127bf05 : SUCCESS in 1h 42m 11s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf47faabb3844c5ca8f605064e50e22f : SUCCESS in 1h 07m 50s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/150eb56f7f634e10ba029cb15327a0e3 : SUCCESS in 1h 40m 12s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e62a93870ce446b4be4d6daf923c069a : SUCCESS in 21m 05s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33fbc3d1bf674de7a8b0ea5d0af756c8 : SUCCESS in 15m 48s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2bb63be8c9884a779597fa94a7d2acb5 : SUCCESS in 2h 11m 33s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/943a1b7247404d6580500f764e52ca15 : SUCCESS in 21m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/676f40bc63e74823b9cd37169deb8f58 : SUCCESS in 15m 44s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6b5e89a82b66420b85c602a54cc52e7d : SUCCESS in 32m 24s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/fa302c75b59c42af8607345a50bcb5a2 : SUCCESS in 14m 46s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58365aaad4424747a3630a45942e876c : SUCCESS in 11m 18s","accounts_in_message":[],"_revision_number":1},{"id":"d38f3385d04dc0b83729fccf14f4bb5aa0a4199e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 21:37:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8cdf8a824d66cf50646a3e04ae87db1da2d6e0f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 21:38:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52934d4fc3204017a5187f7ff523204e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cbbb53f04424eb79b96ed9ae5988a5f : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7d0b7096f824c8eac0923a4c0860434 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9af442ea7b66b7b548b69a783b7187ce78f21dd4","revisions":{"9af442ea7b66b7b548b69a783b7187ce78f21dd4":{"kind":"REWORK","_number":1,"created":"2022-09-30 12:48:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/85/859985/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/859985/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/859985/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/859985/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/859985/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/859985/1"}}},"commit":{"parents":[{"commit":"b30efa49bde4b2aafb4874276f017afcdeb89f21","subject":"[CI] Move queue setting to project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b30efa49bde4b2aafb4874276f017afcdeb89f21"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-09-15 11:13:51.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-09-30 12:48:22.000000000","tz":120},"subject":"Update to latest CentOS Stream 8 cloud image","message":"Update to latest CentOS Stream 8 cloud image\n\nChange-Id: I4c3afb71600781dffe8072303d775cbfad16cbac\n(cherry picked from commit 0db7684428d04dc8b5c24905ca3ff52ec6e63bd9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9af442ea7b66b7b548b69a783b7187ce78f21dd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9af442ea7b66b7b548b69a783b7187ce78f21dd4"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
