)]}'
{"id":"openstack%2Fkayobe~860652","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I2c9dfeca7f0d37a96f9cbd9df51d94098cf07258","project":"openstack/kayobe","branch":"stable/wallaby","topic":"ip-allocation-skip","attention_set":{},"removed_from_attention_set":{"25600":{"account":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"last_update":"2022-10-28 14:14:19.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-10-28 14:14:19.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-10-28 14:14:19.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-10-17 08:18:09.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I2c9dfeca7f0d37a96f9cbd9df51d94098cf07258","subject":"Skip IP address allocation and configuration if needed","status":"MERGED","created":"2022-10-07 11:02:08.000000000","updated":"2022-10-28 14:15:19.000000000","submitted":"2022-10-28 14:14:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860652-ip-allocation-skip","cherry_pick_of_change":821129,"cherry_pick_of_patch_set":10,"meta_rev_id":"8ea2710148fefdb31ac5c3f4c004f01987c84dc2","_number":860652,"virtual_id_number":860652,"owner":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-28 14:14:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":1,"date":"2022-10-14 09:12:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":2,"date":"2022-10-28 12:24:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-10-14 09:13:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-17 08:18:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":1,"date":"2022-10-28 12:24:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"}],"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-07 11:02:08.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2022-10-07 12:33:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-07 13:06:21.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-10-13 19:08:39.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2022-10-14 09:13:05.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2022-10-17 08:18:09.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-10-28 12:24:48.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"bcb7784c3e646d4838273e21d83d2e92297a612e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2022-10-07 11:02:08.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"40277eb653f63fdc56bbd247034d86500646e136","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 12:33:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d94996f6efe4051a3f8dd4b91331de3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b50970ab0e87435db26a652106ff1426 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d97a48f9564c4b1bbb3e73acd74efce6 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e0704c3c8294a6892224a5614775d01 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a65ba60dde544638b3f563d9a515a4d : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ec9104e360043d6b858fbf6bd8d964c : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4de173ce86364451b6142e84432e693c : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abfa934f45324d368c8f88554005b595 : SUCCESS in 3m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8ff493b4d4d14f91af6cea39178b8093 : SUCCESS in 5m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/debbc20b049a4b479d357fd46719f166 : SUCCESS in 9m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2e60d38fd2d34fc8bfa769c499bc754d : SUCCESS in 16m 27s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b034d2327c24436eaad50af185255826 : SUCCESS in 1h 11m 11s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c810d1995a914471bbc281018e0d3a42 : SUCCESS in 52m 44s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9003080690464d1ab62a82d6025843d9 : SUCCESS in 1h 09m 26s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/96d04c1451894dd6b2793f1b74f08c98 : SUCCESS in 19m 49s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/063f98573c5d45f9abb81b853919159a : SUCCESS in 14m 07s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/25656ad88cd64369ac3b16693127d3bb : FAILURE in 1h 25m 27s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/33e4843fe85d4d688e190bc812bc4955 : SUCCESS in 19m 02s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa387bac0b2a4f9a874b21ebfc47cb00 : SUCCESS in 14m 52s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/369071814a21421ca66941f2b999e589 : SUCCESS in 46m 30s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a5cf7cf671c34469886080ef6bba2750 : SUCCESS in 26m 29s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/48f5d64cde50468baceab787ea67326a : SUCCESS in 15m 27s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be85f0d3d7dc4ef698874ac177deaf2c : SUCCESS in 13m 19s","accounts_in_message":[],"_revision_number":1},{"id":"51a6d6f11f78ca4c79e0290ec5f6a00681a99eaf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-10-07 13:06:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8af677538d76989521c5fca78d49847f88a00775","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2022-10-13 07:02:41.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ca3bcdcf1032b17579438817cf8b002f20dd1a4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 09:42:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a187e7b9b560489cb5a3e9da7f672a74\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f380321b3fa47e6b15913e4a87bdc73 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b51e871852464c5cb28302467bb6af68 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0026b4e7b892422abbd73fb95a40735f : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07aff6bb79e24c29b7cc5bbc88d1a93f : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48fcd6efc7e54a32b3c5d56ea28f8622 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e64b087c57194d8097bcd07032e33ce5 : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f1d612269e941e2aa7a569577199e6e : SUCCESS in 3m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f488fee372a043bfb074915a5c8fbf45 : SUCCESS in 6m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/739a7b6ee1d944f3b1a520ec81a62df8 : SUCCESS in 10m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f13c63c282dc499cbde862c7ab3d4f5e : SUCCESS in 16m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/33a40841077a4c19808714db25815d5c : SUCCESS in 1h 57m 24s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee0b8f77c8ea4ac28dc7ad30129185e8 : SUCCESS in 52m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4b6e44d526974d8cb5a40e6f121914ae : SUCCESS in 2h 02m 00s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/72cf8d6b1fff4fabaf910227c8c75d6a : SUCCESS in 20m 42s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2472532f9ea3401f842ab693258576f1 : SUCCESS in 14m 09s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ddb682443fd641cba30a2f572848ecfc : TIMED_OUT in 2h 34m 17s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/dd432ecc38c84523870e793e27fac9e8 : SUCCESS in 19m 01s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/557d6d9db9124e0b8498f54349dee77a : SUCCESS in 14m 46s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/f382d6298675458397358dfb447f5242 : TIMED_OUT in 1h 33m 58s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/47087d7312ef407a8f354d70508640a0 : SUCCESS in 25m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/75ee1eb4b7be476b9deb8eeae43ee0eb : SUCCESS in 15m 29s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b5d6fb6bf2443f9923a810ce9fa8068 : SUCCESS in 10m 42s","accounts_in_message":[],"_revision_number":1},{"id":"d0bfc601e521906fdf72c0c2e463385281f9b80d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-10-13 19:08:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c938e66da114a24e3f9402755dc0f660e4a1e1d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-10-14 09:12:00.000000000","message":"Topic ip-allocation-skip-stable/wallaby removed","accounts_in_message":[],"_revision_number":1},{"id":"b1c828955a06e7a6807d0e5d9bbb780568751ac4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-10-14 09:12:19.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"b18fdd39dac5bd4c8905c181665d0eff84335290","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-10-14 09:12:54.000000000","message":"Topic set to ip-allocation-skip","accounts_in_message":[],"_revision_number":2},{"id":"b25d4a854853195e3dde01be261f5ca39b784bad","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-10-14 09:13:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"89e0ded0de38b5195f2457698457f54d1b675bb7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-10-14 09:14:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"10a63da30a9ce6fc69a655b05a37b7391ac2164d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 11:11:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4091156d126942d18fd9a1aad40247ed\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5306385c385405fac3335f8d49fd869 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/704c23fdf91842e4b2da3634ba244a97 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dee471742a7c4936929641ec3953d610 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2f949fb372a40d4ae9e1d819ef435a5 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/479a04e6f44c480bad588b940bbbaaba : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21147d35fdc84faaa75b228b1820d5b9 : SUCCESS in 9m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a024aa11f16447239da6a8ab4c84d0f7 : SUCCESS in 4m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1c8d9f9c37464f538c0e660d6d1196ce : SUCCESS in 5m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7f2945c392e642108e378e1ca4fc8eca : SUCCESS in 10m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7b65c9d4661842469d9de2c9e6676581 : SUCCESS in 18m 12s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b3cabf22e7f94377af3a2aa15efcf5b7 : SUCCESS in 1h 07m 09s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69984d6df78a41b8ab7c74e5fd2a6205 : SUCCESS in 57m 31s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4f003ef07b0c4e1192aec32564954b0f : SUCCESS in 1h 11m 39s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7b49d4eafafa4ee5b1c83e05f6ce7d23 : SUCCESS in 22m 16s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/276a37964196419e8d02c2a338a16cfa : SUCCESS in 13m 49s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ad1b76c23bfc49db8e201af7af21b761 : SUCCESS in 1h 51m 52s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e4876960e8c848209ee14014cd838652 : SUCCESS in 19m 47s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38dc12a4477d40f0b4beea58331a1780 : SUCCESS in 15m 12s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/5be23f9a2d864910b2c2fc3d2cf0e616 : SUCCESS in 45m 22s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d43b041c7f0a482dbe99ecb09ab405e4 : SUCCESS in 28m 03s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c5780983ddbe4531ba4aeff916fb6b7b : SUCCESS in 17m 10s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51c98b8044124007838735d546b01cd6 : SUCCESS in 10m 54s","accounts_in_message":[],"_revision_number":2},{"id":"7ab2bb1afdd1b3ba736ac5fea27ba082d4cf42f6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-10-17 08:18:09.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d46de3ac8e76cf5e3aff936072a37c3235d4ab4f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-28 12:24:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fb2cddc6f3cb43a9fd8bc963ded0aa3035b005f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 12:25:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2c71c37b2ed1260c61a41580220063e9bd8e863c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 14:14:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f12dd3d9ca81408d8d2f4eb460c206c9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71525936bf78483a845d34e77a9c9d7f : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2d81cfb31b248b1855df13b173da80b : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c353d81882344980a22848e258cd7c46 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/581d5d1da19c4c7aa5fdb18873bc1e1a : SUCCESS in 8m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6548ede0f485414b94e366fab95aa8e1 : SUCCESS in 5m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/487953f9a34a4234abee7e083f85528d : SUCCESS in 4m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6ac1ae9cd60547e096b0107cd344072b : SUCCESS in 10m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/44d6fd3ade424b38a2ef9acf1ce60610 : SUCCESS in 17m 05s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7addf1434f514c30b672119c56060a97 : SUCCESS in 1h 04m 11s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5d8432d67cc420db98aa4d53d1cef2a : SUCCESS in 52m 13s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/2c283d0c5db7417a92e2f9104440a860 : SUCCESS in 1h 09m 50s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c096df626d634604aa53de8a9249f724 : SUCCESS in 20m 48s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d45fe4e41854db1a4ac126803021e33 : SUCCESS in 14m 46s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/37a15cff8c614f6a8cc74e28628e692b : SUCCESS in 1h 41m 16s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1e40b18eb50f45459b1315f6392d7ee6 : SUCCESS in 19m 05s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fff341be3cd4ba0b1fe10e99912aa39 : SUCCESS in 14m 33s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d81a95050326406796002483d439adfe : SUCCESS in 24m 56s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a0875a97248c4dbdbe522a6e0055c186 : SUCCESS in 15m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb948af1e8e54766b3fd5ca10dc8fc75 : SUCCESS in 12m 00s","accounts_in_message":[],"_revision_number":2},{"id":"5c78ad579cf17a689e39e25d396ceb5b48d78b76","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 14:14:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8ea2710148fefdb31ac5c3f4c004f01987c84dc2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 14:15:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69071ce5ba2842339f277b9e26284fd6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed7613f210ef4a7b8dbe78e99aeb691b : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/592508756c0f490fa516f1c417b723e7 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e2aadb31c150795a6ed46f8545fa8fef311397d1","revisions":{"ee162781ad9f8120cd93d864d7d9afc9410792c0":{"kind":"REWORK","_number":1,"created":"2022-10-07 11:02:08.000000000","uploader":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"ref":"refs/changes/52/860652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/860652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/860652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/860652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/860652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/860652/1"}}},"commit":{"parents":[{"commit":"b30efa49bde4b2aafb4874276f017afcdeb89f21","subject":"[CI] Move queue setting to project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b30efa49bde4b2aafb4874276f017afcdeb89f21"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-12-08 20:52:25.000000000","tz":180},"committer":{"name":"Ebbex","email":"openstack@slipsprogrammor.no","date":"2022-10-07 11:02:08.000000000","tz":0},"subject":"Skip IP address allocation and configuration if needed","message":"Skip IP address allocation and configuration if needed\n\nSometimes some hosts should be configured with an interface without any\nIP address set (e.g. bridged interface) and to achieve that this change\nadds the new attribute \u0027no_ip\u0027 for the network configuration. Also the\nchange contain a test for this.\n\nChange-Id: I2c9dfeca7f0d37a96f9cbd9df51d94098cf07258\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 1de4f2a4a3f6ef16399d812a544c6da622609e24)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ee162781ad9f8120cd93d864d7d9afc9410792c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ee162781ad9f8120cd93d864d7d9afc9410792c0"}]},"branch":"refs/heads/stable/wallaby"},"e2aadb31c150795a6ed46f8545fa8fef311397d1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-10-14 09:12:19.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/52/860652/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/860652/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/860652/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/860652/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/860652/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/860652/2"}}},"commit":{"parents":[{"commit":"9af442ea7b66b7b548b69a783b7187ce78f21dd4","subject":"Update to latest CentOS Stream 8 cloud image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9af442ea7b66b7b548b69a783b7187ce78f21dd4"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-12-08 20:52:25.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-10-14 09:12:19.000000000","tz":0},"subject":"Skip IP address allocation and configuration if needed","message":"Skip IP address allocation and configuration if needed\n\nSometimes some hosts should be configured with an interface without any\nIP address set (e.g. bridged interface) and to achieve that this change\nadds the new attribute \u0027no_ip\u0027 for the network configuration. Also the\nchange contain a test for this.\n\nChange-Id: I2c9dfeca7f0d37a96f9cbd9df51d94098cf07258\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 1de4f2a4a3f6ef16399d812a544c6da622609e24)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2aadb31c150795a6ed46f8545fa8fef311397d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2aadb31c150795a6ed46f8545fa8fef311397d1"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
