)]}'
{"id":"openstack%2Fkayobe~860705","triplet_id":"openstack%2Fkayobe~master~I249ec3e3417db31d1f0d880ac9b101f88cac2fe5","project":"openstack/kayobe","branch":"master","topic":"host-configure-comments","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-10-13 21:38:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I249ec3e3417db31d1f0d880ac9b101f88cac2fe5","subject":"Update comments for host configure commands","status":"MERGED","created":"2022-10-07 14:05:06.000000000","updated":"2022-10-13 21:38:09.000000000","submitted":"2022-10-13 21:38:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860705-host-configure-comments","meta_rev_id":"9c7ebe550867866687a204a592332f5edb3d4015","_number":860705,"virtual_id_number":860705,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-13 21:38:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-10-13 19:24:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-10-13 19:24:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-07 15:19:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-13 19:24:07.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"599f530b661cb8ecb4c50ecbf4c724e3de0215ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-10-07 14:05:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"737f96b2ac0a362785cf969aed8857cc3a45db43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 15:19:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6321e9d711b14033891193669348cac9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90974953425744eba0306bc0e5b2051d : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b42d45251eb4502b2f1e1fa01193bf5 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7e9327190a741d1ac7bae3533fa79f7 : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee759b7ae01b4c52a68b27671a7642a3 : SUCCESS in 6m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b85c34a83d3e49b2a4590eb657fc1ee5 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51743ec6ba1b4c9194dc50de8699791d : SUCCESS in 9m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/243bfa90114f4ffb9065707d9690653c : SUCCESS in 6m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6a0ba22051a34444bc7ec498e83b3a84 : SUCCESS in 14m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ee40c8b5a0db4926a802566dbfd9088c : SUCCESS in 17m 43s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2ffb3d0936043bbb4cf9c0e0cd59912 : SUCCESS in 58m 40s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b1fab7c6d7a4034b324125f19d67090 : SUCCESS in 17m 49s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a135ab1fea924429b96be4ec3e494b35 : SUCCESS in 1h 12m 18s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32bad312f8db4c75893de3e54fe0b0cd : SUCCESS in 18m 05s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a8694811a2742e7ab5938d5a00e5353 : SUCCESS in 29m 49s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9ea83c206c24be2846a13f51d570290 : SUCCESS in 13m 34s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b1d87d47b8149ba800af3b60684023a : SUCCESS in 13m 52s","accounts_in_message":[],"_revision_number":1},{"id":"4a8be5c1d98d3db5c6fd94dd087e8c0ec4020ab4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-10-13 19:24:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5b9ee4c85d523703e43075c8166e780605afd528","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 19:24:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aee96ed5abfae2a4b23ff810fcebe7205b2abf2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 21:38:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff9e60f530e94d6299f899b3b52da5ed\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/414e7c6068a4440ca40add82d4f56eae : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b96cb4a675546ff93dce026b2c79b35 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae67d4e9aa8641db891782fff50bec85 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/445251e9e9164df1b00319328e81443d : SUCCESS in 8m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/02af462638f34172bce2ed79d9872678 : SUCCESS in 5m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/698c4fb81f0e4d1baa8be8a750fc5e87 : SUCCESS in 10m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/04fd865ff9b54f498d862e61bbe2af65 : SUCCESS in 17m 03s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e059adb19254ed2b8c7bff5adcbf4a1 : SUCCESS in 55m 07s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8e387e27cb1459aafe82d1340df73ef : SUCCESS in 13m 49s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59b89d8ec59f459fa9d58007a99a0ec2 : SUCCESS in 58m 19s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e5797892d1c4475939de1afe22aedc8 : SUCCESS in 17m 35s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5dfcbc318664f32859706d515850410 : SUCCESS in 29m 06s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71ae7e037c1742328a0677b4fcd686e3 : SUCCESS in 13m 11s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f5a8f2c5cf341afb416d448f02aff35 : SUCCESS in 12m 10s","accounts_in_message":[],"_revision_number":1},{"id":"9c7ebe550867866687a204a592332f5edb3d4015","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 21:38:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd46e1524ce4650a382917241704a5e77b2d021d","revisions":{"cd46e1524ce4650a382917241704a5e77b2d021d":{"kind":"REWORK","_number":1,"created":"2022-10-07 14:05:06.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/05/860705/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/860705/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/860705/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/860705/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/860705/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/860705/1"}}},"commit":{"parents":[{"commit":"b3cc98d3028d98a49693723db565d1f65003f522","subject":"CI: Set previous_release to Yoga - part 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b3cc98d3028d98a49693723db565d1f65003f522"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-10-07 14:03:55.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-10-07 14:03:55.000000000","tz":120},"subject":"Update comments for host configure commands","message":"Update comments for host configure commands\n\nChange-Id: I249ec3e3417db31d1f0d880ac9b101f88cac2fe5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd46e1524ce4650a382917241704a5e77b2d021d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd46e1524ce4650a382917241704a5e77b2d021d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
