)]}'
{"id":"openstack%2Fkayobe~860709","triplet_id":"openstack%2Fkayobe~master~I7c863d1875908d2b885918ec7caed747ae6e345b","project":"openstack/kayobe","branch":"master","topic":"whitespace","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-10-13 21:38:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7c863d1875908d2b885918ec7caed747ae6e345b","subject":"Remove trailing whitespace","status":"MERGED","created":"2022-10-07 14:24:05.000000000","updated":"2022-10-13 21:38:07.000000000","submitted":"2022-10-13 21:38:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860709-whitespace","meta_rev_id":"d89c6b7a5a1ff3ba91b752be0a5dc6ae30e1aa75","_number":860709,"virtual_id_number":860709,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-13 21:38:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-10-13 19:23:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-10-13 19:23:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-07 15:36:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-13 19:23:39.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"17fcba376532f11960ccf00ca58de921999b02b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-10-07 14:24:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"39c0996fd2b8d848d455a29c332c09a1ed54f9e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 15:36:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/575688c103dd45dea37be3d5a6f9229b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08e286feac4442cfb444964c91796093 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b66269619ce1475f980f06174302ac03 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6976d3e5d69c4c2f9d0df3d81a286fd9 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5501fc5b9cae411283bcacda76fc4781 : SUCCESS in 5m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e8f30c15d86456c9dab1e3749c6bbd5 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/540c4bd99c604d0ab272830c2dd98181 : SUCCESS in 8m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1344c47b8ee2432187a23fe9598991b1 : SUCCESS in 5m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f3ce06a1eb5f49fcb1a5e52b9093a815 : SUCCESS in 15m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/04fc914895044747b5a79da63d534754 : SUCCESS in 15m 15s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a779bacff7d544a2a3af6bf0c21d7eba : SUCCESS in 59m 36s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9e7ce9e14a54122a660502d41775e85 : SUCCESS in 15m 44s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44010b7c28054c3c95450a692ee08512 : SUCCESS in 1h 08m 20s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33aa233107a8402e8f7b28559d3af44f : SUCCESS in 16m 05s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86b520882b82454e9e71388b82322064 : SUCCESS in 26m 35s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/750876748f6f48cbbf7e54186b113cc4 : SUCCESS in 11m 51s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff3f0e6624a941ba99e53efa7c1f5ebc : SUCCESS in 12m 48s","accounts_in_message":[],"_revision_number":1},{"id":"265a5c3e36dc43ec00573fcc212b889de616a26c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-10-13 19:23:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"28aa111eee22fc44fccd6c8f58c5da1b15b61ee9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 19:24:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"efb9c69196aebf7028979536aff08c0d0cdfab6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 21:38:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/874bf9d364924db4962dc5d9f045738a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/244d6c36b8b34004984ee5b6bc28b241 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a1375e5577c4f50800b7521f407f01b : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebb8f9c6ac2d4dd7a549abbf39b1fc20 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c19b374036fa40498488d152e5c60ec1 : SUCCESS in 8m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d20f435b6fc94554b796083a64952249 : SUCCESS in 5m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b9afa6fa6ae24b9c9e92a70dc2c7d975 : SUCCESS in 15m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cf991a074bbe4c199597eaf71e59562c : SUCCESS in 18m 38s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c8946cc57e343d5b62b9f0dcdeccde0 : SUCCESS in 55m 41s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee976e81cbbc43ea96f729173431959b : SUCCESS in 18m 22s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1e440c10be8478aba09883896c13487 : SUCCESS in 1h 04m 26s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db802a59899d4665bf2844f8eee9c52f : SUCCESS in 16m 24s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/993e559bcf7d445e90fef5b3a0357caf : SUCCESS in 31m 17s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad8cffb679254626824b351c2fcf051c : SUCCESS in 12m 17s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eed43c8bc6424b3eb3ce83e34d8fee81 : SUCCESS in 13m 38s","accounts_in_message":[],"_revision_number":1},{"id":"d89c6b7a5a1ff3ba91b752be0a5dc6ae30e1aa75","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 21:38:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69ef2b3b3ea113b0c920f23cc6985a4831b235c2","revisions":{"69ef2b3b3ea113b0c920f23cc6985a4831b235c2":{"kind":"REWORK","_number":1,"created":"2022-10-07 14:24:05.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/09/860709/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/860709/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/860709/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/860709/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/860709/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/860709/1"}}},"commit":{"parents":[{"commit":"14a04a69cc8d390ad48f804ec495d8fef383b308","subject":"Merge \"Add prometheus-msteams group to kayobe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/14a04a69cc8d390ad48f804ec495d8fef383b308"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-10-07 14:23:56.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-10-07 14:23:56.000000000","tz":120},"subject":"Remove trailing whitespace","message":"Remove trailing whitespace\n\nChange-Id: I7c863d1875908d2b885918ec7caed747ae6e345b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/69ef2b3b3ea113b0c920f23cc6985a4831b235c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/69ef2b3b3ea113b0c920f23cc6985a4831b235c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
