)]}'
{"id":"openstack%2Fkayobe~861397","triplet_id":"openstack%2Fkayobe~master~I4813a1f6ab9cb566596e806bd0ada6dff342d119","project":"openstack/kayobe","branch":"master","topic":"kolla-passwords-overrides-issue","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-11-08 16:55:53.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-11-08 16:55:53.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-11-08 16:55:53.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-11-08 16:55:53.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-11-08 16:55:53.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2023-11-08 16:55:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4813a1f6ab9cb566596e806bd0ada6dff342d119","subject":"Fix an issue when user forgot combine custom passwords","status":"MERGED","created":"2022-10-14 15:02:06.000000000","updated":"2023-11-08 16:57:14.000000000","submitted":"2023-11-08 16:55:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"861397-kolla-passwords-overrides-issue","meta_rev_id":"6f62c47800c599e8a472f16ec1281cd6c58e18ce","_number":861397,"virtual_id_number":861397,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-08 16:55:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-11-08 14:13:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2023-11-01 12:36:25.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-11-01 11:58:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2023-11-08 14:13:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-11-01 11:45:01.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-14 16:12:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-14 16:16:43.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-10-14 16:16:43.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-04-20 10:04:44.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-04-26 13:40:32.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"CC"},{"updated":"2023-05-25 08:44:24.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2023-06-07 12:12:19.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-11-08 14:13:17.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"546f73716f2a83169d7ef75f34ba3aa855d74f04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-10-14 15:02:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2c122bb2e582928c6aa5518c2a1b72b12154eef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 16:12:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de3eaa2cf5f7494bafa878a7b97df00b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecfc3fd4ede04f909834786f154dca70 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dae6c17d58cb4a92bb46e96b117331d8 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd27a72188f743d294e9d41388c308e2 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e47430fd9ec649bf88ce0f1fc314e03b : SUCCESS in 6m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11e0794ad71949fe88cfb399fd35decd : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d7376627faa4358bd2ff5bb63116150 : SUCCESS in 9m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/baa7f8df71034933b217c3463a7789d3 : SUCCESS in 3m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/766562dd2e99455787d14c7b86fb8abb : SUCCESS in 6m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5278e2c40ed3476d8cbea7881a4122a1 : SUCCESS in 14m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/811d921483be4bf088ec499a08994b65 : SUCCESS in 16m 34s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c34f3e9fbe7149abb5d3af5c8844bae3 : SUCCESS in 51m 42s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2c4d732bacd429d98c36d93c7785feb : SUCCESS in 17m 04s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45965cc200164a1b827ffbdaf75d346b : SUCCESS in 1h 06m 13s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b960ef99565a4d108d5c545bf5c3ed42 : SUCCESS in 15m 41s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c152db2223f4676a1135a8b0ed67669 : SUCCESS in 26m 33s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1427f3e97d3b4fc6a9deac195a335f2a : SUCCESS in 12m 52s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8dd316a2cd3a42fa89284b5a1badd091 : SUCCESS in 13m 11s","accounts_in_message":[],"_revision_number":1},{"id":"c67370f55f2d486db3867c944e70ab621d8adc2f","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-04-26 13:40:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"732fff4dd307443554b4b5439314e2cdbcac4225","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-26 13:42:25.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e6f2647f3cf9d0003c91c1d6842d3192c3782688","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-26 13:42:33.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"8b23a70ba27321215b4b59e76a4d5e9847fb6110","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-26 13:42:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9a14ee18b25c17d84b9f728b5f88245c80929a69","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-04-26 14:36:38.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"785338564d1c86fec10ae5da59c59fc711d1f810","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-04-26 14:37:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"96ad68bf6499c010ed17c9d7df38ca17ba3a2ee0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 15:23:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d41780afc3ce47c09c8f5694446afd92\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0a236045d134a2ca3f2c3f8e544d614 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee4f4e1062174aa19602df7aaa741801 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee1d5accc2e84ad0bd09c2f608d00ba1 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26b49b1d908346f699e8a82a7512c45c : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cba4e4d8f2cf40be833e13ff190b6bb4 : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e21f52be6e3445b8b10f5dd791108549 : SUCCESS in 2m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/024f551786da478e8b6bdf4264ef9b80 : SUCCESS in 5m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/49b92752f70d44ce8152dd5cbcd0e0bb : SUCCESS in 10m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6e9fdbf707c243d5a1f55c853dce6f9f : SUCCESS in 13m 03s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/69be5213a08e40598b97ba7239882e1f : FAILURE in 42m 52s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/cbf5b93474ec4cc89a298052de61e926 : SUCCESS in 59m 13s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8267edcf373c43e58d3a2e89d406c16a : SUCCESS in 52m 09s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/93ce4022756a45999477f18ecd582079 : SUCCESS in 57m 21s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/61a1411c14374a78a2ddb6a81454ce1a : SUCCESS in 15m 08s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/3d8138828630476e9942d5ada3b0fdc0 : SUCCESS in 18m 44s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/223818fc3c4c4494bb53a805fdd5f5ad : SUCCESS in 14m 21s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/335d118582c44eeda2ae4fe5ebd00b54 : SUCCESS in 1h 39m 37s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b9c48d0652f491db51db9fe43950b4f : FAILURE in 37m 59s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3453c53a1cd14d15b42c9b5b04595ebe : SUCCESS in 16m 13s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b072349e658442a895d96859e5bf739 : SUCCESS in 13m 56s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/936869d28a484e61a8ff17b19f542f34 : FAILURE in 31m 16s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/44096660fa4d4769ba4956ece415d83a : SUCCESS in 26m 47s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69958840ca77479784448840efd43339 : SUCCESS in 20m 56s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1353406e08a544bf815ec6487db78fd2 : SUCCESS in 13m 51s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f5148b3d4f34584aac7f264e5ce84b6 : SUCCESS in 11m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f4e0c8b72b3e43e99bfbbb5db7b02742 : SUCCESS in 15m 40s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c9ee168fee04f0182f93c24f1737545 : SUCCESS in 12m 20s","accounts_in_message":[],"_revision_number":3},{"id":"05851c21963f9488eae3fb7d6e97ad5297d5017b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-26 16:53:12.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a6e2ba523ebece1d5780d57fe397d9e3bdf1b9f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-26 16:53:14.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"ee255de0ebc49b6b3f417f90a20090e4771015d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 18:15:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/80bb78e9d21b4df4a50e92fa9013ab78\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/004c4dfdb70e4ecd87eb3d8f0e3eb6e9 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85a9a811ed84464e94e0f742b89a0ae7 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/992209577ff6465aaf496afb2ed0232a : SUCCESS in 6m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bdbaf6c084a64fa8888cacf591a491db : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d59286615c3c48b6b636d26f8fef868d : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/840b48e64840455f99eb3892d14a26cd : SUCCESS in 4m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/92cb87181b844b408de7a2b5877cc8d3 : SUCCESS in 4m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b1f48840e0ff4df79e5103349d863042 : SUCCESS in 12m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/584a4211e5394d26abb11fd441a8f6d0 : SUCCESS in 13m 32s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a60f497c81094bdfbef99db9959fd0f2 : SUCCESS in 1h 02m 17s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/4ae19df58f3e43ecb8e6aecc19270032 : SUCCESS in 1h 03m 15s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af5e3e179015426790f240ce854f20f2 : SUCCESS in 41m 17s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/edf8462d655e4570bf90cdfb90172111 : SUCCESS in 1h 04m 41s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c9741bb4a1a7442a8224f8a59ba7fffc : SUCCESS in 16m 49s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/1fbc45b77b5f4a41a18d46737ece07a4 : SUCCESS in 17m 32s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7612c4ae04d4aa68b64b0fa4828259a : SUCCESS in 12m 58s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a271c06a8f2e417baad1426497405e0e : FAILURE in 46m 25s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3487f532686e4fca8b337fa5015e22cc : SUCCESS in 55m 34s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/31b878671729437cbd24923167516f97 : SUCCESS in 15m 42s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28182b3640194c47b98486b996e52f43 : SUCCESS in 13m 38s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f7b7b17bdb0f4c9184e56406aa6dab2d : FAILURE in 28m 17s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0cc676bf906c439f859072093891ab9f : SUCCESS in 22m 08s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8378151c9cf43cd97e3aa7cd89d7fac : SUCCESS in 22m 01s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2c4b7d884b0146988e735991c8ec38e3 : SUCCESS in 12m 36s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1fd9e6c96cfe4734ba04f0f4355494e3 : SUCCESS in 10m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fdc81b46812841259ac8767c31e1cd3c : SUCCESS in 15m 07s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c179ecbddf0d4c3e9f33f3468a3b6f74 : SUCCESS in 12m 14s","accounts_in_message":[],"_revision_number":4},{"id":"eaeecc73d08858cce47fd4a2e84e3928619f7aeb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-26 18:20:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1dfcaf01291dfc6a85820d31ceb9b65d91637af9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 19:52:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45b8e6df97c443afb8908f96de382ddf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8de82af29ad04bab8f97102b6cf94927 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d49dfbbc4504966886dd79815a32798 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/884d55e715f9431b9a0a5bf64342464c : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b67443efbcaa4fe6a1938acb8826724d : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6df4d10cbd947a1b1e5f1e58ec5390a : SUCCESS in 7m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebbb1b6a62384768b1977d1f99fe0e82 : SUCCESS in 2m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aa50af2aea084379a20fc02b9e34ebbd : SUCCESS in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/68e9e25880464426b9cc104cc93471ad : SUCCESS in 11m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/366f2e28bf95486789892eabbdf456a3 : SUCCESS in 14m 13s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8dfc3778e1604b46ad53ccc20608b5d6 : SUCCESS in 57m 22s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/13ab72cdde324ba68ca7bcf4095e1b0d : SUCCESS in 1h 06m 13s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8edf4f2106404a56bb563ead35f29136 : SUCCESS in 50m 04s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f639685e7a6f40c9909b5d1ca39b2221 : SUCCESS in 57m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f9d8d62bc1aa45788a1b3f4fc78e2015 : SUCCESS in 15m 35s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/bbb80e1d8db34822a59c135d8c7455d8 : SUCCESS in 16m 42s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5c53b66e70f480b9fe83d309d73e2a9 : SUCCESS in 12m 39s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/63758b5edcb34877938c313a87f70feb : SUCCESS in 1h 20m 37s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eaea9e64c26346dc8f8fe258ba59b9f6 : SUCCESS in 1h 09m 56s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/53ce6a4204bc41b88fe0a686c7297de5 : SUCCESS in 15m 50s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70fc27a17a9e41c69a57bc4062a12f09 : SUCCESS in 10m 24s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/9a970afaa1244822b366f3b5404ccc86 : FAILURE in 25m 32s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/939ec5d61a0f4cc4a58d90521dc7b0f9 : SUCCESS in 24m 04s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf10469ac5f349669dcd6cea385cec99 : SUCCESS in 20m 21s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/680e21a4166c4f75a74148968e318425 : SUCCESS in 11m 51s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a08ea299fe9340acaff215c5c9d24ff4 : SUCCESS in 10m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/167f7d8f71bc4d368e95355dda26cb58 : SUCCESS in 15m 22s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c372404b3174971907e9014e3bf9ab4 : SUCCESS in 13m 55s","accounts_in_message":[],"_revision_number":4},{"id":"526b5f598662b86b6451a212c6606de09f8a8eec","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-27 08:26:03.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"00516bfacf9da6ebb57f6a800537fd309658a0dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-24 18:12:12.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"0bf6dfff4d4dd6206eee2bfeb7c347b22ab28167","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-24 18:12:52.000000000","message":"Patch Set 5: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"35b50ee0fe7666f7fb5a147e60dac32f434c5c2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-24 18:28:30.000000000","message":"Uploaded patch set 6.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n","accounts_in_message":[],"_revision_number":6},{"id":"23fd8c00818012419a17027463c9612ac855e1bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 20:01:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4248f2a41c6f40dd88e377c39772414d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/187b7fec5fe74ab280334cfd1311306c : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c9f8178acba4388a5a04bb4a2f56719 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be74fe4ce3b944a7812dc0bf38bd8a6d : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c14c13ea57e436db8e2feaf40731935 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c782008e233a428fb9f7165dc2bae08e : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53f1d5f30fed46339208fe3ff577e894 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44a17789d24c40f0b25f0cf62fa4e535 : SUCCESS in 3m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/883473546dac41e88aa3602ab6286192 : SUCCESS in 3m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/32905c62f2bf4dc7a56e57da9a541615 : SUCCESS in 9m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c4b844b89f884d7a8736785bd5258de0 : SUCCESS in 13m 37s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a356eb27ce6240dab18e7110554e766a : SUCCESS in 50m 47s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/0ce0374d8fcf4f72a76ccd7a734dd358 : SUCCESS in 52m 52s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/174f78ff1e7a479abfbc3daf49419dbb : SUCCESS in 51m 33s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/393979d4651d49ea92102d760a98c5eb : SUCCESS in 58m 02s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/bb3da0e028e14b29b18df80566ecab73 : SUCCESS in 18m 59s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d563f2f9ad764afbadb4c7cb3708d1cf : SUCCESS in 15m 10s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82ab54c146784e8981bc8cc30333e333 : SUCCESS in 15m 17s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/831471fd05524c8ca68993a64ee9171c : SUCCESS in 1h 11m 17s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/091e638019ba47b4a490d2fef94a4659 : SUCCESS in 1h 30m 32s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/baca1542bcea4b6692bc9d86cbdc14a9 : SUCCESS in 15m 03s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/810f48a379d14de181fdd48f9802cd65 : SUCCESS in 13m 34s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6e6c01cad75d4dd7966452ad7fc9b21b : FAILURE in 28m 58s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d48505c478794a6d9cfe0e43f6eef41f : SUCCESS in 25m 11s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3e978d6f3484bfca245a31ba56aca7f : SUCCESS in 27m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6fdecde1642e42b9a0af5db5fee3e0d9 : SUCCESS in 12m 07s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1ed7757adcf4e2bb8c90dcfd057ab60 : SUCCESS in 9m 56s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b5cb9b49e0844a38a2035d4b4aeea133 : SUCCESS in 16m 14s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0cfd0f7c5fec4d2e89cc2c491d114726 : SUCCESS in 13m 14s","accounts_in_message":[],"_revision_number":6},{"id":"e0058f680456ddfe4ec6dd28b20190528692b53f","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-05-25 08:44:24.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3b616d4e2740d808409bba6b2e95c07bccb03845","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-25 11:40:08.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"5fb3d621cf0c992eedef7af333305f1bf9bf81ce","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-25 11:40:35.000000000","message":"Patch Set 7: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"546ac3b63f212954ab799862a7722d910113b684","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 13:20:56.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f61568ff87c4dfeaacfe317fea67f25\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3f9ad6d46ec47c8933529d2e7a1f320 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5daecb3de5f44fc8a387e6dd6c4f521a : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2852b4fd5e86417ea859f13a9056c806 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec0316d2b84c46b091f63eddc8edd7a5 : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8dc0c2b61d2147a284796e12b08ac617 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d752a668c8e64ca09a8bef67afc11329 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/258b07ecf1654183878492ab9d4018f7 : SUCCESS in 3m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/706e5dac07d741d29629cc29c994db58 : SUCCESS in 4m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/21b15f69ae31459ca312501f0336669d : SUCCESS in 12m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c4c42743ba554baa9116765cbbb568a2 : SUCCESS in 13m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/88f33a94595e41958cf83ab5d830df19 : SUCCESS in 52m 53s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/3963b62897784a68991055d683530bfd : SUCCESS in 50m 14s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/183454b735134c708ae2767d8fb8f973 : SUCCESS in 43m 38s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2652f8ad08b64e5c96ae17d9014f6077 : SUCCESS in 54m 35s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7a1522addaf7418ba2c0cfe87b3eab89 : SUCCESS in 14m 00s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/bd1bbb4e11c54549a616fdc8a550667c : SUCCESS in 17m 41s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6477d446aa74c23a339509a9f43a9bc : SUCCESS in 13m 12s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e99bab89aad8421a8494828d00156a94 : SUCCESS in 1h 09m 34s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1100236a541f42a7827294ae1c37c78f : SUCCESS in 1h 28m 09s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/bc73c026ea8246dc803148e5b2e060ab : SUCCESS in 15m 40s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db1e6500907542d08bbc019c2166258b : SUCCESS in 13m 54s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/706e4f419eaf4b79b54c44e864a9458a : FAILURE in 25m 03s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d19832900bae433e835dc5bacd562dd8 : SUCCESS in 27m 21s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d275fded2124624bb4ad8a23486d593 : SUCCESS in 25m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/97a822829f284090969aa66c6e361673 : SUCCESS in 13m 23s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b03c516a775d4b4892c8bd3f51732ef4 : SUCCESS in 10m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/321c605a9f4f4923a99e96786153d86f : SUCCESS in 15m 55s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c892e930affb441693815b9209b02c21 : SUCCESS in 13m 55s","accounts_in_message":[],"_revision_number":7},{"id":"c4f5c9e3c39ad5d03a2237f814ac3425a946f533","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-05-30 13:48:05.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"ab6b190ad563dd40cf9b23e4eea0329afd80f59b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-18 14:11:36.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"94c30ca5eb75b1b4cc12e0f29c1b7ec2f6a96f10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 16:18:43.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d233e3b279de46019b77b4086517ebc7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0f29866aa644bceacc17bccc7182fe4 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/911479367a274b5f8739e0011abd55ad : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d667852e479458a91d71b0ed40e8910 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1778631ae6a847b2ae0853378fb8fb52 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc45311689264c9cafc628bf6d6dd20f : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7733eb360b2b4da8a7b78c67bda3dcef : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80f5fb11f0b64fc0bc41b167ab246756 : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a114f0c93a84982a5d16b1a34786712 : SUCCESS in 3m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5111ea829d5848e1805817969f9aa626 : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2c73655e1e8c4601b2b9836075b74870 : SUCCESS in 11m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7473d7e26ff549588e1fe889cf6998a2 : SUCCESS in 13m 50s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6f8e16e1bd2d46f8940e7e538a761662 : SUCCESS in 50m 38s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f91e956124e141d1abb65ddac3243c94 : SUCCESS in 41m 15s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14abee938903413fb5bdf73fd6c9d2f4 : SUCCESS in 46m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a21b75ca32c24c449d781d6ee0287d54 : FAILURE in 49m 13s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0949b3230fdf43c8baf3e5d8eb69fb49 : SUCCESS in 18m 34s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/00e6ecbeca0e4127be148ddb320329d9 : SUCCESS in 16m 24s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/387d842f2a2d48a99d5457584831904b : SUCCESS in 13m 15s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/434c574a594c45368551e24d6f5390be : SUCCESS in 1h 51m 02s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56ec68be972f4155981eb7b687054f42 : SUCCESS in 1h 55m 09s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5f76451303084a558f9f05d194af8a58 : SUCCESS in 14m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c37aceedb0124e359c5afd8de104eba5 : SUCCESS in 13m 07s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/705ba430ec3b49c3982971bd19661b1d : SUCCESS in 34m 51s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/008b943dd3fe476993b8f9e3c74835f4 : SUCCESS in 29m 09s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4e33cc532954ada88bfec83beeadde7 : SUCCESS in 24m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/72f4dd8316f549678164d44c05ef732a : SUCCESS in 10m 58s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9d39bfb7da1429fb71fbe414a541acf : SUCCESS in 9m 10s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8260edc7277e41bf8dcaef73720f24c7 : SUCCESS in 31m 40s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/841bb38c05e646dca2e82576107ff7b9 : SUCCESS in 27m 11s","accounts_in_message":[],"_revision_number":8},{"id":"dbd732025cf014429b07a4840a0224847c8967c0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-18 21:05:27.000000000","message":"Patch Set 8: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"74f2ca7343d1c4039e48dd5ab60c22bf754d53ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 22:52:50.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f7f0d0f15fc4530b4cf4f29669d7120\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/631477692dfc4344937479d059532083 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32099e3d25f645958095a1b7a3c1163b : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59e0ba3306b94c8180340f382e2a5a4d : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33ac0600a91149b48feff6050816857f : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca6ee8a992a34c00bbf05c84a2475233 : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c53d9109c2a84b748d7abf53da3f80b5 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e03159b25ded44649a5f4def1bf48bad : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c121d71538c849399a286d873db87062 : SUCCESS in 3m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2bbe55d22bd945fb9fdca9edc06917fc : SUCCESS in 4m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d0f6e2cede984370bfdda0c1b464e96b : SUCCESS in 11m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3a20201d6c7c4330ab60805c3731bd9e : SUCCESS in 18m 49s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0a2f83916b1d4aaea276271f45d100f1 : SUCCESS in 57m 06s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/4c61db56bc434486b9488779c140b975 : SUCCESS in 1h 02m 27s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7647bb85debe40d381a6b1a8f4ac6838 : SUCCESS in 53m 13s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/48af915591c94dfa97930603c8105acd : SUCCESS in 1h 03m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/08e1b04ebccd4b619696d3361983464e : SUCCESS in 17m 33s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/6a0223bc820f48ffa82e03c537d369cb : SUCCESS in 17m 15s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c58beacd995c41b4a8c548e05cd3cafa : SUCCESS in 11m 12s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/14fd5d240edc441cb6c6f85359485d7b : SUCCESS in 1h 41m 50s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/588a53d9160944d888b9769be87b6f7e : SUCCESS in 1h 24m 33s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5d03027e94cb46fb95c41301348f94e1 : SUCCESS in 16m 47s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0aeda9bfbaf40fdb7e183f1c38f0216 : SUCCESS in 11m 44s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/96992fc2080e47c39364e03f202f6847 : SUCCESS in 31m 11s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/929aea79fcd940c5867c152b969dfe57 : SUCCESS in 29m 08s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a99a89952e874c358f01f3fe7cc858fa : SUCCESS in 19m 14s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0a1ed0e1c2094a119cab1b6fe70cf217 : SUCCESS in 13m 04s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ea38c7e355a4cd3bd84bf4e5fcf189f : SUCCESS in 9m 06s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b8d7f842c842402b8dfc9d81d0cfae65 : SUCCESS in 34m 07s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d7bcf87e7c84b8297aabb15e4c73c8e : SUCCESS in 17m 58s","accounts_in_message":[],"_revision_number":8},{"id":"41ca9d3620999bdfc89f215f466dca9a502c8052","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-30 13:33:46.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"a78ca2eb8def239ef92d0a17c7d310af351129fc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-01 11:30:53.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b8f7f6d0d70a3dc887cdf3d27a02e0fcb375db90","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-01 11:31:08.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a3de965ce4f00026a444a8f860820880bc0217fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-01 11:31:12.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"fe181cf32a65d659d4f7b6d32dce11557b4b3de0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-01 11:45:01.000000000","message":"Patch Set 9: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"9584b74dc951d4a6550f15eba939089c7a33bdf3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-01 11:58:02.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"475be53fb431f6e2c8c80709f09222a943ba1b71","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-01 12:36:25.000000000","message":"Patch Set 9: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"862c31a11ecff0663ae0662669d0cdcb4611574b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 14:39:07.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c920ab06449041a8b4805638374b2dc8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5d06e167c8c4d05b765ddaf3da52f33 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f162be4b4f474596a0928a51bcbf2d8d : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/640cc499fc4a4fcdbcf43fac2fc3498c : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4c7e1abb6ebf48c787169a62aa66e5d1 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf253a8159b0432ba4256f14d56d24cc : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5df79fc69e04aeebff62a30c5df4cd5 : SUCCESS in 3m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/54356de95f434d0593f1f7dc9f3c9e0a : SUCCESS in 4m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fbf0c6170d0b4c8aaef089d6d7b6b08d : SUCCESS in 11m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7cdcbcbc9a3944ddbe0d5e9dd53b98f6 : SUCCESS in 10m 24s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/dd7791e76e0b446aac149345024219aa : SUCCESS in 1h 05m 38s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/76a7df09a93443f0abf95cbfea974f74 : SUCCESS in 46m 35s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da344339111d4bc1abfb0af149134795 : SUCCESS in 51m 48s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/bacc4181c7c74d779e74b66e666b7eda : SUCCESS in 1h 09m 37s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3c50db6cbbba4098afbc29f77141eca3 : SUCCESS in 18m 08s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/46d7e75dbacd4cd6a4d6db45e19d5c3e : SUCCESS in 13m 02s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a249d0bf708f4e0aacfa5f62dc44d6ff : SUCCESS in 12m 49s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6f42d2cd37534b828405a4ba6b812a56 : SUCCESS in 2h 30m 39s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b6d1905e96247bbaf37311a5442e275 : SUCCESS in 2h 15m 43s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d50ed1ace618428d860cc60961aa2250 : SUCCESS in 14m 31s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4eb876736e3c40aca209a8c3f4473d98 : SUCCESS in 13m 57s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/422430d42e504c669bbd50c8d72d06d6 : SUCCESS in 32m 33s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/257c81cdbb734e30bb445e8f2aa4465f : SUCCESS in 29m 38s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f83eb025143423b9f8e3f700546853d : SUCCESS in 24m 37s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c795a8043db44c77a644f21599c25822 : SUCCESS in 9m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e19dad3b1864a5d8c8590365feb97c5 : SUCCESS in 8m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fb3d96612bfc4f1e981985819eb86320 : SUCCESS in 17m 26s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b10656273a45493f8be0e04b970af8c6 : SUCCESS in 15m 49s","accounts_in_message":[],"_revision_number":9},{"id":"e39945c9a14087c11e9fd23798b6d172bb2f9ce9","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-11-08 14:13:17.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"eb2d74be0bdd505bfdaa7a9a92bf00e8f71420a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 14:13:47.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"13e4ad2544da21e705eab5800edb13cd924ec31e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 16:55:53.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9a9b965afbf48a280db90751afc1e56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd441c17beea461cb9a3c70dc308dec8 : SUCCESS in 7m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0092a2436c814a3f9d8d50f34344a073 : SUCCESS in 3m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5e87eb3cabe2486caac2420cd81062e2 : SUCCESS in 5m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8ef761ce16fd4291ae899f816663abc7 : SUCCESS in 12m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4f98d72fe2454a359d99d1af43423559 : SUCCESS in 10m 24s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3b401e970bbf423d976ab0bb876607fa : SUCCESS in 1h 03m 07s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3f7287649554190bbcdcebe0ccf35fb : SUCCESS in 52m 45s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/af2fafc0a30c4cd5bf80f130020ccc1e : SUCCESS in 1h 05m 31s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/be614ea2697b4783a0250def3a2b9078 : SUCCESS in 19m 37s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e2ebcbc5fa942fc86526a2d8f2d8589 : SUCCESS in 14m 13s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b9fba5f61def4b7b9d85b6ab0928e29f : SUCCESS in 2h 24m 09s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af356dfa0d6840f49966aef27a67f8b3 : SUCCESS in 2h 33m 52s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3d0df0842cee43a19e4fcfde7976b6aa : SUCCESS in 15m 50s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1cdc98ad34a74e718de3212bd4db2021 : SUCCESS in 14m 28s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/40fe8b792984480f8ecfebf6c1cb176f : SUCCESS in 30m 52s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b49e22c9a104e129db4e6d03c5416c6 : SUCCESS in 26m 57s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b41ac450ce994809b8059a0fac96ccb2 : SUCCESS in 13m 22s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/844f4c3af3b84845adc50c386e53ea03 : SUCCESS in 11m 10s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/61aba63b86634418ae3bf169d1f5f6db : SUCCESS in 16m 29s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e5b5b9c2e424163bfa38e09002d409f : SUCCESS in 15m 19s","accounts_in_message":[],"_revision_number":9},{"id":"9a27705cecf2ec860fbd7ddd7aa09a9c2a6ab326","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 16:55:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"6f62c47800c599e8a472f16ec1281cd6c58e18ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 16:57:14.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/192a0f1a0d1744d4925c93d69a6e2f33\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cb9c430f3d7462db65ab1a5a41b53cb : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5681b17f5df945c0ada49f0634a0f9c8 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"73abf0e57ee3aaeefa53eaf4fb7b1ed5f5c96933","revisions":{"ce5cf58e5db0e41c40fc606f42167487a7e83b01":{"kind":"REWORK","_number":1,"created":"2022-10-14 15:02:06.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/861397/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/861397/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/861397/1"}}},"commit":{"parents":[{"commit":"18ff3195fc30c5fa00e6b23071b4bafc60213cbe","subject":"Merge \"Fix bug trying to decrypt a plaintext file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18ff3195fc30c5fa00e6b23071b4bafc60213cbe"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-10-14 14:56:20.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-10-14 15:01:03.000000000","tz":180},"subject":"Fix an issue when user forgot combine custom passwords","message":"Fix an issue when user forgot combine custom passwords\n\nFixes an issue when user forgot to combine\n\u0027kolla_ansible_custom_passwords\u0027,\n\u0027kolla_ansible_default_custom_passwords\u0027 and own dictionary with\ncustom passwords in configuration files. Now\n\u0027kolla_ansible_custom_passwords\u0027 should provide only user custom\npasswords to add or override in the passwords.yml.\n\nChange-Id: I4813a1f6ab9cb566596e806bd0ada6dff342d119\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ce5cf58e5db0e41c40fc606f42167487a7e83b01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ce5cf58e5db0e41c40fc606f42167487a7e83b01"}]},"branch":"refs/heads/master"},"d0256c699aa9cf76dce12783739062b7b3f24885":{"kind":"REWORK","_number":2,"created":"2023-04-26 13:42:25.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/861397/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/861397/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/861397/2"}}},"commit":{"parents":[{"commit":"18ff3195fc30c5fa00e6b23071b4bafc60213cbe","subject":"Merge \"Fix bug trying to decrypt a plaintext file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18ff3195fc30c5fa00e6b23071b4bafc60213cbe"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-10-14 14:56:20.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-26 13:42:25.000000000","tz":0},"subject":"Fix an issue when user forgot combine custom passwords","message":"Fix an issue when user forgot combine custom passwords\n\nFixes an issue when user forgot to combine\n\u0027kolla_ansible_custom_passwords\u0027,\n\u0027kolla_ansible_default_custom_passwords\u0027 and own dictionary with\ncustom passwords in configuration files. Now\n\u0027kolla_ansible_custom_passwords\u0027 should provide only user custom\npasswords to add or override in the passwords.yml.\n\nChange-Id: I4813a1f6ab9cb566596e806bd0ada6dff342d119\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d0256c699aa9cf76dce12783739062b7b3f24885"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d0256c699aa9cf76dce12783739062b7b3f24885"}]},"branch":"refs/heads/master"},"387c049e218136acb15d7ceada4198e218f4ffa5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-04-26 13:42:33.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/861397/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/861397/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/861397/3"}}},"commit":{"parents":[{"commit":"9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac","subject":"Merge \"Revert back to MichaelRigart.interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-10-14 14:56:20.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-26 13:42:33.000000000","tz":0},"subject":"Fix an issue when user forgot combine custom passwords","message":"Fix an issue when user forgot combine custom passwords\n\nFixes an issue when user forgot to combine\n\u0027kolla_ansible_custom_passwords\u0027,\n\u0027kolla_ansible_default_custom_passwords\u0027 and own dictionary with\ncustom passwords in configuration files. Now\n\u0027kolla_ansible_custom_passwords\u0027 should provide only user custom\npasswords to add or override in the passwords.yml.\n\nChange-Id: I4813a1f6ab9cb566596e806bd0ada6dff342d119\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/387c049e218136acb15d7ceada4198e218f4ffa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/387c049e218136acb15d7ceada4198e218f4ffa5"}]},"branch":"refs/heads/master","description":"Rebase"},"a657b8efc7752a8aa11a9899beb1ffe5149649e5":{"kind":"REWORK","_number":4,"created":"2023-04-26 16:53:14.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/861397/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/861397/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/861397/4"}}},"commit":{"parents":[{"commit":"9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac","subject":"Merge \"Revert back to MichaelRigart.interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-10-14 14:56:20.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-26 16:43:43.000000000","tz":0},"subject":"Fix an issue when user forgot combine custom passwords","message":"Fix an issue when user forgot combine custom passwords\n\nFixes an issue when user forgot to combine\n\u0027kolla_ansible_custom_passwords\u0027,\n\u0027kolla_ansible_default_custom_passwords\u0027 and own dictionary with\ncustom passwords in configuration files. Now\n\u0027kolla_ansible_custom_passwords\u0027 should provide only user custom\npasswords to add or override in the passwords.yml.\n\nChange-Id: I4813a1f6ab9cb566596e806bd0ada6dff342d119\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a657b8efc7752a8aa11a9899beb1ffe5149649e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a657b8efc7752a8aa11a9899beb1ffe5149649e5"}]},"branch":"refs/heads/master"},"66eaee56625edef95c63ab560d63a7b131c6b49d":{"kind":"REWORK","_number":5,"created":"2023-05-24 18:12:12.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/861397/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/861397/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/861397/5"}}},"commit":{"parents":[{"commit":"9b58320f6b40de3c33bee179d97dd0bf5bf61986","subject":"Show an error when *_bridge_ports is not a list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b58320f6b40de3c33bee179d97dd0bf5bf61986"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-10-14 14:56:20.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-05-24 18:11:25.000000000","tz":180},"subject":"Fix an issue when user forgot combine custom passwords","message":"Fix an issue when user forgot combine custom passwords\n\nFixes an issue when user forgot to combine\n\u0027kolla_ansible_custom_passwords\u0027,\n\u0027kolla_ansible_default_custom_passwords\u0027 and own dictionary with\ncustom passwords in configuration files. Now\n\u0027kolla_ansible_extra_custom_passwords\u0027 should provide only user\ncustom passwords to add or override in the passwords.yml.\n\nChange-Id: I4813a1f6ab9cb566596e806bd0ada6dff342d119\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/66eaee56625edef95c63ab560d63a7b131c6b49d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/66eaee56625edef95c63ab560d63a7b131c6b49d"}]},"branch":"refs/heads/master"},"03352360d8d324d819d8b8d666156db6613eaee7":{"kind":"REWORK","_number":6,"created":"2023-05-24 18:28:30.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/861397/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/861397/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/861397/6"}}},"commit":{"parents":[{"commit":"9b58320f6b40de3c33bee179d97dd0bf5bf61986","subject":"Show an error when *_bridge_ports is not a list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b58320f6b40de3c33bee179d97dd0bf5bf61986"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-10-14 14:56:20.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-05-24 18:28:17.000000000","tz":180},"subject":"Fix an issue when user forgot combine custom passwords","message":"Fix an issue when user forgot combine custom passwords\n\nFixes an issue when user forgot to combine\n\u0027kolla_ansible_custom_passwords\u0027,\n\u0027kolla_ansible_default_custom_passwords\u0027 and own dictionary with\ncustom passwords in configuration files. Now\n\u0027kolla_ansible_extra_custom_passwords\u0027 should provide only user\ncustom passwords to add or override in the passwords.yml.\n\nChange-Id: I4813a1f6ab9cb566596e806bd0ada6dff342d119\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/03352360d8d324d819d8b8d666156db6613eaee7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/03352360d8d324d819d8b8d666156db6613eaee7"}]},"branch":"refs/heads/master"},"6e7436891bc3e2e6de8736c6d85760707d412324":{"kind":"REWORK","_number":7,"created":"2023-05-25 11:40:08.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/861397/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/861397/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/861397/7"}}},"commit":{"parents":[{"commit":"9b58320f6b40de3c33bee179d97dd0bf5bf61986","subject":"Show an error when *_bridge_ports is not a list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b58320f6b40de3c33bee179d97dd0bf5bf61986"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-10-14 14:56:20.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-05-25 11:40:08.000000000","tz":0},"subject":"Fix an issue when user forgot combine custom passwords","message":"Fix an issue when user forgot combine custom passwords\n\nFixes an issue when user forgot to combine\n\u0027kolla_ansible_custom_passwords\u0027,\n\u0027kolla_ansible_default_custom_passwords\u0027 and own dictionary with\ncustom passwords in configuration files. Now\n\u0027kolla_ansible_extra_custom_passwords\u0027 should provide only user\ncustom passwords to add or override in the passwords.yml.\n\nChange-Id: I4813a1f6ab9cb566596e806bd0ada6dff342d119\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e7436891bc3e2e6de8736c6d85760707d412324"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e7436891bc3e2e6de8736c6d85760707d412324"}]},"branch":"refs/heads/master"},"3c6c57f8bb06dab78f750afb89b3cb0d39f7d9af":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-08-18 14:11:36.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/861397/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/861397/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/861397/8"}}},"commit":{"parents":[{"commit":"d52b764a37cded5c6523d55fbf7fa6a27a48d1bf","subject":"Merge \"Add STP option for bridge interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d52b764a37cded5c6523d55fbf7fa6a27a48d1bf"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-10-14 14:56:20.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-08-18 14:11:36.000000000","tz":0},"subject":"Fix an issue when user forgot combine custom passwords","message":"Fix an issue when user forgot combine custom passwords\n\nFixes an issue when user forgot to combine\n\u0027kolla_ansible_custom_passwords\u0027,\n\u0027kolla_ansible_default_custom_passwords\u0027 and own dictionary with\ncustom passwords in configuration files. Now\n\u0027kolla_ansible_extra_custom_passwords\u0027 should provide only user\ncustom passwords to add or override in the passwords.yml.\n\nChange-Id: I4813a1f6ab9cb566596e806bd0ada6dff342d119\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c6c57f8bb06dab78f750afb89b3cb0d39f7d9af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c6c57f8bb06dab78f750afb89b3cb0d39f7d9af"}]},"branch":"refs/heads/master","description":"Rebase"},"73abf0e57ee3aaeefa53eaf4fb7b1ed5f5c96933":{"kind":"REWORK","_number":9,"created":"2023-11-01 11:31:12.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/97/861397/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/861397/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/861397/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/861397/9"}}},"commit":{"parents":[{"commit":"d52b764a37cded5c6523d55fbf7fa6a27a48d1bf","subject":"Merge \"Add STP option for bridge interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d52b764a37cded5c6523d55fbf7fa6a27a48d1bf"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-10-14 14:56:20.000000000","tz":180},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-11-01 11:30:45.000000000","tz":0},"subject":"Fix an issue when user forgot combine custom passwords","message":"Fix an issue when user forgot combine custom passwords\n\nFixes an issue when user forgot to combine\n\u0027kolla_ansible_custom_passwords\u0027,\n\u0027kolla_ansible_default_custom_passwords\u0027 and own dictionary with\ncustom passwords in configuration files. Now\n\u0027kolla_ansible_extra_custom_passwords\u0027 should provide only user\ncustom passwords to add or override in the passwords.yml.\n\nChange-Id: I4813a1f6ab9cb566596e806bd0ada6dff342d119\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/73abf0e57ee3aaeefa53eaf4fb7b1ed5f5c96933"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/73abf0e57ee3aaeefa53eaf4fb7b1ed5f5c96933"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
