)]}'
{"id":"openstack%2Fkayobe~861644","triplet_id":"openstack%2Fkayobe~master~Ifa89e7307eb445b4f1708f0c6ac3409b3f96aafe","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2022-12-07 15:33:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifa89e7307eb445b4f1708f0c6ac3409b3f96aafe","subject":"Fixes an issue with --wipe-disks","status":"MERGED","created":"2022-10-17 15:29:01.000000000","updated":"2022-12-07 15:34:46.000000000","submitted":"2022-12-07 15:33:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"861644","meta_rev_id":"64646b22efb5a1fe0efdd0e4b1b2c66598bcac35","_number":861644,"virtual_id_number":861644,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-07 15:33:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-12-07 13:34:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-07 14:28:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-12-07 14:28:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-17 15:33:17.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-10-17 18:10:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-07 13:34:42.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-12-07 14:28:36.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"46a9cdca35ad05c49f0806fdef1ec90a7c425628","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-10-17 15:29:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03d6bc64b0b7dfe101606cf03e3b980fd31cf3ff","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-10-17 15:33:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3cf0e0dbf37dc12adc7516584ebaa0cbcf455581","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-10-17 15:37:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9a1fc3276b11e7ece5803f51a83954d181c9133b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-10-17 15:44:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9dacbf9e64642a4404dced3e29e2ca69a190eb91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-10-17 15:55:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a1f01f30821fc4fa6aa845cb04a89dfc43ec3681","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-10-17 16:31:03.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"eb74c1e8ea9caa12edd39b0c12ead76a63b177cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-10-17 16:38:16.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e0278ff5dabf4d1eabf61be1ea244114c347a566","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 18:10:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8df3cc7bc5ae4e918397f86ca9d57a91\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01179674e4be4985a079418b072e70fb : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17a01ee5e15b44a88d2e3b7f66324c9b : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc806ac4d0e840e6bbce0c4375157dfa : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c24993defce48399a866318bd878fb0 : SUCCESS in 6m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4393918b66ff4cb692882f446df4cfdb : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f20daf3b9f814ded8ff9741fc20a90e1 : SUCCESS in 9m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97f35f7c6b524189a975f1e00e651c37 : SUCCESS in 2m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5c08920f65af4929977fb9933d35cdbc : SUCCESS in 5m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/523b734c581c43fc9f6a7786f26b4cac : SUCCESS in 16m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/da5b81e5e45a4f47a763940b862a5e89 : SUCCESS in 17m 03s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21c18b06ed1a4de6b6f56a5b0549c424 : SUCCESS in 59m 04s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c55a5da7f44a4034a1793f7275ab8637 : SUCCESS in 16m 43s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf27b15480ef4d3f928048562f8331b0 : SUCCESS in 1h 25m 31s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7136ecbe2cc842b98295688c4b11ce6f : SUCCESS in 17m 36s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5dc2e32693546ce87c91d4f3c24021e : SUCCESS in 31m 04s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5675fa2dc8834d40b2ec65ee77e38cdc : SUCCESS in 12m 37s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95a67ed1f27f44179bf52b39fa64b5ff : SUCCESS in 16m 09s","accounts_in_message":[],"_revision_number":6},{"id":"af256c8f2ef7c6e34efe4c5cc5f9dcaff850eb69","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-12-07 13:34:42.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"567f9bd0ca6927f10e7b5e36a22f76d59a4a034d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-07 14:28:36.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"1ff694fe1c787b556b1661d8d7411ea0ef66b6ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 14:29:02.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"cf6f5988be017cb82aa3f7b2f071129ed4622b8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 15:33:39.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e127b206b26f41f09b39777335b4547f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8602aa01a41c472d95896918bfa5e9fa : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b5a46989ba5492bb4a4b8d337501db8 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca35782958824caeb159d3a7ba58ba3a : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d9293ef5eeb4a50a33962a4eb295b84 : SUCCESS in 9m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95051aaf00434f929a9ba68db0bcc160 : SUCCESS in 3m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9430b51e75184ba9b76955df0afcc2fb : SUCCESS in 5m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5d70e9f1107c4862a85f587789df4c1c : SUCCESS in 13m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bda315121c524135aa914987bf209ec8 : SUCCESS in 21m 54s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e115e9398fe1476aaafffdcba4bc3854 : SUCCESS in 59m 29s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a8775603f0d4ed0b6783b9a605e51de : SUCCESS in 16m 32s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a046514334a94dee944d06ce5332acb3 : SUCCESS in 58m 04s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da0c499bc14a4fd0b1589e38ae542137 : SUCCESS in 16m 57s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/920b78b856874f64a142a4561e9ba536 : SUCCESS in 58m 45s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0aa54ee3d9ef40d48922b4dbd48be20a : SUCCESS in 11m 43s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93ba24a8227f48e5b11acad19bfa2074 : SUCCESS in 13m 03s","accounts_in_message":[],"_revision_number":6},{"id":"965a07fb9fece14bdeef1824d0712890f327cc6f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 15:33:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"64646b22efb5a1fe0efdd0e4b1b2c66598bcac35","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 15:34:46.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44722bd1a51f4efabeedea5c1dab967b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04297b0c7fa34fdcb63753ace152bcdb : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5990111c0c94d84a6e7df80e71a60dd : SUCCESS in 42s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"faa5b33fef42954b552148eb425d78f19ab45f27","revisions":{"a68a6a204578e467a8d76e896d2c26db7e3f9c26":{"kind":"REWORK","_number":1,"created":"2022-10-17 15:29:01.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/44/861644/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/861644/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/861644/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/861644/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/861644/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/861644/1"}}},"commit":{"parents":[{"commit":"18ff3195fc30c5fa00e6b23071b4bafc60213cbe","subject":"Merge \"Fix bug trying to decrypt a plaintext file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18ff3195fc30c5fa00e6b23071b4bafc60213cbe"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-10-17 15:23:34.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-10-17 15:23:34.000000000","tz":60},"subject":"Fixes an issue with --wipe-disks","message":"Fixes an issue with --wipe-disks\n\nThe awk expression that was in use prior to this change was fragile\nand in some cases could incorrectly identify the unmounted block\ndevices. This change switches to parsing the json output which\nshould hopefully be more robust.\n\nChange-Id: Ifa89e7307eb445b4f1708f0c6ac3409b3f96aafe\nStory: 2010367\nTask: 46578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a68a6a204578e467a8d76e896d2c26db7e3f9c26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a68a6a204578e467a8d76e896d2c26db7e3f9c26"}]},"branch":"refs/heads/master"},"57b20f0cdca5f2b457964a013ccc6da6c52e22c0":{"kind":"REWORK","_number":2,"created":"2022-10-17 15:37:32.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/44/861644/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/861644/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/861644/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/861644/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/861644/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/861644/2"}}},"commit":{"parents":[{"commit":"18ff3195fc30c5fa00e6b23071b4bafc60213cbe","subject":"Merge \"Fix bug trying to decrypt a plaintext file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18ff3195fc30c5fa00e6b23071b4bafc60213cbe"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-10-17 15:23:34.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-10-17 15:36:29.000000000","tz":60},"subject":"Fixes an issue with --wipe-disks","message":"Fixes an issue with --wipe-disks\n\nThe awk expression that was in use prior to this change was fragile\nand in some cases could incorrectly identify the unmounted block\ndevices. This change switches to parsing the json output which\nshould hopefully be more robust.\n\nChange-Id: Ifa89e7307eb445b4f1708f0c6ac3409b3f96aafe\nStory: 2010367\nTask: 46578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/57b20f0cdca5f2b457964a013ccc6da6c52e22c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/57b20f0cdca5f2b457964a013ccc6da6c52e22c0"}]},"branch":"refs/heads/master"},"cb56776037ce4294eb3261aadf33c0440877f5df":{"kind":"REWORK","_number":3,"created":"2022-10-17 15:44:31.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/44/861644/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/861644/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/861644/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/861644/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/861644/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/861644/3"}}},"commit":{"parents":[{"commit":"18ff3195fc30c5fa00e6b23071b4bafc60213cbe","subject":"Merge \"Fix bug trying to decrypt a plaintext file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18ff3195fc30c5fa00e6b23071b4bafc60213cbe"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-10-17 15:23:34.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-10-17 15:44:16.000000000","tz":60},"subject":"Fixes an issue with --wipe-disks","message":"Fixes an issue with --wipe-disks\n\nThe awk expression that was in use prior to this change was fragile\nand in some cases could incorrectly identify the unmounted block\ndevices. This change switches to parsing the json output which\nshould hopefully be more robust.\n\nChange-Id: Ifa89e7307eb445b4f1708f0c6ac3409b3f96aafe\nStory: 2010367\nTask: 46578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cb56776037ce4294eb3261aadf33c0440877f5df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cb56776037ce4294eb3261aadf33c0440877f5df"}]},"branch":"refs/heads/master"},"ae43dccf4ae567800de307649400a7787421c594":{"kind":"REWORK","_number":4,"created":"2022-10-17 15:55:54.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/44/861644/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/861644/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/861644/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/861644/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/861644/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/861644/4"}}},"commit":{"parents":[{"commit":"18ff3195fc30c5fa00e6b23071b4bafc60213cbe","subject":"Merge \"Fix bug trying to decrypt a plaintext file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18ff3195fc30c5fa00e6b23071b4bafc60213cbe"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-10-17 15:23:34.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-10-17 15:55:00.000000000","tz":60},"subject":"Fixes an issue with --wipe-disks","message":"Fixes an issue with --wipe-disks\n\nThe awk expression that was in use prior to this change was fragile\nand in some cases could incorrectly identify the unmounted block\ndevices. This change switches to parsing the json output which\nshould hopefully be more robust.\n\nChange-Id: Ifa89e7307eb445b4f1708f0c6ac3409b3f96aafe\nStory: 2010367\nTask: 46578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ae43dccf4ae567800de307649400a7787421c594"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ae43dccf4ae567800de307649400a7787421c594"}]},"branch":"refs/heads/master"},"87d33c5b8f0a4375cbfc14791fbf8779a64095af":{"kind":"REWORK","_number":5,"created":"2022-10-17 16:31:03.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/44/861644/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/861644/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/861644/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/861644/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/861644/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/861644/5"}}},"commit":{"parents":[{"commit":"18ff3195fc30c5fa00e6b23071b4bafc60213cbe","subject":"Merge \"Fix bug trying to decrypt a plaintext file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18ff3195fc30c5fa00e6b23071b4bafc60213cbe"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-10-17 15:23:34.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-10-17 16:00:32.000000000","tz":60},"subject":"Fixes an issue with --wipe-disks","message":"Fixes an issue with --wipe-disks\n\nThe awk expression that was in use prior to this change was fragile\nand in some cases could incorrectly identify the unmounted block\ndevices. This change switches to parsing the json output which\nshould hopefully be more robust.\n\nChange-Id: Ifa89e7307eb445b4f1708f0c6ac3409b3f96aafe\nStory: 2010367\nTask: 46578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87d33c5b8f0a4375cbfc14791fbf8779a64095af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87d33c5b8f0a4375cbfc14791fbf8779a64095af"}]},"branch":"refs/heads/master"},"faa5b33fef42954b552148eb425d78f19ab45f27":{"kind":"REWORK","_number":6,"created":"2022-10-17 16:38:16.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/44/861644/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/861644/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/861644/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/861644/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/861644/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/861644/6"}}},"commit":{"parents":[{"commit":"18ff3195fc30c5fa00e6b23071b4bafc60213cbe","subject":"Merge \"Fix bug trying to decrypt a plaintext file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18ff3195fc30c5fa00e6b23071b4bafc60213cbe"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-10-17 15:23:34.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-10-17 16:37:55.000000000","tz":60},"subject":"Fixes an issue with --wipe-disks","message":"Fixes an issue with --wipe-disks\n\nThe awk expression that was in use prior to this change was fragile\nand in some cases could incorrectly identify the unmounted block\ndevices. This change switches to parsing the json output which\nshould hopefully be more robust.\n\nChange-Id: Ifa89e7307eb445b4f1708f0c6ac3409b3f96aafe\nStory: 2010367\nTask: 46578\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/faa5b33fef42954b552148eb425d78f19ab45f27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/faa5b33fef42954b552148eb425d78f19ab45f27"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
