)]}'
{"id":"openstack%2Fkayobe~863454","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I840a56339c826d05da663f8cc8eb0623ad12c468","project":"openstack/kayobe","branch":"stable/yoga","topic":"decryption_bug_fix-stable/yoga","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-11-04 13:45:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I840a56339c826d05da663f8cc8eb0623ad12c468","subject":"Fix bug trying to decrypt a plaintext file","status":"MERGED","created":"2022-11-03 16:45:24.000000000","updated":"2022-11-04 13:46:40.000000000","submitted":"2022-11-04 13:45:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863454-decryption_bug_fix-stable/yoga","cherry_pick_of_change":858925,"cherry_pick_of_patch_set":4,"meta_rev_id":"4ea9f99b15233a2ba1b86f5c7c2619520354c39f","_number":863454,"virtual_id_number":863454,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-04 13:45:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-11-03 20:05:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-04 11:44:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-04 11:44:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-03 16:45:24.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"state":"CC"},{"updated":"2022-11-03 18:50:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-04 11:44:57.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"3ca1dae1628232f20a88b25642853d88af54621c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-11-03 16:45:24.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"57ef44c8bf63e1807ab8633a84f15b2be599b79c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 18:50:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be7e151762d041acbc58afdf5ddf24a8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc81653d0263435383f5855d467d7aaf : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0b5e6e20e4a4048a3520fcb098d3b09 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ef58b6f1a404437a0db0cf346a80e89 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35f850e835024ac78df3fb85b24be9cf : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbc8bd48ea314e4498dfbcf3c272f40e : SUCCESS in 9m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bceda07909d24f33a416602d8383f4ad : SUCCESS in 3m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/00d8b4a6e042407ab9b6310cde003c9c : SUCCESS in 5m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/be2653e956e24e319455deed61e2d066 : SUCCESS in 14m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e76fea985885451b8ec7169c6ac3b4e3 : SUCCESS in 16m 22s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4478523453a44077bc2843a28dae0cd6 : SUCCESS in 1h 15m 34s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/44008c6d099541438f0a80fe33ce5eff : SUCCESS in 1h 09m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a30922325eec48e89c42b78cd93039df : SUCCESS in 52m 26s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0e5b63601b464822b087be62415ae907 : SUCCESS in 1h 08m 03s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/af8ec998ab434865ba07041cb2c8bf02 : SUCCESS in 22m 58s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/6b08063af6ed41ba8451adba28faf94c : SUCCESS in 20m 57s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a282afec6f44dee8e4618a233f3acf9 : SUCCESS in 13m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0b785395b45f4cd08526635417938cf6 : SUCCESS in 1h 59m 11s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5eca03ed57384cd6891f5cc7e6a12f56 : SUCCESS in 1h 22m 26s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/29ab015b64214679b6733e794bb06b71 : SUCCESS in 20m 26s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/bc6ad68873044dd1ae977e640cea91eb : SUCCESS in 21m 15s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/371d46c31ddd4938b275d102da887d92 : SUCCESS in 16m 44s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/3237d2df8fd8408cba898b7f2ca41578 : SUCCESS in 47m 00s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8245c4591a17465fa6272b3e109b70ce : SUCCESS in 32m 45s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bad3207f8214f5bbfa403c61f63b0ac : SUCCESS in 31m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a475e163e9354ef68db4d8de56df43c6 : SUCCESS in 16m 40s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/1e157b58fc624600bda1bc7920d3b9dd : SUCCESS in 14m 19s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc0a565c1edc435b9f5fb28ce1bb3de1 : SUCCESS in 12m 27s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/5c114ac0ca45422183ad08e843a5070a : SUCCESS in 14m 07s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e9ca0ac2bf70406789e2ced8781f6ead : SUCCESS in 17m 55s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/987be73039e941d7913952ed0ddd0d50 : SUCCESS in 10m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d7c3087aab7342ef66d35aad405ace463df5b8a4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-11-03 20:05:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2694f67093271ac8ce911a9b035aadbbbc215d8f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-11-04 11:44:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9474dcd6361ba3344711ba23e90b0a419fdaba91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 11:45:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"77e62f1f9e8200a4719dd86b6c5eb4ec49cf2c58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 13:45:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bb6ae9709b3498297ca2f703bc271de\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bdaedcba3ee44abbb7a624333674b8d : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27ff166563fe41858a6ab1b5d51e19dd : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60a53da7d50d4d0ebe6cf535a8f7adcb : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2db4d9c630a649d7a511d656a7073ac3 : SUCCESS in 8m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e9888f640aa4c6f88bd639639f4ab32 : SUCCESS in 3m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e8145438a59c4ad78075a51db5b09e24 : SUCCESS in 5m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8001a681985641aea11977990d65cf76 : SUCCESS in 14m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3880072b703945fa9711a3e3d0146dfd : SUCCESS in 16m 22s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9fae216dd4c14712901a18da705dda8c : SUCCESS in 1h 13m 01s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/a5d97c7ea3c14df0a6325931f9f5ff11 : SUCCESS in 1h 06m 38s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4c288e42d8b4565b3e4b1fefdcb1ddd : SUCCESS in 52m 46s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c062ab1ae937479593a0a0829a1ccff8 : SUCCESS in 1h 10m 07s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/08767d6f19864c6d82cd8de741109533 : SUCCESS in 23m 30s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/4f21cad552dd4becae30a2b10a3d1fd8 : SUCCESS in 22m 53s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7864002421d945fdaaeb9bd60696f3fa : SUCCESS in 13m 29s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0b98ae7e2f864eb1ad2a0985d8ea7409 : SUCCESS in 1h 54m 27s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0597e389570942248c96b3856032cd46 : SUCCESS in 1h 13m 59s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a7bbf77a8d974db8b33069a7fced37e2 : SUCCESS in 20m 17s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/68a17b4d4596410aab46b1402a47715c : SUCCESS in 20m 21s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1253750f5619426eac216d13be888218 : SUCCESS in 15m 55s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/00f54e3ec7e04a8e8d5c725d1540106c : SUCCESS in 35m 30s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f98ae006ccfd409ca7e5feab5970a3d4 : SUCCESS in 25m 13s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c6cfaa94a1344b3eb45437d3871dbfe2 : SUCCESS in 16m 43s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/5d9897e33c70499ea5da94040d85b4b8 : SUCCESS in 16m 19s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/792fcb69a3de403fb730697cfee792d4 : SUCCESS in 10m 05s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/a47aca37b22f41d39fcb07bf744da03d : SUCCESS in 17m 28s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/ab599583be2a473aaa72b91f9d89f020 : SUCCESS in 19m 12s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b1929bef3a545d79532c2e7d0b53e1d : SUCCESS in 12m 57s","accounts_in_message":[],"_revision_number":1},{"id":"94b17cc79d219d69e7e39f4aa7b0e3b9eaf60739","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 13:45:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4ea9f99b15233a2ba1b86f5c7c2619520354c39f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 13:46:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab05139f1e1641d9bd4a75398a8f6a49\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e3d5c2f7f8b4cff87e301cf24029637 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d506b71185345549fff1fbfd09512f3 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f9a0bbbab03dcd76f6fde11b59429652b60409db","revisions":{"f9a0bbbab03dcd76f6fde11b59429652b60409db":{"kind":"REWORK","_number":1,"created":"2022-11-03 16:45:24.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/54/863454/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/54/863454/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/54/863454/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/54/863454/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/54/863454/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/54/863454/1"}}},"commit":{"parents":[{"commit":"fd25eb9e04c86ce5faec70ef5cff15057bea3ca3","subject":"Update to latest CentOS Stream 8 cloud image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fd25eb9e04c86ce5faec70ef5cff15057bea3ca3"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2022-09-22 13:16:27.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-11-03 16:45:24.000000000","tz":0},"subject":"Fix bug trying to decrypt a plaintext file","message":"Fix bug trying to decrypt a plaintext file\n\nStory: 2010315\nTask: 46382\nChange-Id: I840a56339c826d05da663f8cc8eb0623ad12c468\n(cherry picked from commit c81d9e8b2c887171fbf5e62a1cd66051fc295bd9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9a0bbbab03dcd76f6fde11b59429652b60409db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9a0bbbab03dcd76f6fde11b59429652b60409db"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
