)]}'
{"id":"openstack%2Fkayobe~863455","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I840a56339c826d05da663f8cc8eb0623ad12c468","project":"openstack/kayobe","branch":"stable/xena","topic":"decryption_bug_fix-stable/xena","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-11-04 13:45:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I840a56339c826d05da663f8cc8eb0623ad12c468","subject":"Fix bug trying to decrypt a plaintext file","status":"MERGED","created":"2022-11-03 16:45:40.000000000","updated":"2022-11-04 13:46:42.000000000","submitted":"2022-11-04 13:45:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863455-decryption_bug_fix-stable/xena","cherry_pick_of_change":858925,"cherry_pick_of_patch_set":4,"meta_rev_id":"501599d9c6a435a27c4e4364b5e798d28677a7eb","_number":863455,"virtual_id_number":863455,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-04 13:45:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-11-03 20:05:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-04 11:45:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-04 11:45:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-03 16:45:40.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"state":"CC"},{"updated":"2022-11-03 19:11:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-04 11:45:26.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"a7ba00860795eef91737dc8dd80a6ea4d7461d64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-11-03 16:45:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d94b5b6a10e28a6a7eec9de5117dea7470668c2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 19:11:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0452fe1dfb5a455cbee756f25095d4c1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da75a0e57ace4fdfb0a4370a7922f374 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22d59b4d011d4704b00e0f363f230a40 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/677e46e8d2a94f03b799e831d8fb7f2f : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99dcc005c50d4fd99275cf3f0b980fb1 : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec67f2f9156d4af6af7535712c53a2a9 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e3b1497d97d468fb51e0537f06563d7 : SUCCESS in 10m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5cae71cd57541c18fbab431c0887ed9 : SUCCESS in 5m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/706a88f1660643dab2705dd7052c7727 : SUCCESS in 6m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6e3b651e78f04fffb8e20548667458cc : SUCCESS in 14m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0bb54dc90cbf4098ba3d918dc2a9442a : SUCCESS in 19m 06s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/bee61c957ce14094a231992e9d777d54 : SUCCESS in 51m 38s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/1048e27aec49425fb835e09727f7b149 : SUCCESS in 1h 11m 21s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7aedce9ab7c54aea955b7203aa72edd0 : SUCCESS in 49m 57s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/803ddb346c3a48348ab182b92656a908 : SUCCESS in 1h 04m 17s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/eb8a33c512ce4f25862eba2fb4724244 : SUCCESS in 24m 21s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/561b5b7ef5b24fc18621afd5d2eeb899 : SUCCESS in 25m 04s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fccf8eba3420426ab5ac31d3e5ca0e4f : SUCCESS in 14m 44s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3245bf97d42549ee824beeaa5c5af757 : SUCCESS in 2h 07m 08s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f72b804db60f468899187e52708bbbb2 : SUCCESS in 1h 14m 57s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/cd751cf4218d44bfa0ced7bba25e8d31 : SUCCESS in 15m 47s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/6e25174e4aa34c218ded3be89684b587 : SUCCESS in 19m 25s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb48be053cd845d3a2e901a4204380a3 : SUCCESS in 15m 50s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/f968852e543841bba4460e66095ff8a7 : SUCCESS in 39m 35s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7e38013e8cca4f34af75884245010335 : SUCCESS in 29m 07s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4719a3b5e2af484bbf8b43247b18ada9 : SUCCESS in 25m 16s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a1ce7e2a1c7248d29c6208259168b9e5 : SUCCESS in 13m 17s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/fec17ed1befc4993b6aba92765f287d8 : SUCCESS in 18m 46s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92cba06d62474f949cf63fc25587468d : SUCCESS in 11m 38s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/ae97f278603b4a2f9b392d2482059dd3 : SUCCESS in 18m 33s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/66a2e6f1f7864831bb2ba74af1784de6 : SUCCESS in 18m 04s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68eeb26a4f7643969c944f3e52114854 : SUCCESS in 14m 24s","accounts_in_message":[],"_revision_number":1},{"id":"adfbc4cd86fcbabe06d7882d62692c4dc768a365","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-11-03 20:05:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d0fc4834773080f95d3c94419f5ecd0340defda7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-11-04 11:45:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4c0c217312d2cb92d62ce8105de70291a3314a63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 11:46:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ca5e271d135a36903f4e81d9365b5b56e1e19b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 13:45:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e16a883e7aa4e8aae09b314372992d4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76fee5f5a2b543e1955af4ca21936b3c : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a064599b0b4447928fa3df6dd0f1d31a : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3eef00d380d64e65af4a99d646985e3d : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc68ac81e7c34c6a985503ad4538f9ea : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b8ba9938e214bddb6802f4b87d4f2f7 : SUCCESS in 3m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9cb061b8822d405a8b355c284dfd9687 : SUCCESS in 5m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1e05412ac1f9438c8d094ab9411eefa2 : SUCCESS in 14m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/98f2933f59e54bc9964c6f601b01cb7c : SUCCESS in 18m 12s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d90c1a6ef11e4be08c80d0b5bdc7dd35 : SUCCESS in 1h 06m 06s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/64391659a27549dc9def7a33d1c34f10 : SUCCESS in 1h 19m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a82e09e130d742e79ec4e323106afa3a : SUCCESS in 59m 04s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/58095806309a4639b878a20afba07826 : SUCCESS in 1h 21m 12s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/47586c10a5284856b5e38c9afacdb47c : SUCCESS in 23m 15s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/c71588ddbf544628a65736e62bdf948d : SUCCESS in 24m 05s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5702e6a432da4c26b825d811b4c50241 : SUCCESS in 16m 51s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ce0f6d4a85c0437f946a3e6bfcc92ed8 : SUCCESS in 1h 52m 36s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/289aec2e1fa541349c029fd37bf4e40e : SUCCESS in 1h 27m 05s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b793ac64e2044300a703124e2fafd635 : SUCCESS in 21m 50s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/316a5a68af9e4b85a313e84e57dc37c1 : SUCCESS in 20m 31s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35db963149964289a4427e3942528baf : SUCCESS in 16m 16s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/572fb1a8b68d4ac9bb5342809b6e7e22 : SUCCESS in 30m 24s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6baccf1b35854f2f9878cb798cf4ad85 : SUCCESS in 21m 44s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4fcbe308400a4155a307e3275019cb29 : SUCCESS in 15m 21s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e3aaed94984b48aea108b78385da184b : SUCCESS in 16m 24s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c79ab624cbd4f5fb8c4f0d92e5983b3 : SUCCESS in 11m 50s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/2dfc9927f5c4490c9661e5022ac34bee : SUCCESS in 17m 45s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/aa9ad1917f5542a29ae508c48738ea58 : SUCCESS in 19m 49s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db0e944ce7b8477bbd025965c7609ac4 : SUCCESS in 10m 36s","accounts_in_message":[],"_revision_number":1},{"id":"b9d92cee1aef92413ae7d9168e6334d0178ec969","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 13:45:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"501599d9c6a435a27c4e4364b5e798d28677a7eb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 13:46:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8f12e5e0e954e1bb40f4094d123510d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d9a1b685b6644878a768350158de6e0 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d4d1344eb8c4d45bdecf6b7dea08987 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2db4e4a1c479b748498fcfeef5fc4fa580c84192","revisions":{"2db4e4a1c479b748498fcfeef5fc4fa580c84192":{"kind":"REWORK","_number":1,"created":"2022-11-03 16:45:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/55/863455/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/55/863455/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/863455/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/863455/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/863455/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/55/863455/1"}}},"commit":{"parents":[{"commit":"41a65ecaffe130e3d677ce1a9d763abe2adc394b","subject":"Merge \"Update to latest CentOS Stream 8 cloud image\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41a65ecaffe130e3d677ce1a9d763abe2adc394b"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2022-09-22 13:16:27.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-11-03 16:45:40.000000000","tz":0},"subject":"Fix bug trying to decrypt a plaintext file","message":"Fix bug trying to decrypt a plaintext file\n\nStory: 2010315\nTask: 46382\nChange-Id: I840a56339c826d05da663f8cc8eb0623ad12c468\n(cherry picked from commit c81d9e8b2c887171fbf5e62a1cd66051fc295bd9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2db4e4a1c479b748498fcfeef5fc4fa580c84192"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2db4e4a1c479b748498fcfeef5fc4fa580c84192"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
