)]}'
{"id":"openstack%2Fkayobe~863456","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I840a56339c826d05da663f8cc8eb0623ad12c468","project":"openstack/kayobe","branch":"stable/wallaby","topic":"decryption_bug_fix-stable/wallaby","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-11-04 13:45:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I840a56339c826d05da663f8cc8eb0623ad12c468","subject":"Fix bug trying to decrypt a plaintext file","status":"MERGED","created":"2022-11-03 16:45:55.000000000","updated":"2022-11-04 13:46:55.000000000","submitted":"2022-11-04 13:45:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863456-decryption_bug_fix-stable/wallaby","cherry_pick_of_change":858925,"cherry_pick_of_patch_set":4,"meta_rev_id":"02cccebec52df54ea39df7e7b5117e946cc3de63","_number":863456,"virtual_id_number":863456,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-04 13:45:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-11-03 20:05:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-04 11:45:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-04 11:45:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-03 16:45:55.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"state":"CC"},{"updated":"2022-11-03 18:39:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-04 11:45:34.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"01e51eb6195bdd87276307a7f9f8de6e8fc31bf4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-11-03 16:45:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"88bf3eead28ec70c5a5cf56464782c51cc6ba482","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 18:39:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf846586aff645b5b1cdb1812af1fce6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/455d852a6e5a490aa0d9208c84208d41 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d47ff90dc98248338903360d1d61fcc5 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ed6b95555724be0bf17410d783cf82f : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dad2891b0a444c35af00570024eb2ed0 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f79c90fac2134cf5b4783edd90d12cc3 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edc4d128ca6948caa6b6f7e6afd5c3b8 : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afa7a62730924adfae4751ba3637c5d5 : SUCCESS in 3m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2b37e501f43c4f92a076104384c59d6f : SUCCESS in 5m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0e6bc39484264799ba46b9a793bbda22 : SUCCESS in 10m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/111adae0620147109c79d810d1781a61 : SUCCESS in 18m 44s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/712475b58f4d47ca9a2479447c018866 : SUCCESS in 1h 11m 26s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e1d1209d74242108f767c9e1eb7281c : SUCCESS in 58m 45s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7e6a3238da8a43fb8ac320b35332789a : SUCCESS in 1h 14m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/393982af41d04171b6ecbeea2ee8e4a9 : SUCCESS in 21m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2186a27cdc23425f8377c1ac13be9ce1 : SUCCESS in 14m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c5cb3a3841db4dd18194aa6543b65771 : SUCCESS in 1h 41m 52s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/066cd7f899664e7ea0a676388554dc8e : SUCCESS in 18m 44s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/497b413c13dc46269a32a9b5c4c7ff14 : SUCCESS in 15m 55s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/826002813d3643eeba5820fe8bd7bd9b : SUCCESS in 48m 37s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8ef53946b9ca4ef49d36475d2fe18f25 : SUCCESS in 26m 12s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f1fed8405e7e45c4beec9223fe070b67 : SUCCESS in 16m 50s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/483ab55b738c4d00807550013c571d9c : SUCCESS in 11m 09s","accounts_in_message":[],"_revision_number":1},{"id":"4cb4dd1c1ba8b6ee25fa0966952306e3e0803446","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-11-03 20:05:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"28fd688452503ced644a5cb4a85b5d2da2950c3d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-11-04 11:45:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"183d09a6ce997a74de33bd2e236df7e33e97fa01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 11:46:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bde0600bc4998a1a625e20204a9a37bb19840751","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 13:45:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a6311048094468faffbb204b600ad02\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c4c0c410bfb46a0bd534972d7b41190 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91fb1309b7dc41798e835b8e05a5b65c : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89e612a8ed224726940cc0f3c168f640 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca50f0f9ed974e5baed49f1a356630b9 : SUCCESS in 8m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aacc657a07ab40e9af00e049d73d338e : SUCCESS in 3m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1480bd4eb151467993f43cabe27762c4 : SUCCESS in 5m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/866dc064fa504f6ab665bc8c529e5682 : SUCCESS in 9m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/350eee06a6504755b35e0025325beb7f : SUCCESS in 17m 35s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/722e8e1b06304d2f81d1224f6b9373f2 : SUCCESS in 1h 16m 54s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cca295c469641b292a940f34750a622 : SUCCESS in 54m 56s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c41a33aa5a4c40d78032a00535ddb06f : SUCCESS in 55m 29s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/202edf61ada04f7db623aa3a00facab0 : SUCCESS in 20m 32s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2cf8f6d2b7d441083763100023a3b2a : SUCCESS in 13m 52s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c94cd7e510984bdfa3eb7fbc4ca32a27 : SUCCESS in 1h 45m 08s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0ffe963c5b774eaeb3a3beeb71bbc2b8 : SUCCESS in 18m 33s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06b45f733ce648e9b28fe7b2663460e5 : SUCCESS in 14m 45s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/056446fb57c74af3a663cc22b28b99c8 : SUCCESS in 25m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3fb1205d5ece46c492198c2707a30d4f : SUCCESS in 14m 14s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/702a7ac67390474298ffa94009149647 : SUCCESS in 11m 55s","accounts_in_message":[],"_revision_number":1},{"id":"7b1445ced6f1225ee4016bf2ca2dc106b3e9aa68","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 13:45:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"02cccebec52df54ea39df7e7b5117e946cc3de63","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 13:46:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d6e470319d44230a4a783c1286d8587\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a7735f4400740d5861a20b3b0c19f43 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7068519b6fa04a49905d105a27a4d2b2 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7ab1de390e4c86be5221929a1342294e09462f07","revisions":{"7ab1de390e4c86be5221929a1342294e09462f07":{"kind":"REWORK","_number":1,"created":"2022-11-03 16:45:55.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/56/863456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/56/863456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/56/863456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/56/863456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/56/863456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/56/863456/1"}}},"commit":{"parents":[{"commit":"e2aadb31c150795a6ed46f8545fa8fef311397d1","subject":"Skip IP address allocation and configuration if needed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2aadb31c150795a6ed46f8545fa8fef311397d1"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2022-09-22 13:16:27.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-11-03 16:45:55.000000000","tz":0},"subject":"Fix bug trying to decrypt a plaintext file","message":"Fix bug trying to decrypt a plaintext file\n\nStory: 2010315\nTask: 46382\nChange-Id: I840a56339c826d05da663f8cc8eb0623ad12c468\n(cherry picked from commit c81d9e8b2c887171fbf5e62a1cd66051fc295bd9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ab1de390e4c86be5221929a1342294e09462f07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ab1de390e4c86be5221929a1342294e09462f07"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
