)]}'
{"id":"openstack%2Fkayobe~864518","triplet_id":"openstack%2Fkayobe~master~I2f4f02dbd6834ffd0fa5189b818bdc91a5c168a2","project":"openstack/kayobe","branch":"master","topic":"nova-compute-release-file","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-12-08 16:25:00.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I2f4f02dbd6834ffd0fa5189b818bdc91a5c168a2","subject":"Support copying nova_compute Kolla configuration","status":"MERGED","created":"2022-11-15 12:22:30.000000000","updated":"2022-12-08 18:51:47.000000000","submitted":"2022-12-08 18:50:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864518-nova-compute-release-file","meta_rev_id":"09e375f1f5c304016a0016238c2508c1349f45aa","_number":864518,"virtual_id_number":864518,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-08 18:50:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-12-08 14:43:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-12-07 13:30:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-12-08 14:43:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-15 13:20:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-07 13:30:49.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-12-08 14:43:54.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"5c345d9f52458326fb3675d7be4cd7942c94ca6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-11-15 12:22:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1dc8f4e2e64c541bd6120bb6234e04b7b7b1def","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 13:20:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1e79a6e85c14bbea2c3aeb5804d951e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c35a071f8ca431999aac01975ce260f : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf2fb967a4c54b2093bfc5a425477225 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34d4b9fcb191475eb204f40baed1cdb3 : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a97399cb0c154cf8811c4ebb2f10e236 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8fc9712bedf2469289d8911f9d8c480a : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80aa4ba6ae9241e598baafb1284e63b3 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9dc41301cbf44c14a96baef697a1fcdd : SUCCESS in 3m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d189e6d542b94ae2b88b7f1730de45ad : SUCCESS in 5m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2401a3e759ed4f1e8419cd524b002d90 : SUCCESS in 9m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/42e611f4811549c3a199de1165b91008 : SUCCESS in 15m 49s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e06c1a5b7bb4a0980d13916616ee691 : SUCCESS in 56m 21s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c893891b498c4c6b9b65f408090a3a54 : SUCCESS in 12m 58s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c0645742cd34356949345ed6e024d2b : SUCCESS in 53m 52s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cff05c167fe94c689b56c038f4191ace : SUCCESS in 18m 19s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af3480fa44ac4c8c89c4416430531b18 : SUCCESS in 21m 31s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8416f29df4d442c8273cd55750f7077 : SUCCESS in 13m 18s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19ec126610bd4557b636becedbd5c29c : SUCCESS in 10m 56s","accounts_in_message":[],"_revision_number":1},{"id":"3d78fae54adea14a6be48d0101bf7229e9e29645","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-12-07 13:30:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"414b1a331f4aad352f849c8684ed7f5cf69f2773","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-08 14:43:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a20c38845d87afd9fd8517ad10fc92cd89d88b7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 14:44:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ee37f22a6dcb3feaa151a28e903665144b9a979","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 16:05:01.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1537184df7d4771a6aa6c7a5b7c524b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f972dbd6d0b745139d2d9fbad8a6fb66 : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f51b4497cbde4e899a2fb0941678fe48 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22daca8dac864ce79cc7740fc670f598 : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a96698b5b8784dacbc82fde36996d574 : SUCCESS in 9m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c415cdcc616c4962a89bb8b1326a77ae : SUCCESS in 3m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0093ace8f844453d9c2dd084f725f82c : SUCCESS in 7m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/21a96375b66e4182ab96f8c25bcca28b : SUCCESS in 16m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1bfb5d0d9a594181b5f2ad2fa84c59a7 : SUCCESS in 17m 07s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39d1209358bc4549ad8c0a8da26fd01f : SUCCESS in 54m 02s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0664b5a2efb743ce814b7509c43b370b : SUCCESS in 15m 49s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edac0e5ada774a5f8f3ede701962ca4c : SUCCESS in 1h 15m 55s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd71ba43dd3a443cb222e5c5b4ea26a4 : FAILURE in 19m 36s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3dbf90050e254777afdc16d9b66c77f6 : SUCCESS in 26m 31s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/054e88aac20344129f1a77f15ca9449f : SUCCESS in 11m 45s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/107064a90b8946638a58047613f57761 : SUCCESS in 14m 51s","accounts_in_message":[],"_revision_number":1},{"id":"c843c95f35706a455b922aa1c60436c7ebb12a00","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-08 16:25:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2d9675f5c520d5dff664cee43ac6f122f8bbd8a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 17:36:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb94563851fa488681dfa467a1a9f996\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82feb03ffe2c48e49b149d81bc596769 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e34ccc62007142219ffb87bd94c139b2 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/810f49e4465e497a829ae825313b44b2 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83585a8b279f46238829a067324a870e : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a61835c179b488f9e47a83c16675d05 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0bcd8b4c22e475ead697dba453c97de : SUCCESS in 9m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac624a5a17f4462894a703875b78d167 : SUCCESS in 2m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/180f411ab990466e88629d656057656e : SUCCESS in 5m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/31c9380cccc24b0cafadc9371c6cf3d8 : SUCCESS in 14m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b7ccf09548bb471d80d42f53f6ea3498 : SUCCESS in 15m 31s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36c30b9af836411187df642d5699243a : SUCCESS in 45m 34s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/251e91a45b4a46e9a46c42b70f4f2585 : SUCCESS in 14m 41s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40636222088d403193e5a4c8af840254 : SUCCESS in 1h 07m 17s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/480f013e012e4b579e3f6ecfdd109dce : SUCCESS in 16m 52s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/782dab2b52b14d249a8ac9c5a09ca494 : SUCCESS in 25m 15s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c1f72037b6e44f6894357a6b384bb68 : SUCCESS in 11m 26s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47e4fb7e55af43e89f6226c5ac0389ce : SUCCESS in 10m 53s","accounts_in_message":[],"_revision_number":1},{"id":"fd55b5503140deee9ffacaa1f863f09864e90f41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 17:37:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bda88988363576d09d82105dc20fb9e2a271da0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 18:50:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a1a667ad26c421890da802b07bed98e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87e8d25230ea4200b178de39105bd6d9 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d787660699d4d759be37916a04ba5a0 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c51a263dffb4c8d8b418b3b240f90e5 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af2cf0305e1b493797f1c51af72a70df : SUCCESS in 8m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4f6893485e144a787d5b3ff56157c4a : SUCCESS in 3m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/37a810682c2b4439bfa4cfce05908797 : SUCCESS in 5m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2127249ac9374e308dbd999e8724e052 : SUCCESS in 15m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/affba83c97214a249ebd9172ef366d09 : SUCCESS in 16m 31s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34bae420b99d4dea88768aafa3710726 : SUCCESS in 55m 53s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cec879b2b6dd481ea629b392ca566fb0 : SUCCESS in 14m 55s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e275caa9c7d04e6d85ee46d820133e97 : SUCCESS in 1h 12m 52s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e51786bbf7dc405083ccdb41bc9a1f3f : SUCCESS in 17m 19s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/562eb846998144ffab431d1036e77ec6 : SUCCESS in 29m 57s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a85206fecc4047808fedd9fb8461ca62 : SUCCESS in 12m 55s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e24dd18cf3044f1858195d45cbafca3 : SUCCESS in 13m 04s","accounts_in_message":[],"_revision_number":1},{"id":"c423879fb05aa06741b66ac34fee4d885567e8d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 18:50:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"09e375f1f5c304016a0016238c2508c1349f45aa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 18:51:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5641ecfb7eb34097b8408dce625568a4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e36a6f420c184056894ea16ab0151cd3 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c54784e25e6496891c9a911313198dd : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"59aaec13f16163b438963ff72f2f0edc1ac76c74","revisions":{"59aaec13f16163b438963ff72f2f0edc1ac76c74":{"kind":"REWORK","_number":1,"created":"2022-11-15 12:22:30.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/18/864518/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/864518/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/864518/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/864518/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/864518/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/864518/1"}}},"commit":{"parents":[{"commit":"d4ce41baa22abf8ef9224acb0432a38fb0b286ea","subject":"Merge \"Adds missing variables to ssh.yml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d4ce41baa22abf8ef9224acb0432a38fb0b286ea"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-11-15 12:21:36.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-11-15 12:21:36.000000000","tz":60},"subject":"Support copying nova_compute Kolla configuration","message":"Support copying nova_compute Kolla configuration\n\nThis folder can be used to store a Nova release file [1] specific to\nnova_compute containers.\n\n[1] https://opendev.org/openstack/nova/src/branch/master/etc/nova/release.sample\n\nChange-Id: I2f4f02dbd6834ffd0fa5189b818bdc91a5c168a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/59aaec13f16163b438963ff72f2f0edc1ac76c74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/59aaec13f16163b438963ff72f2f0edc1ac76c74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
