)]}'
{"id":"openstack%2Fkayobe~866474","triplet_id":"openstack%2Fkayobe~master~I4034376cfe78f6a520e035be8df028ff5a747b43","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2022-12-07 09:57:10.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-12-07 09:46:47.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2022-12-07 09:57:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4034376cfe78f6a520e035be8df028ff5a747b43","subject":"[release] Tidy up release notes for Zed","status":"MERGED","created":"2022-12-02 13:59:31.000000000","updated":"2022-12-07 09:58:21.000000000","submitted":"2022-12-07 09:57:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"866474","meta_rev_id":"f9cf1e778f65c6e0d8247bd447ed3f61536d1dac","_number":866474,"virtual_id_number":866474,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-07 09:57:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2022-12-07 09:37:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2022-12-07 09:46:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2022-12-07 09:46:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-02 14:10:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-05 09:21:12.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-12-07 08:36:58.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-12-07 08:36:58.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"d34b77c22c0cbbff382df33f85a7886860fd5bb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2022-12-02 13:59:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1908c38ce692a79b5dd409e743f7e05013b0f0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 14:10:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da25b4e2c1cd4964881ab91de1c58c72\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3769ead98fd4677a4f1638a325dfd29 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7fb6522fe344541bf2d40ee2e388ba7 : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c4161512d2e4af2bbb1409784f1176e : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":1},{"id":"90ccae7fd5c5728d00a86e0d4b86a7f015bb469f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2022-12-02 14:14:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c321e94b8e06dd012d30d2a02d646e97c6333055","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 14:24:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8025b5735bcd49a982a6e143d6a4568a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69598831d6ca40368904acf59b413f60 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/885480bfb7c6437abd92da32fa6935b5 : SUCCESS in 9m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f807019caf6d4f3783c831dfeb5bcd7f : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":2},{"id":"20a7d8d20a1a5e3411d159c840ec3753e9bd444a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-05 09:21:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"160e62c106af97ecb1ab28acf89ef2497c270935","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2022-12-07 09:37:05.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"62d29d720ff90728a0c9fbf4e70267a6c0014610","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2022-12-07 09:37:21.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8dd618c2b548cbba09e15681e55283abd041b76b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-12-07 09:46:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b04d74b6f9a17888905cb0904cb9947145e95b7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 09:46:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/beb36f622b6847ebbcb55d30db1884d2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d019f455b8674fd4a89100c949f1d7c1 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0a43fc741664190a6a872e3f9cb3af7 : SUCCESS in 8m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c867bf0c9339416d98d8a2df06a4e004 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":3},{"id":"05662467596fcdaf23af3eaf7d830f5c87031f0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 09:47:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3290254d2522ae927ad0c1fbd274cafc3187bbca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 09:57:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8422f58df7aa41c19582560f5de7b5d4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85feb3e95d1944f797ca4dd7a8261d85 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/204c6d85b0804582922254f5836f572c : SUCCESS in 9m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c39b03929bde4e448046f9971213300b : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":3},{"id":"ea19b96609e56480fb54cb527823a8c3e7bd8cb2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 09:57:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f9cf1e778f65c6e0d8247bd447ed3f61536d1dac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 09:58:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a75776b13c841d2ab1900113fe8fe91\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cba3f386459f4a939ab303f79dcb26e8 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56736995fc15444abd228d59487441ae : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"31085438b1f1913b49d42b37e3c315867e68c0dc","revisions":{"68f70d151323358315331d0296bc3182608fcea7":{"kind":"REWORK","_number":1,"created":"2022-12-02 13:59:31.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/74/866474/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/866474/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/866474/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/866474/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/866474/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/866474/1"}}},"commit":{"parents":[{"commit":"744ca0918f21f1df684d0e99d17971db22f33813","subject":"Merge \"CI: Fix molecule job on Ubuntu Jammy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/744ca0918f21f1df684d0e99d17971db22f33813"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-12-02 13:59:16.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-12-02 13:59:16.000000000","tz":60},"subject":"[release] Tidy up release notes for Zed","message":"[release] Tidy up release notes for Zed\n\nChange-Id: I4034376cfe78f6a520e035be8df028ff5a747b43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/68f70d151323358315331d0296bc3182608fcea7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/68f70d151323358315331d0296bc3182608fcea7"}]},"branch":"refs/heads/master"},"7dfec04b01c5af0b59b3af98ac329e6f92e9c44e":{"kind":"REWORK","_number":2,"created":"2022-12-02 14:14:04.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/74/866474/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/866474/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/866474/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/866474/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/866474/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/866474/2"}}},"commit":{"parents":[{"commit":"744ca0918f21f1df684d0e99d17971db22f33813","subject":"Merge \"CI: Fix molecule job on Ubuntu Jammy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/744ca0918f21f1df684d0e99d17971db22f33813"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-12-02 13:59:16.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-12-02 14:13:59.000000000","tz":60},"subject":"[release] Tidy up release notes for Zed","message":"[release] Tidy up release notes for Zed\n\nChange-Id: I4034376cfe78f6a520e035be8df028ff5a747b43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7dfec04b01c5af0b59b3af98ac329e6f92e9c44e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7dfec04b01c5af0b59b3af98ac329e6f92e9c44e"}]},"branch":"refs/heads/master"},"31085438b1f1913b49d42b37e3c315867e68c0dc":{"kind":"REWORK","_number":3,"created":"2022-12-07 09:37:05.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/74/866474/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/866474/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/866474/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/866474/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/866474/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/866474/3"}}},"commit":{"parents":[{"commit":"744ca0918f21f1df684d0e99d17971db22f33813","subject":"Merge \"CI: Fix molecule job on Ubuntu Jammy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/744ca0918f21f1df684d0e99d17971db22f33813"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-12-02 13:59:16.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2022-12-07 09:37:05.000000000","tz":0},"subject":"[release] Tidy up release notes for Zed","message":"[release] Tidy up release notes for Zed\n\nChange-Id: I4034376cfe78f6a520e035be8df028ff5a747b43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/31085438b1f1913b49d42b37e3c315867e68c0dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/31085438b1f1913b49d42b37e3c315867e68c0dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
