)]}'
{"id":"openstack%2Fkayobe~866844","triplet_id":"openstack%2Fkayobe~master~Ie3ea200b2bbe07271942a06fa1887561a9c04fda","project":"openstack/kayobe","branch":"master","topic":"zed-release-sync","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-12-13 10:05:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie3ea200b2bbe07271942a06fa1887561a9c04fda","subject":"[release] Sync Kolla Ansible feature flags for Zed","status":"MERGED","created":"2022-12-07 12:03:19.000000000","updated":"2022-12-13 10:06:13.000000000","submitted":"2022-12-13 10:05:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866844-zed-release-sync","meta_rev_id":"620de651156459fe536f86f1ff753e2a4d3b0272","_number":866844,"virtual_id_number":866844,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-13 10:05:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-12-13 08:45:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-12-13 09:08:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-12-08 14:49:29.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-12-13 08:45:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-12-08 14:42:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-07 13:07:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-07 13:28:44.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-12-08 14:42:39.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-12-08 14:49:29.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"d7f54f8acc1abe8f24fd6137f9fc9a85c4bb2811","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-07 12:03:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88a2fa9a41fa33bd608594cd1ad45e7aaa26a1b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 13:07:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba25c55515dc46d4ac9ff9ab5b695cf9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5c9d49279fb4d89a7cda325356bae4d : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d6c0f63c4414e8fa2f8201da298b318 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa0b796c2aed47ae8f786d7ef2d39367 : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e89aedaffb32472ebc6b0f281c78135d : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0667de08f3cc4511a41317e80438a656 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/698eecdec6134956acea9de869b71b63 : SUCCESS in 11m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/862e2bc804344ef4b700e112c8697a80 : SUCCESS in 6m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/05538ae6c1d54ebdba98658da03cb91e : SUCCESS in 13m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/94f187e809634f82a584f628a5233b16 : SUCCESS in 19m 11s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/627ce29b5c8a441fa7a0ba7c9b2b3bd4 : SUCCESS in 55m 29s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8196c66f8f24e3fa1436544c99f6cf0 : SUCCESS in 16m 55s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65c07abc5b83458c946e15fc8323d554 : SUCCESS in 1h 02m 33s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad0c2bad0db4410c8f0d0b9cd2476a63 : SUCCESS in 16m 47s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/200b9b11108e4cdc949d591d2f9b2092 : SUCCESS in 31m 19s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be53709a0e1f4c488ba6bcb1f2ad6a18 : SUCCESS in 11m 40s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d5948fe424440c697f3db8f2fc89ad2 : SUCCESS in 14m 08s","accounts_in_message":[],"_revision_number":1},{"id":"7b4fd4eae9df94c1d9727026df0b85cb98bd0823","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-12-07 13:28:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4225646d7713a0bb812afa72dfd4eb9b65a4b5eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-07 15:15:54.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"7066e3608f8ff015d341add4a3c725e2c7206b1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 16:27:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d39b4a9aa324922a926b334d9547ef6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b278b0153c044958f382bdb6557f022 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b64b4982e604d7db56838730dbb13ff : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8118e11c3ca54580afe74f1c1e3fcf2c : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/667d274e3cde4ec18371da6d87896512 : SUCCESS in 5m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/825ebca4900244edb9592650ea26c0eb : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f48ddd9050114fbb9e82c88bf7bad19f : SUCCESS in 11m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/61493bf11eec47a0bcd676e27d9f9c5d : SUCCESS in 6m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2d7199c1e2c7458cb4c8bde1f46e95c0 : SUCCESS in 17m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f4165f5f91ff42c59be01336cccc0c1b : SUCCESS in 18m 10s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9807bb9814042f99f55ff40b685891f : SUCCESS in 50m 47s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de9c49ec515040a59c2726c6031e61a3 : SUCCESS in 19m 46s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4dff468604f240a4a67597b1242e1148 : SUCCESS in 1h 06m 49s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/616ccdfacc294b75b31b90d00aff5139 : SUCCESS in 16m 05s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4aff683fe07e49a59e196ca1d9a88611 : SUCCESS in 27m 40s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc97ae01fcd74283bd148a2a21887a4c : SUCCESS in 10m 39s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2fc1ce63ce048f09520a79ca8b53bc4 : SUCCESS in 15m 53s","accounts_in_message":[],"_revision_number":2},{"id":"10d4f75c523e7c96c3a8a8e613d574656334ad50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-07 21:37:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b15eb4e68c1fac10181c3cdef657101fbb354f18","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-07 21:38:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"71a57ac5776f38a401851897432339499043f354","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-07 21:38:46.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"f3e0c5bc5de655c45da01e0f521ffd64e4da0e28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 22:50:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a063e568c40475fae8d990ff8eea41d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/118ca51c6f51494fabc6730414ac4712 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0873071d32946739897dc12a4a8247b : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2fe84e4121b74d9999f661fb72ed4385 : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f04a8eca49db4a9a8906bf6812acc62a : SUCCESS in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13c27ce77bc64f07a2cd341c3d0d317d : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d03d0f70adfc4f0c908addf3bb178e1e : SUCCESS in 10m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/431ce2dd387642168d53dc0b4234baf5 : SUCCESS in 6m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c1e42ca3e9854342b0822773f0cad882 : SUCCESS in 16m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6de7dda3647e47169520f7e152b41944 : SUCCESS in 18m 16s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cef1d66ae4d6494f87031c199c7cad46 : SUCCESS in 57m 13s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f81f9e70fb4e42cf881232796a4723ce : SUCCESS in 14m 51s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c287792fd21443dad8a77f0dce2d110 : SUCCESS in 1h 04m 34s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba4aef79a30b4e599737ee0ca7457b83 : SUCCESS in 14m 40s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e61bf1eb8b9947a5b67f92e7e2fbe6a1 : SUCCESS in 28m 49s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45dd58645dc94fc483c52b6911bd4c58 : SUCCESS in 13m 09s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af793725c8a64d78a2b7d5a46ba96353 : SUCCESS in 12m 39s","accounts_in_message":[],"_revision_number":4},{"id":"02982abba7f78875354bc456bab112caba45b3db","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-08 14:42:39.000000000","message":"Patch Set 4: Review-Priority+2","accounts_in_message":[],"_revision_number":4},{"id":"40cf71381f5529fe97757f08e2898e732540c469","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2022-12-08 14:49:29.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"3fc97657a9faf5e0df9b0b3fce2fda87cacae6c6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-13 08:45:41.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"68dbc80e8d7fadc133ec703f8ebdc11f55a9b28c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 08:46:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"fb2349499b9bf20c6b3cf305cdb463f2283b7ed5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-12-13 09:08:43.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"6609601f7d43b2239b75770fb5df653fdb6e79f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 10:05:16.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47c0838d547e48668003581b16796249\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/218f8e8f42d84c1981e27ca1fd7829d3 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ae5e1e041164ec2b2493a65a2726602 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2aa17568d05e41c3b17a3f685281dd26 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ba438bad0be4a20af4d794a1a2ab607 : SUCCESS in 8m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3f5da6f5cebb4c56ad6e3d5cda693e28 : SUCCESS in 5m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b218e0a01de44f1cb781dc4eef496754 : SUCCESS in 14m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9abaed1892664688abd55ff694e93046 : SUCCESS in 15m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/37e0c547a34642fead6104800bd83332 : SUCCESS in 1h 08m 15s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d49334428dc3485081ab16a81a09b2c1 : SUCCESS in 53m 26s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fcf3c1542c104919a5d593dde9d362b5 : SUCCESS in 1h 00m 39s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3bf52baa33ee48d9935851060c696891 : SUCCESS in 16m 16s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/830042918c464d388a7a73969f5b9203 : SUCCESS in 20m 12s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1791e5cbcd2a46d6aed7402d15cae97f : SUCCESS in 1h 14m 53s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/fa02adda8f3d4522b419be36012fbdf7 : SUCCESS in 19m 01s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2b6d632ce704c3ab3776707885a05f7 : SUCCESS in 18m 09s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6490105fa2644bf9cfe28a9239ee67e : SUCCESS in 29m 46s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8dfc47bed1bb451aaa6da8d7b938b327 : SUCCESS in 18m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dcd7a1e8c2e74868a6fb4778ff1fa60b : SUCCESS in 14m 43s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/86d32254cb014433b045796394b52cda : SUCCESS in 18m 47s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39c15abdb8634726a5634da7812bce6c : SUCCESS in 13m 45s","accounts_in_message":[],"_revision_number":4},{"id":"79b9dfbebe4ddbbdcf81a5ec144eb13071902297","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 10:05:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"620de651156459fe536f86f1ff753e2a4d3b0272","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 10:06:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f674c635fe744a089a01b50ababf2c2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13543c0a8cae4f5e9cae7fff307c97d2 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3d0e1a3818016bf74ef6a12ba31f16bf4c30cd93","revisions":{"494e1a39221d121e25d95a2fb08fd5a65e5e1b26":{"kind":"REWORK","_number":1,"created":"2022-12-07 12:03:19.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/44/866844/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/866844/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/866844/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/866844/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/866844/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/866844/1"}}},"commit":{"parents":[{"commit":"3a4b3bea5197e688a59c60b89119037102d1e0e6","subject":"Merge \"Remove support for Ironic boot option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a4b3bea5197e688a59c60b89119037102d1e0e6"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 12:02:08.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 12:02:08.000000000","tz":60},"subject":"[release] Sync Kolla Ansible feature flags for Zed","message":"[release] Sync Kolla Ansible feature flags for Zed\n\nChange-Id: Ie3ea200b2bbe07271942a06fa1887561a9c04fda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/494e1a39221d121e25d95a2fb08fd5a65e5e1b26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/494e1a39221d121e25d95a2fb08fd5a65e5e1b26"}]},"branch":"refs/heads/master"},"dbcced8be0c5fe267c295572904e1c6070b71e24":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-12-07 15:15:54.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/44/866844/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/866844/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/866844/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/866844/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/866844/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/866844/2"}}},"commit":{"parents":[{"commit":"a50fdb6b7f859aab20bcdb668b0de77bed3d0010","subject":"Merge \"Remove kolla_install_type variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a50fdb6b7f859aab20bcdb668b0de77bed3d0010"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 12:02:08.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 15:15:54.000000000","tz":0},"subject":"[release] Sync Kolla Ansible feature flags for Zed","message":"[release] Sync Kolla Ansible feature flags for Zed\n\nChange-Id: Ie3ea200b2bbe07271942a06fa1887561a9c04fda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dbcced8be0c5fe267c295572904e1c6070b71e24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dbcced8be0c5fe267c295572904e1c6070b71e24"}]},"branch":"refs/heads/master","description":"Rebase"},"ea3022b5d645d3cf3a67a8dd7887afc0fcf34dd9":{"kind":"REWORK","_number":3,"created":"2022-12-07 21:37:22.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/44/866844/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/866844/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/866844/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/866844/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/866844/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/866844/3"}}},"commit":{"parents":[{"commit":"a50fdb6b7f859aab20bcdb668b0de77bed3d0010","subject":"Merge \"Remove kolla_install_type variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a50fdb6b7f859aab20bcdb668b0de77bed3d0010"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 21:36:29.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 21:36:29.000000000","tz":60},"subject":"[release] Sync Kolla Ansible feature flags for Zed","message":"[release] Sync Kolla Ansible feature flags for Zed\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/866884\nChange-Id: Ie3ea200b2bbe07271942a06fa1887561a9c04fda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea3022b5d645d3cf3a67a8dd7887afc0fcf34dd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea3022b5d645d3cf3a67a8dd7887afc0fcf34dd9"}]},"branch":"refs/heads/master"},"3d0e1a3818016bf74ef6a12ba31f16bf4c30cd93":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-12-07 21:38:46.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/44/866844/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/866844/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/866844/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/866844/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/866844/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/866844/4"}}},"commit":{"parents":[{"commit":"ba38d7d6cecc9251828b5ab0b145a3badfc858ea","subject":"Merge \"Fixes an issue with --wipe-disks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ba38d7d6cecc9251828b5ab0b145a3badfc858ea"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 21:38:41.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 21:38:41.000000000","tz":60},"subject":"[release] Sync Kolla Ansible feature flags for Zed","message":"[release] Sync Kolla Ansible feature flags for Zed\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/866884\nChange-Id: Ie3ea200b2bbe07271942a06fa1887561a9c04fda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d0e1a3818016bf74ef6a12ba31f16bf4c30cd93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d0e1a3818016bf74ef6a12ba31f16bf4c30cd93"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[]}
