)]}'
{"id":"openstack%2Fkayobe~866845","triplet_id":"openstack%2Fkayobe~master~If7171c925be052e59e64423d62c9416485ca9113","project":"openstack/kayobe","branch":"master","topic":"elasticsearch-opensearch","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-12-13 10:05:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If7171c925be052e59e64423d62c9416485ca9113","subject":"[release] Sync Kolla Ansible inventory for Zed","status":"MERGED","created":"2022-12-07 12:08:52.000000000","updated":"2022-12-13 10:07:00.000000000","submitted":"2022-12-13 10:05:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866845-elasticsearch-opensearch","meta_rev_id":"14bd9052ea57990ee9a8789936e4ebf6647cc513","_number":866845,"virtual_id_number":866845,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-13 10:05:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-12-13 08:45:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-12-07 22:08:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-12-08 14:49:47.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-12-13 08:45:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-12-08 14:42:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-07 13:29:22.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-12-07 13:32:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-08 14:42:30.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-12-08 14:49:47.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"6549c0e4b59dcbaf0576593a1c3cff90634ff024","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-07 12:08:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90d10a48209c78e40bc96aef59a11dd1787bd1b9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-12-07 13:29:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"272fd36dc21e51abd47c20e1fcda6e2e40003473","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 13:32:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f7e06fc95834372b8f05cea9dd58633\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ce717ba07a34af891a5f0e38b24e1dd : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c556cc765a644e489f6bdde7507d604b : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3322bf8d2154b019d99690233cfb8bb : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28138aaed1354fb79d13f9bc6f26ba34 : SUCCESS in 6m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b35010db764d4d9991dca3d94c0117e1 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/463eb70126804091874394741cdadd17 : SUCCESS in 11m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f24858e71b2548db86ffda4998e69036 : SUCCESS in 7m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/79872bde8c234161bbf1bbc16b680a39 : SUCCESS in 15m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aa5491c597e645998873c0fbc71a00ea : SUCCESS in 20m 03s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73307044917b456f9d40347da0877c6a : SUCCESS in 58m 18s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6296b702016422495134964940b4c4b : SUCCESS in 18m 23s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df5b3cd479dd46189d95d200b39aff23 : SUCCESS in 1h 15m 38s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/110ff8e8af0a43229037919f16f3b0b0 : SUCCESS in 17m 02s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa7c281f719646d88c9cce2ac7b90b5e : SUCCESS in 29m 09s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d830938ec03d4e2bac9a6cb345f99f83 : SUCCESS in 14m 33s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16983c626ee442a082d76d8a359e9f26 : SUCCESS in 16m 04s","accounts_in_message":[],"_revision_number":1},{"id":"3923a4683db04f28fa3f656886cc909259985caf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-07 15:16:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"49b0bc66081a438d35dc4ca627ddf68e139bb827","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 16:30:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de2c29611f1e4e709fc983988f8692f7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ddb9f5ecea0d4612965a20abeb8dc076 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c061a8c754148bfbd2cb912843fb390 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6466f0a8f11e495db6802a6383957453 : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83911f2193bb4fa4bf62607b256db5cd : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d49198878d08408d82a32ef3527c228d : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6313482dc7e4492ebb66b784967fd2ca : SUCCESS in 9m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e3409da3b9aa44c18308605381226584 : SUCCESS in 6m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/74524d6df4ce43c2816509c05edcfae6 : SUCCESS in 12m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0e9146667db940bca2e2ec022c688f37 : SUCCESS in 19m 05s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b5ea8465f034f4dacfc3dcd64011fd9 : SUCCESS in 50m 53s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ffe8b77bdab4a84ae79a264a04db0ed : SUCCESS in 13m 33s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f7c1e574dd748a280e489a0beee8d8f : SUCCESS in 1h 12m 58s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/673426e5f8284a86adbab5d3443ea73d : SUCCESS in 16m 38s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f67dc76c213a4b5b97cdca7298a7dcf5 : SUCCESS in 23m 53s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ffb2e84a76024ed8bb5eceb9275c4073 : SUCCESS in 13m 10s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98a1618434ce4d5d9b465721c6c45887 : SUCCESS in 12m 49s","accounts_in_message":[],"_revision_number":2},{"id":"babfed10f1267420b13a28b3a5f950d8e2cfc4fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-07 21:42:57.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"e94d27afdfeee239bfc8746054421c0ec27edcc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-07 22:08:32.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent, and commit message as Patch Set 3.","accounts_in_message":[],"_revision_number":4},{"id":"967544cdc1d7ee0910e2cbd11248d916a41a7f9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 23:21:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b93514d00894559938711c45de91cec\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df535355e6c54488bfb826bbfadd29af : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29ec21793ab14a9994757d4c31825a96 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f52b397bbdc0463db62173a77f8f0cc9 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95df02a8d60a46bcab9e99389f152fff : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b16330254304a4a8952d0420992b570 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bc821c49cdb48ec8c108dcc32eafba7 : SUCCESS in 9m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d21213f1584f4e5894c39679221162d7 : SUCCESS in 6m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/43c8879fb0064cfd9913c701c7e1fec3 : SUCCESS in 13m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d7d67128bfd34372a69ff4f369fb402a : SUCCESS in 19m 39s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61bd551b1d1044559da056e7a961f613 : SUCCESS in 57m 08s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b44ab466700e4969888aee6d6d53d7b2 : SUCCESS in 12m 44s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2619d3a3124841e390a5348967dc2244 : SUCCESS in 1h 09m 39s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17846efc3cd04648a56b93159c058888 : SUCCESS in 17m 25s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/562ed6768335409fbf0aa1b684ab824b : SUCCESS in 28m 10s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c82716865ec8416f897b7b076990fbbf : SUCCESS in 12m 28s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85a933ca488144978aa8021a2d992c8e : SUCCESS in 12m 29s","accounts_in_message":[],"_revision_number":4},{"id":"f6a7986d147ac881199d974e14ed40e795f8a093","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-08 14:42:30.000000000","message":"Patch Set 4: Review-Priority+2","accounts_in_message":[],"_revision_number":4},{"id":"030b4119d45f4751b773e4159d6696b56c57de28","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2022-12-08 14:49:47.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"8e473e855a75ee488403669c6dcca2a61e1629cc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-13 08:45:15.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"58c1a41f1462c62ca8f78e9ec45f9a54576853ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 08:46:14.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9c5fc038a5985fa7fe06c6596d6122698ba5b093","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 10:05:19.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b43d13e7f6fc48d5a027b6011da4ff41\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57c428cb893e471a8e05ab94a1bd0ec8 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30a3f7f507bd4e63b354a49f597eacad : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62fe945338cd46a0942db6818fd627ec : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fbdbcf4e8af4494a95a953805ca7564 : SUCCESS in 9m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/55798b14a4d143f8afe4432f1a7b20b5 : SUCCESS in 6m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2d1f54db953843ce8af6829b9507cddc : SUCCESS in 13m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/72def35405964820b0b32522653f24fa : SUCCESS in 14m 57s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0fcd518caf644cb2802a3be6e0f7651f : SUCCESS in 1h 06m 35s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f7d6c8ae7d44d1fa0c404c2c64f7bc9 : SUCCESS in 49m 22s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e2e4584b234e489786a77e0e562a903d : SUCCESS in 1h 08m 07s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ef038b2f716f444f8cc9d70792a91e8c : SUCCESS in 17m 41s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26639b03b4ef4756baa91375fc63935a : SUCCESS in 17m 18s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a097768704647d796c4224ffbb683d3 : SUCCESS in 1h 05m 40s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8cfe8ddd5c0d42fd973cc41ab99bff30 : SUCCESS in 18m 50s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1842c8d215dd4dedae734a3ca91ad034 : SUCCESS in 18m 09s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e26c69e14cf4aa3bb7f8f1baab18b7e : SUCCESS in 23m 06s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a8cb5237813946fb8b7f7c16a6581a75 : SUCCESS in 17m 39s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b37196777084bcf972a467ff44d6ccf : SUCCESS in 12m 47s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0e395f7efdf042648465f2c9121446c3 : SUCCESS in 18m 20s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d93b3ab953654f34aa6d383245573eac : SUCCESS in 12m 26s","accounts_in_message":[],"_revision_number":4},{"id":"632bc54f50f1bee9c7d76cd0897f3ec937af4706","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 10:05:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"14bd9052ea57990ee9a8789936e4ebf6647cc513","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 10:07:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c26e084024f8409092a388e0b85828dd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a9ea6a9b0154315a5affa9589b8c5d4 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f3ab02280d20048a5c74346883ec2db076502435","revisions":{"95a6bf612d5ddfdea88b29a455d5914f77114173":{"kind":"REWORK","_number":1,"created":"2022-12-07 12:08:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/45/866845/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/866845/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/866845/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/866845/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/866845/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/866845/1"}}},"commit":{"parents":[{"commit":"494e1a39221d121e25d95a2fb08fd5a65e5e1b26","subject":"[release] Sync Kolla Ansible feature flags for Zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/494e1a39221d121e25d95a2fb08fd5a65e5e1b26"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 12:08:11.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 12:08:11.000000000","tz":60},"subject":"[release] Sync Kolla Ansible inventory for Zed","message":"[release] Sync Kolla Ansible inventory for Zed\n\nSyncs with Kolla Ansible inventory for Zed release, at commit\nce28cbc0d8bd907982212f4963b3a3e48215f688.\n\nChange-Id: If7171c925be052e59e64423d62c9416485ca9113\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95a6bf612d5ddfdea88b29a455d5914f77114173"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95a6bf612d5ddfdea88b29a455d5914f77114173"}]},"branch":"refs/heads/master"},"65baca18bf0f0667d9f3f37bb3de96224e11e985":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-12-07 15:16:00.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/45/866845/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/866845/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/866845/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/866845/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/866845/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/866845/2"}}},"commit":{"parents":[{"commit":"dbcced8be0c5fe267c295572904e1c6070b71e24","subject":"[release] Sync Kolla Ansible feature flags for Zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dbcced8be0c5fe267c295572904e1c6070b71e24"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 12:08:11.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 15:16:00.000000000","tz":0},"subject":"[release] Sync Kolla Ansible inventory for Zed","message":"[release] Sync Kolla Ansible inventory for Zed\n\nSyncs with Kolla Ansible inventory for Zed release, at commit\nce28cbc0d8bd907982212f4963b3a3e48215f688.\n\nChange-Id: If7171c925be052e59e64423d62c9416485ca9113\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65baca18bf0f0667d9f3f37bb3de96224e11e985"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65baca18bf0f0667d9f3f37bb3de96224e11e985"}]},"branch":"refs/heads/master","description":"Rebase"},"80286e5f335187817fc96fe4895a136d62f71536":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-12-07 21:42:57.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/45/866845/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/866845/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/866845/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/866845/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/866845/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/866845/3"}}},"commit":{"parents":[{"commit":"3d0e1a3818016bf74ef6a12ba31f16bf4c30cd93","subject":"[release] Sync Kolla Ansible feature flags for Zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d0e1a3818016bf74ef6a12ba31f16bf4c30cd93"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 12:08:11.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 21:42:57.000000000","tz":0},"subject":"[release] Sync Kolla Ansible inventory for Zed","message":"[release] Sync Kolla Ansible inventory for Zed\n\nSyncs with Kolla Ansible inventory for Zed release, at commit\nce28cbc0d8bd907982212f4963b3a3e48215f688.\n\nChange-Id: If7171c925be052e59e64423d62c9416485ca9113\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/80286e5f335187817fc96fe4895a136d62f71536"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/80286e5f335187817fc96fe4895a136d62f71536"}]},"branch":"refs/heads/master","description":"Rebase"},"f3ab02280d20048a5c74346883ec2db076502435":{"kind":"NO_CHANGE","_number":4,"created":"2022-12-07 22:08:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/45/866845/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/866845/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/866845/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/866845/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/866845/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/866845/4"}}},"commit":{"parents":[{"commit":"3d0e1a3818016bf74ef6a12ba31f16bf4c30cd93","subject":"[release] Sync Kolla Ansible feature flags for Zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d0e1a3818016bf74ef6a12ba31f16bf4c30cd93"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 12:08:11.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 21:39:08.000000000","tz":60},"subject":"[release] Sync Kolla Ansible inventory for Zed","message":"[release] Sync Kolla Ansible inventory for Zed\n\nSyncs with Kolla Ansible inventory for Zed release, at commit\nce28cbc0d8bd907982212f4963b3a3e48215f688.\n\nChange-Id: If7171c925be052e59e64423d62c9416485ca9113\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f3ab02280d20048a5c74346883ec2db076502435"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f3ab02280d20048a5c74346883ec2db076502435"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[]}
