)]}'
{"id":"openstack%2Fkayobe~866871","triplet_id":"openstack%2Fkayobe~master~I8102fafb00db178f1ae6801d37c43a39033cbfe6","project":"openstack/kayobe","branch":"master","topic":"remove-monasca","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-12-13 14:19:49.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-12-13 14:52:05.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2022-12-13 14:09:59.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-12-13 16:08:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8102fafb00db178f1ae6801d37c43a39033cbfe6","subject":"Remove support for Monasca and its dependencies","status":"MERGED","created":"2022-12-07 15:17:18.000000000","updated":"2022-12-13 16:09:17.000000000","submitted":"2022-12-13 16:08:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"866871-remove-monasca","meta_rev_id":"4584f958b08bfb28bf796826ee9c3b7a620e2206","_number":866871,"virtual_id_number":866871,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-13 16:08:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-12-13 14:19:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-12-13 14:52:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2022-12-13 14:19:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-12-13 14:52:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-12-13 14:19:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-07 16:19:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-08 14:42:49.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-12-13 11:00:22.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-12-13 14:01:22.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"cea3a77d5cabea01374c540b78847d9c4b533e82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-07 15:17:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c402fdbfdd165a743117a088ea80e630e50b206","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 16:19:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c09f02c2ac8d49c3a11c03045c98d053\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dd526677f7fa4e94b6ae2fe9cdd662f1 : SUCCESS in 2m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fc2da07db63435ebc0fd4d562112b3e : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98201ff56f5c432aa77f901a07fba2ff : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fa59da96f664f97be1568b1c6fd5c6c : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6926b192d4e34cf188b59ef508963551 : SUCCESS in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/256faedb56594e2ea176231c54078fb9 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/373c5701f6d740c9a4f7999281b2a02b : SUCCESS in 9m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88722333f2b749e8b0b089765898b59a : SUCCESS in 3m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f17cdc2d61ef446b9825505752574eb7 : SUCCESS in 6m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/73f8f910186c439c8b411fc33b62c5f7 : SUCCESS in 15m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7f45fd52d6db47b3869667b99dfc6fe6 : SUCCESS in 21m 38s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa2d5cedbedc400b82bbbdc934f60428 : SUCCESS in 49m 54s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c1a710d83c34cef842db57116146941 : SUCCESS in 18m 20s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/297a1e18f05341b9ba4381068c1c4d36 : SUCCESS in 1h 00m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8112e836b2a74eb5913d86ca6dc3c24d : SUCCESS in 18m 46s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94cb462da58e46248271347dfaa5b0a1 : SUCCESS in 24m 45s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ded3e8e9850482ea1ae9c7f1fada45b : SUCCESS in 12m 28s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/237573ccf03c4f76b26d1f136bcc14b5 : SUCCESS in 16m 02s","accounts_in_message":[],"_revision_number":1},{"id":"79c4e14586dafe25cfb2311008f6d530085f040e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-08 14:42:49.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"60f275e11ef6fefcbff82656859aee758910a58f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-13 08:46:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e716535c79d720908462c260e65386536763c8fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 10:07:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9717cabb5ed640ffac032a4b31158902\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a529bb62b7b8407a8527c8ab059214d9 : SUCCESS in 1m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9b0a468da1840a391ee464524cda7e8 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6da2629f819244e1b7aa1b3cc3168bd3 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6fccd125f044dfa9bc0b1cc51d4727d : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca274adcf4084000b1151d2726580339 : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a66b509d98847a6a28aa76cd02538b4 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfa2bd5511d44368a250fdb4cfc107f0 : SUCCESS in 10m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33dff6ae62ad4924878f5d0f806b3171 : SUCCESS in 3m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c778cb37068e44d3baf62f33b2001049 : SUCCESS in 5m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a846c534c7404b20bd2cdf6e386c659b : SUCCESS in 13m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9cbf6c33897245c28a86636ae7982af2 : SUCCESS in 16m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9778e8326d7b4222b9c6679caf0a812a : SUCCESS in 54m 20s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/457d4f4812e247169d0ce92a50be4b44 : SUCCESS in 1h 00m 55s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/527faed10bb24585a173b814bb5caa64 : SUCCESS in 54m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/def115007eca41bba32b632dc447fd19 : SUCCESS in 1h 10m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d01c7a160caf46a49472b610cb899dbe : SUCCESS in 15m 18s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e5b3eccb648840d7b214610ec27c3edb : SUCCESS in 21m 00s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d900ded1baca4c4489857c9b365fc814 : SUCCESS in 16m 12s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87bcef19e5e54dad9e3ac8fd4c2b54e0 : SUCCESS in 59m 23s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e7e565e50d9f426284e509c5ce93b31c : SUCCESS in 19m 16s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70f58d326b934c648432be58ad5afac0 : SUCCESS in 19m 18s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/3020c18e4d1b4cb58ad4ea15179cc1aa : SUCCESS in 43m 13s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4053d86fc5b44f61a82adc4a93aa107f : SUCCESS in 29m 28s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/42bf1b9b5bef4d18a6273df3aacd347d : SUCCESS in 16m 35s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96b2680d375b43a19889ab1e1c59290b : SUCCESS in 10m 58s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d9cf4ced32a747358e2403480de97af4 : SUCCESS in 16m 28s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3779257c9be4b12945aa58c3f60f9d5 : SUCCESS in 15m 37s","accounts_in_message":[],"_revision_number":2},{"id":"832bf2d44d5743651b76da6941e1b5ffc135cc58","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-12-13 11:00:22.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f90a81e816c43ee86e37489475a863ad85b04ee0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-13 12:44:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8533b866af1493e5043bb617e2ae917c939a1b5f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-13 12:45:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6aedec724c11e067242c77ae225ed889881e19b7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-13 12:49:33.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a4aa22935118d5dc5b65df5a0dafb0d0aa42c4f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-13 13:32:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e18ccc66311a7a387f8a5cea692b77a5d5a56c79","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-13 13:33:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a694c583ea59db322e71eb5681290ee16bda6628","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2022-12-13 14:01:22.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1509fb6a8bf36cd3f6e78c1efaac28028e8aa747","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-13 14:06:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fe92123f1a399f55563671fc4fabaeea1d9d35b0","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2022-12-13 14:09:59.000000000","message":"Patch Set 4: -Code-Review","accounts_in_message":[],"_revision_number":4},{"id":"7071fd9555cc165659cfda06a62cf8620e3c8da4","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2022-12-13 14:19:15.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"45ae17eada3cc2f5e883ff38a22ae0a29d54d18f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-13 14:19:49.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":4},{"id":"550d4ec487b352ab9458d59241fcc21c7610a4e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 14:46:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c09f2d776f92418a88567c9fc6145013\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c843dcb77aa84450bb4360c8d7094fbc : SUCCESS in 1m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/369896c84f684debb348f78ee5f97143 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f94566c7b0c48a58884a4af10a8981f : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09e8ff457300416e8b3065fbd95c8edf : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a11611a275b46eeaaa8354824730e4c : SUCCESS in 6m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4c2f11afbb646c7b8c6527c66cffb1f : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/194ab7acec054aa6b3e2d66ab47a75a2 : SUCCESS in 10m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8831ab905b3647aca5aa64c2cf040c1b : SUCCESS in 4m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bde8d1569db54197a1b468b1d5439976 : SUCCESS in 5m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d7476caf680041a88d6ad6732c2c9ee3 : SUCCESS in 14m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4d711933eeb94a3192ee6f34358e0e32 : SUCCESS in 17m 35s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/17831138fcd347be9e9b043bc1c36791 : SUCCESS in 59m 38s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/a9019738ab474b52a9249aed132e1de9 : SUCCESS in 1h 01m 15s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/faf63a46fec6416c936853b0ac9c4a60 : SUCCESS in 45m 31s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/46f09ab64c4b43e4bba9090d6a38e6d4 : SUCCESS in 1h 05m 41s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b5af84c237a84dcb9fb51c892d1bcd5c : SUCCESS in 23m 44s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/db2adb764ae5428bb42064f883995e93 : SUCCESS in 19m 45s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2f51498e1ae4d79a00fd60e9e27a481 : SUCCESS in 19m 47s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/946e275a074444c98dc1e5c807791004 : SUCCESS in 1h 11m 47s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/082f54783f9443c38db46d2a741f5dfd : SUCCESS in 21m 45s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69599f2483914832aa27d568e20ce644 : SUCCESS in 20m 50s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8f52baa671f54872ad2cf6a6f48c6f2e : SUCCESS in 37m 55s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f0a65ee5d794e248878a892781c847f : SUCCESS in 29m 16s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/da6c2a7f28cd471dbf44666aa7409c0e : SUCCESS in 14m 54s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b619b588b6214e23ac9c38c2c530aab6 : SUCCESS in 13m 56s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/874e2ce60dd048728bfe5e515d8be050 : SUCCESS in 16m 15s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/58be18671cd34fbfa17f9fbccf081da0 : SUCCESS in 12m 43s","accounts_in_message":[],"_revision_number":4},{"id":"1c497cd7a636bffa62b1189d4d1d8216fe959111","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-12-13 14:52:05.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2c03f4d9ca919381666c05e10af004146f025eca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 14:52:32.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1fce194c7be13f61f7aa9913bfa388ea0e414503","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 16:08:08.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3417a017e6814a858b745552a268f71a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/81fe883734ab4b7f9bf34a6fc29af20f : SUCCESS in 2m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/471fb666e71a40448e327f3962e3fa85 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0faf3b448fc4dd2b98a09f19ab4d966 : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/775cb78312ea47d3aed1fe4842a6e0be : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f11427fd0354b64818f74fb6e64dd3c : SUCCESS in 10m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8619f3d5bdd84638b87b708a0f5337d7 : SUCCESS in 2m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d06243b675d9479aa3cba4396f85efb9 : SUCCESS in 6m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f4e5615d68194b0b8828c9c8f193efc7 : SUCCESS in 13m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/521db99289f74219b02184b66bfd1e47 : SUCCESS in 20m 04s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/359f034646fb4d749f87aa7fad84a757 : SUCCESS in 54m 46s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd5b58081efc4f51944617cc0dc12382 : SUCCESS in 1h 05m 23s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ae10d7f443994958b51ab0197cf1e1ad : SUCCESS in 1h 05m 44s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5c184c4b85ef4d6480446ce150075885 : SUCCESS in 20m 53s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ae6a03071b64fedb28e46fa3356726f : SUCCESS in 21m 36s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94f14c9276164a15a8cc81799bd7945e : SUCCESS in 1h 14m 22s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4a8b9ed833624f49b41c19897fbb960e : SUCCESS in 20m 17s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4642348762ca43979ba5e988bcadb2fd : SUCCESS in 18m 16s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bce0ca960dd44d859bb4ff195cbc6a0c : SUCCESS in 27m 38s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/712bbdd6c4c44a618cc4f9ddf9428111 : SUCCESS in 15m 29s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/471f6b0a40bc42cd947ac0be9aa06cb8 : SUCCESS in 14m 31s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b656cadc07df45e085ef6224eabc45f0 : SUCCESS in 16m 34s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/928be0bc15a8413aa3883537d2fd5feb : SUCCESS in 12m 10s","accounts_in_message":[],"_revision_number":4},{"id":"962dee41d4f7ef2ca857147a7a21224567016971","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 16:08:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"4584f958b08bfb28bf796826ee9c3b7a620e2206","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 16:09:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6239ede50b7492da212c0defcc5c7a7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b39f9e90261641fe8ab7d9ab2282b660 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8f29a7cd72e466d8c4cafe5375d25a9 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f6660bbff87684ec3ee09fe124bb2c71a3074225","revisions":{"66d63320491f5647ae6aa2f25f57b3c331fc82b4":{"kind":"REWORK","_number":1,"created":"2022-12-07 15:17:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/71/866871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/866871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/866871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/866871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/866871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/866871/1"}}},"commit":{"parents":[{"commit":"65baca18bf0f0667d9f3f37bb3de96224e11e985","subject":"[release] Sync Kolla Ansible inventory for Zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65baca18bf0f0667d9f3f37bb3de96224e11e985"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 15:14:50.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 15:16:53.000000000","tz":60},"subject":"Remove support for deploying Monasca","message":"Remove support for deploying Monasca\n\nThis follows removal of support from Kolla Ansible. This also removes\nsupport for configuring Grafana with overcloud post configure.\n\nChange-Id: I8102fafb00db178f1ae6801d37c43a39033cbfe6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/66d63320491f5647ae6aa2f25f57b3c331fc82b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/66d63320491f5647ae6aa2f25f57b3c331fc82b4"}]},"branch":"refs/heads/master"},"d16eaa3046f6a905f72bc890faa63a3b803646bc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-12-13 08:46:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/71/866871/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/866871/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/866871/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/866871/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/866871/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/866871/2"}}},"commit":{"parents":[{"commit":"f3ab02280d20048a5c74346883ec2db076502435","subject":"[release] Sync Kolla Ansible inventory for Zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f3ab02280d20048a5c74346883ec2db076502435"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 15:14:50.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-12-13 08:46:04.000000000","tz":0},"subject":"Remove support for deploying Monasca","message":"Remove support for deploying Monasca\n\nThis follows removal of support from Kolla Ansible. This also removes\nsupport for configuring Grafana with overcloud post configure.\n\nChange-Id: I8102fafb00db178f1ae6801d37c43a39033cbfe6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d16eaa3046f6a905f72bc890faa63a3b803646bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d16eaa3046f6a905f72bc890faa63a3b803646bc"}]},"branch":"refs/heads/master","description":"Rebase"},"00dd84975e5b94788eed2d9dff669f2b66c7708b":{"kind":"REWORK","_number":3,"created":"2022-12-13 12:44:40.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/71/866871/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/866871/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/866871/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/866871/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/866871/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/866871/3"}}},"commit":{"parents":[{"commit":"ad1a107fa55c914891f042fa51f10cb137cd8a99","subject":"Merge \"Support OpenSearch configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad1a107fa55c914891f042fa51f10cb137cd8a99"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-13 12:44:18.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-13 12:44:18.000000000","tz":60},"subject":"Remove support for deploying Monasca","message":"Remove support for deploying Monasca\n\nThis follows removal of support from Kolla Ansible. This also removes\nsupport for configuring Grafana with overcloud post configure.\n\nChange-Id: I8102fafb00db178f1ae6801d37c43a39033cbfe6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/00dd84975e5b94788eed2d9dff669f2b66c7708b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/00dd84975e5b94788eed2d9dff669f2b66c7708b"}]},"branch":"refs/heads/master"},"f6660bbff87684ec3ee09fe124bb2c71a3074225":{"kind":"REWORK","_number":4,"created":"2022-12-13 13:32:51.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/71/866871/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/866871/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/866871/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/866871/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/866871/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/866871/4"}}},"commit":{"parents":[{"commit":"ad1a107fa55c914891f042fa51f10cb137cd8a99","subject":"Merge \"Support OpenSearch configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad1a107fa55c914891f042fa51f10cb137cd8a99"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-13 13:32:33.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-13 13:32:33.000000000","tz":60},"subject":"Remove support for Monasca and its dependencies","message":"Remove support for Monasca and its dependencies\n\nThis follows removal of support from Kolla Ansible. This also removes\nsupport for configuring Grafana with overcloud post configure.\n\nChange-Id: I8102fafb00db178f1ae6801d37c43a39033cbfe6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6660bbff87684ec3ee09fe124bb2c71a3074225"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6660bbff87684ec3ee09fe124bb2c71a3074225"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[]}
