)]}'
{"id":"openstack%2Fkayobe~867676","triplet_id":"openstack%2Fkayobe~master~Ibc2b9b647766231029fdc2c0e3671e872b1ae366","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-12-15 09:28:11.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-12-15 09:35:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibc2b9b647766231029fdc2c0e3671e872b1ae366","subject":"Some Rocky Linux 9 cleanups","status":"MERGED","created":"2022-12-14 16:27:03.000000000","updated":"2022-12-15 09:36:32.000000000","submitted":"2022-12-15 09:35:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"867676","meta_rev_id":"89e6811211c562a8c616e74d33aed68b0dd4d7ce","_number":867676,"virtual_id_number":867676,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-15 09:35:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-12-15 09:28:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-15 09:22:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-12-15 09:27:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-14 16:35:27.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2022-12-14 17:47:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-14 21:43:44.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-12-15 08:04:53.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-12-15 09:28:11.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"61df007b6e1d462821a2af271f47336395daef5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-12-14 16:27:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c408c46f1fc679ddbdb030bb6aded57e5b591d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 17:47:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45b55fd9c9d74ba48e34c3acd500c365\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fa0a6bf5fb942b590a95deb953665f4 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2436201ab7da4f0493e99d9554e89f9e : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c70c34392d842c0bf223ad536acc3d7 : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50fc2603617541b89eda2b8f727495e4 : SUCCESS in 6m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/afff7752dc8946f4a2d161d1caff44fc : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/220ec4fc47014cc6b57eb842056f114a : SUCCESS in 10m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/53d00de2f7414f13a8cd1c5376489d02 : SUCCESS in 7m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fb83713c56b54d2481cb662b9f4f2b99 : SUCCESS in 15m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e184b8bbfcaf4610bd48c7b7d7bd7147 : SUCCESS in 16m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/168d6b65ce694617a3c47ff41173e450 : SUCCESS in 1h 12m 45s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/52170500d8004ae4950b2777196e7812 : FAILURE in 19m 23s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/524a497d2d8a4862b852876ff59654ab : SUCCESS in 1h 04m 10s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3e6ef77765474c78b079200ed3847e48 : SUCCESS in 1h 13m 34s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/93b1bde924eb4962bac6e777eff13b68 : SUCCESS in 21m 28s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/31f927ab72f8483cb7d6571e340631a6 : SUCCESS in 23m 39s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e98086e483b49cf8cde9686da814a9c : SUCCESS in 18m 10s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba01612d42d442669df6459920fda69c : SUCCESS in 1h 12m 16s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/57c9fd62e97146fca02ef1086003639f : SUCCESS in 21m 30s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b9b348c9c154846ac9b1b13458a2a85 : SUCCESS in 19m 00s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ed6c1a507077441d9529c5ad8663c9d6 : SUCCESS in 38m 24s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e807af4f3b434077a1b977b28e1ace58 : SUCCESS in 30m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/35ccddd1505c4a5e901fd6603db3dceb : SUCCESS in 17m 09s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f41e09f8c011437a8701a921d192b3b8 : SUCCESS in 15m 15s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/eeb7184cc8d44a409f7e0c7995641a40 : SUCCESS in 17m 28s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4288b47598140d2bbc2004dae4b9435 : SUCCESS in 15m 51s","accounts_in_message":[],"_revision_number":1},{"id":"e1698f9e5ae4be086138d050d86ed94ad165c119","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-14 21:43:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"71186b2f37f5297d6e9905079962c919bc951874","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-15 08:04:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"43a85c955550ebc34c06ca6e8116b7d7f262f2e9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-15 08:16:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b3a91ba982ff8d6e782d6acef61fc67f529e71eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-15 08:32:02.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"349fe62eb54bd8f81ad28fc186b00da6e9ac81c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 08:40:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59cade8984644650afdc6d7ed303cb69\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4652912514444afc8827e290aaa598cc : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7f551d03079496680a5e89fc25be05d : SUCCESS in 7m 08s","accounts_in_message":[],"_revision_number":2},{"id":"a071ef4f85eb074fa23987d144fb71538791d1c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-15 08:59:36.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"79e1cab355c16dff57b3fb7b45ba02453112975a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-15 09:00:09.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"e2e903d3c4d99e95e2b09e760abaab7453a4d027","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 09:06:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b3057634a3d4c10a404b83f26cb92fd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d05b59fe3a684219ae1c9296641804da : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/072a3c229afb4d3a99765cea6c95faa5 : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":4},{"id":"1844334a79157b97438a6dfed1ddbaa23fc46bad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-15 09:13:57.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"13d50e032625ca28e0e44e13f5b6e8534fce26c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-15 09:14:46.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b107b2141d10ff28b21e336933b6ffc6450646f8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-15 09:22:38.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"398cbcf6d68b46f82ef1e63f47688eafeff9abb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 09:22:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a785c58960f49c48e8823d435a44831\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c41bd704b5144ab822140eb85ec8577 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ade83d8cf8047c18343af7740caa771 : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":6},{"id":"14db5e10f97ea50292189c8c9857c8f87b37108e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-15 09:27:29.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"079be34b8188ecb71fdbf2e06394dd1bd334d41d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 09:27:49.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"5989305b070d4927049aab45517d99b878861ba4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-15 09:28:11.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"226038183020481fdf54d60af52444581259fc12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 09:35:29.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0db53c929c284449b33d2117ed848b67\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/966ba6f1ebaa4b288f75994de61977fd : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a2dc5b55dc74813b479a5c0d5263242 : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":6},{"id":"25feb2cfe49afaa0199cd805b43d21658d3aca3f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 09:35:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"89e6811211c562a8c616e74d33aed68b0dd4d7ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 09:36:32.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/debc9b623dcd42cd8bad28568f52d43b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a895e425a2234524a9eddbfb75a3e93a : SUCCESS in 40s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"5846d81c53b75966fb84b29813b7a98bbeacdd8d","revisions":{"b20cbe6fce7bb295dbf40a6b065c424f2dc61633":{"kind":"REWORK","_number":1,"created":"2022-12-14 16:27:03.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/76/867676/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/76/867676/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/867676/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/867676/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/867676/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/76/867676/1"}}},"commit":{"parents":[{"commit":"4b20bb62a8746f87b24aeb7c5b5db8a81b27b2c6","subject":"Merge \"CI: Add kayobe-seed-images-centos9s to experimental jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4b20bb62a8746f87b24aeb7c5b5db8a81b27b2c6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-12-14 16:26:57.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-12-14 16:26:57.000000000","tz":0},"subject":"Some Rocky Linux 9 cleanups","message":"Some Rocky Linux 9 cleanups\n\n* Update support matrix notes\n* Drop kolla_base_distro CI overrides\n\nChange-Id: Ibc2b9b647766231029fdc2c0e3671e872b1ae366\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b20cbe6fce7bb295dbf40a6b065c424f2dc61633"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b20cbe6fce7bb295dbf40a6b065c424f2dc61633"}]},"branch":"refs/heads/master"},"a4bb32e9628d486f9d4cc68ba1c71acf93c4db38":{"kind":"REWORK","_number":2,"created":"2022-12-15 08:32:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/76/867676/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/76/867676/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/867676/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/867676/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/867676/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/76/867676/2"}}},"commit":{"parents":[{"commit":"4b20bb62a8746f87b24aeb7c5b5db8a81b27b2c6","subject":"Merge \"CI: Add kayobe-seed-images-centos9s to experimental jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4b20bb62a8746f87b24aeb7c5b5db8a81b27b2c6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-12-14 16:26:57.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-12-15 08:31:57.000000000","tz":0},"subject":"Some Rocky Linux 9 cleanups","message":"Some Rocky Linux 9 cleanups\n\n* Update support matrix notes\n* Drop kolla_base_distro CI overrides\n\nChange-Id: Ibc2b9b647766231029fdc2c0e3671e872b1ae366\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a4bb32e9628d486f9d4cc68ba1c71acf93c4db38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a4bb32e9628d486f9d4cc68ba1c71acf93c4db38"}]},"branch":"refs/heads/master"},"d282c6ec79636bd314b086c6d37862f490e89f49":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-12-15 08:59:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/76/867676/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/76/867676/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/867676/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/867676/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/867676/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/76/867676/3"}}},"commit":{"parents":[{"commit":"4b20bb62a8746f87b24aeb7c5b5db8a81b27b2c6","subject":"Merge \"CI: Add kayobe-seed-images-centos9s to experimental jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4b20bb62a8746f87b24aeb7c5b5db8a81b27b2c6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-12-14 16:26:57.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-15 08:59:36.000000000","tz":0},"subject":"Some Rocky Linux 9 cleanups","message":"Some Rocky Linux 9 cleanups\n\n* Update support matrix notes\n\nChange-Id: Ibc2b9b647766231029fdc2c0e3671e872b1ae366\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d282c6ec79636bd314b086c6d37862f490e89f49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d282c6ec79636bd314b086c6d37862f490e89f49"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ca9fd31be4d12d43d028cbf828ec53bad205094f":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-12-15 09:00:09.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/76/867676/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/76/867676/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/867676/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/867676/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/867676/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/76/867676/4"}}},"commit":{"parents":[{"commit":"4b20bb62a8746f87b24aeb7c5b5db8a81b27b2c6","subject":"Merge \"CI: Add kayobe-seed-images-centos9s to experimental jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4b20bb62a8746f87b24aeb7c5b5db8a81b27b2c6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-12-14 16:26:57.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-15 09:00:09.000000000","tz":0},"subject":"Some Rocky Linux 9 cleanups","message":"Some Rocky Linux 9 cleanups\n\n* Fix description of kolla_base_distro\n* Update support matrix notes\n\nChange-Id: Ibc2b9b647766231029fdc2c0e3671e872b1ae366\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca9fd31be4d12d43d028cbf828ec53bad205094f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca9fd31be4d12d43d028cbf828ec53bad205094f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6100ff65d979ac89ce6138058d816a5d47c24501":{"kind":"REWORK","_number":5,"created":"2022-12-15 09:13:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/76/867676/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/76/867676/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/867676/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/867676/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/867676/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/76/867676/5"}}},"commit":{"parents":[{"commit":"4b20bb62a8746f87b24aeb7c5b5db8a81b27b2c6","subject":"Merge \"CI: Add kayobe-seed-images-centos9s to experimental jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4b20bb62a8746f87b24aeb7c5b5db8a81b27b2c6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-12-14 16:26:57.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-12-15 09:13:55.000000000","tz":60},"subject":"Some Rocky Linux 9 cleanups","message":"Some Rocky Linux 9 cleanups\n\n* Fix description of kolla_base_distro\n* Update support matrix notes\n\nChange-Id: Ibc2b9b647766231029fdc2c0e3671e872b1ae366\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6100ff65d979ac89ce6138058d816a5d47c24501"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6100ff65d979ac89ce6138058d816a5d47c24501"}]},"branch":"refs/heads/master"},"5846d81c53b75966fb84b29813b7a98bbeacdd8d":{"kind":"REWORK","_number":6,"created":"2022-12-15 09:14:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/76/867676/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/76/867676/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/867676/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/867676/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/867676/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/76/867676/6"}}},"commit":{"parents":[{"commit":"4b20bb62a8746f87b24aeb7c5b5db8a81b27b2c6","subject":"Merge \"CI: Add kayobe-seed-images-centos9s to experimental jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4b20bb62a8746f87b24aeb7c5b5db8a81b27b2c6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-12-14 16:26:57.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-12-15 09:14:40.000000000","tz":60},"subject":"Some Rocky Linux 9 cleanups","message":"Some Rocky Linux 9 cleanups\n\n* Fix description of kolla_base_distro\n* Update support matrix notes\n\nChange-Id: Ibc2b9b647766231029fdc2c0e3671e872b1ae366\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5846d81c53b75966fb84b29813b7a98bbeacdd8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5846d81c53b75966fb84b29813b7a98bbeacdd8d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
