)]}'
{"id":"openstack%2Fkayobe~867923","triplet_id":"openstack%2Fkayobe~stable%2Fzed~Ie1c9e1ada50774cbcaec528bd64d8df066c21d2e","project":"openstack/kayobe","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-12-16 19:29:07.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2022-12-16 20:52:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie1c9e1ada50774cbcaec528bd64d8df066c21d2e","subject":"Run handlers before configuring veth interfaces","status":"MERGED","created":"2022-12-16 18:15:20.000000000","updated":"2022-12-16 21:35:27.000000000","submitted":"2022-12-16 20:52:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867923","cherry_pick_of_change":867965,"cherry_pick_of_patch_set":1,"meta_rev_id":"722ddb4eeee18ce3e3a7d559390a6e20a69dabff","_number":867923,"virtual_id_number":867923,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-16 20:52:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-12-16 19:29:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-16 21:35:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-12-16 19:29:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-16 18:15:20.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-12-16 18:15:49.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-12-16 19:38:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-16 21:35:27.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"4b7864ccf77a24bf72d896d4236b519337db0dc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2022-12-16 18:15:20.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5005e68179441b32ec962277bfb86ff1ddf11835","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-16 19:29:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"960e67516f85f1b43e95b9ac1b898554318ad981","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 19:38:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3d803d8198243ce8e7121e0f2169264\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a8375a1d1d64b199142a73cb0de92fa : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d54a32de5159405e9090bc87b503e3ad : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d522525e422d405f803e60c16d56ebd1 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a1c18b77e5b483f9820d35f62bb70c5 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86eb092972b9418381ac8ebb44452955 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d66604d12af43368857871e6e6ed032 : SUCCESS in 7m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d534e9a7aaa44e3385d4cb30364ea258 : SUCCESS in 4m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/84406c32eae34a5dba7f56a1921b0a19 : SUCCESS in 11m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7e75c5ce2b0a4b6babfe027f7bc85572 : SUCCESS in 15m 38s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/44ddfec5b871417288cb9909eba61214 : SUCCESS in 48m 30s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1454f4e4d48f421985fe08ab467a6fe1 : SUCCESS in 1h 11m 44s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2b34f168cd1405598123bbc1f10d243 : SUCCESS in 53m 02s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f9ccab12239a4385ad9bbbd9de0f654d : SUCCESS in 1h 03m 29s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/13913d99749249d49b21bf1d3635be07 : SUCCESS in 15m 23s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/32ed9106dba24bcaae4d111d4a87885f : SUCCESS in 18m 26s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17b1980ff09b45f3a00901819c6facc9 : SUCCESS in 11m 23s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c15a9e64a67f4998afd121dd7c2eebef : SUCCESS in 1h 17m 51s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4cf5ce47b02849a0b97869d5d5f24bf3 : SUCCESS in 17m 19s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f853956b467e42d48785f7e7094b5ab1 : SUCCESS in 13m 58s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/5cba5db820d64622b0ec0d46554e46cc : SUCCESS in 36m 13s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b617d1454d045d3ad0d94794c31af12 : SUCCESS in 27m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2b6eddefb6c24e40b301048edeb77017 : SUCCESS in 13m 05s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93283c95ff214a009c1b0a1a5362b942 : SUCCESS in 10m 13s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/99097a83e2e241b599164a4e6cace793 : SUCCESS in 14m 43s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2653ba14693640d79881cf292e27ff2d : SUCCESS in 13m 30s","accounts_in_message":[],"_revision_number":1},{"id":"e08452685f95f449fb3474c823b9b01c9ef2d729","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 19:38:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b138d5c557382ee23e3f23dc8be1f50e2ac61014","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 20:52:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/522afceb431f44d09785208c23ba2a56\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf138528fca84e7e82310fdd75852cbd : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9349f35af16b44e59ea8e54e1a4c7cf5 : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26910cf660144802ba17264bfc361753 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32f5063ff2304380af2192820fac0867 : SUCCESS in 6m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3390508934234a98a1c27020d074d5ca : SUCCESS in 3m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6b2100687264466696536a020dbbff02 : SUCCESS in 11m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5e3f390309ac4798bfbf3a37eee7d19a : SUCCESS in 13m 36s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/eb962b0248b4428c93180855fe654874 : SUCCESS in 1h 08m 36s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75ec11498abc4bfe8d3b3dc4c19c3541 : SUCCESS in 59m 15s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ba99ca96f5e34ae98421711dba4c54eb : SUCCESS in 57m 12s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f27bc826b44c4ec6a11a5d723eb07c78 : SUCCESS in 17m 28s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a26f7061a1bb453897ae6abd097d8127 : SUCCESS in 11m 07s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ac957b31ea342dc9677140294710875 : SUCCESS in 1h 11m 55s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9d2f5c8e86634904a44e581e7ea3468b : SUCCESS in 16m 20s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7cf925793d364a499e2a2a3e73529420 : SUCCESS in 15m 20s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/677f5ca0e9fd4793b9d09a355b3cbe72 : SUCCESS in 24m 10s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/43d350f25fc140f2b5685101a9f27249 : SUCCESS in 14m 29s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/892c02212672458a8a6a3886e07e4182 : SUCCESS in 10m 42s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d4043396609742048b74e08041178ed9 : SUCCESS in 14m 41s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/922b98393a5d4b7d84883ed8ed5f8390 : SUCCESS in 13m 11s","accounts_in_message":[],"_revision_number":1},{"id":"786ec24d1d92065563580483dda08d130be98ea5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 20:52:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f19ac6716d66d6be4448deb25f426de97c04a48a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 20:53:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a214aa665944048bd7bcf88be60f922\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35ab70ffedba4ad2a8ec4eb74336a492 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1},{"id":"722ddb4eeee18ce3e3a7d559390a6e20a69dabff","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-16 21:35:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"faf67f0eddf251939c7d48871740f73e0574de55","revisions":{"faf67f0eddf251939c7d48871740f73e0574de55":{"kind":"REWORK","_number":1,"created":"2022-12-16 18:15:20.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/23/867923/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/867923/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/867923/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/867923/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/867923/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/867923/1"}}},"commit":{"parents":[{"commit":"0a69b7ca178c357f383a829555b90bb05f974ba0","subject":"Merge \"Update .gitreview for stable/zed\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a69b7ca178c357f383a829555b90bb05f974ba0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-16 10:26:28.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-12-16 18:15:20.000000000","tz":0},"subject":"Run handlers before configuring veth interfaces","message":"Run handlers before configuring veth interfaces\n\nHandlers from MichaelRigart.interfaces are not run until the end of the\nplay, which results in the veth role trying to create veth interfaces\nagainst a non-existing bridge.\n\nChange-Id: Ie1c9e1ada50774cbcaec528bd64d8df066c21d2e\n(cherry picked from commit fbf000689559c434157d49b147da60725ba9562a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/faf67f0eddf251939c7d48871740f73e0574de55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/faf67f0eddf251939c7d48871740f73e0574de55"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
