)]}'
{"id":"openstack%2Fkayobe~867965","triplet_id":"openstack%2Fkayobe~master~Ie1c9e1ada50774cbcaec528bd64d8df066c21d2e","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2022-12-16 11:02:03.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-12-16 16:25:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie1c9e1ada50774cbcaec528bd64d8df066c21d2e","subject":"Run handlers before configuring veth interfaces","status":"MERGED","created":"2022-12-16 10:26:53.000000000","updated":"2022-12-16 16:26:08.000000000","submitted":"2022-12-16 16:25:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867965","meta_rev_id":"97e3ab04c149cf127b7a523fdfcdf2d8c5fd5576","_number":867965,"virtual_id_number":867965,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-16 16:25:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-12-16 10:51:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-12-16 11:02:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-12-16 15:01:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-16 10:44:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-16 10:51:32.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-12-16 11:01:29.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2022-12-16 11:38:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c5538bfff83a4b2e8459ab053b8da73289dc0c31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-16 10:26:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9554ce471558f0f2fcc7c7eb4b1715b9903c44c8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-12-16 10:44:15.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"42f2f929b3ee67cb58dc115fe8fdbda0d391fc8e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-16 10:51:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"87f72c002e66ee07316595cceb08374d34a16aa7","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2022-12-16 11:02:03.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"220e9b8c8cf06720bc3fb5ee57ceafa1d19fda69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 11:38:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8460e17389334772b6ba92a60914c86e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1351f2bf82224b679575484be2ac5fd3 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/504c1e4dd10449ecb90b69b019457d72 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ddb904cf946f444d966e24e327f34e65 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71ca7aea23b54cfd925063f337341a17 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bc4d31dff5e4e8c821262ece93dc111 : SUCCESS in 7m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d4f329a7fa1842288597896940e76af0 : SUCCESS in 5m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/387c891116e14b13a8876428ae3711fc : SUCCESS in 10m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1a8f16e70ebb4b398317b581b8431a17 : SUCCESS in 13m 06s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d1a50e248689422abce7bac5f2fd5df0 : SUCCESS in 47m 15s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/2c7dda707cca41a58165311e742b7df2 : SUCCESS in 52m 07s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a34fc4312a64b3092485003a9e8e50a : SUCCESS in 48m 45s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b3f61415a8a94924981f2dc154bdefbe : SUCCESS in 58m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3ef8e31fb41f4b1bbbc312a794d0b921 : SUCCESS in 13m 46s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/4b01b69f00424b7c870d7919f446db78 : SUCCESS in 16m 02s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c9b3d3c65384badb4ae424c87f43ab5 : SUCCESS in 12m 41s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0ce469bf55a432aaea3f35da2bcf827 : SUCCESS in 1h 10m 34s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/fece6dcfdcb64c6a8387c97949493d45 : SUCCESS in 15m 51s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68799290aa0a4b2c8d7b2f6dda62bc6a : SUCCESS in 14m 07s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/0ff7366837e245a7a49bde74fdf367ec : SUCCESS in 32m 57s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d359e18efd9340f497bd78396168a54c : SUCCESS in 28m 45s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b6aef3784f5e4531a1a8f06aa6a92849 : SUCCESS in 12m 20s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e814e8db13b84114b6b55887581de2cc : SUCCESS in 12m 33s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8b8363ba935e4d36bf251de534989350 : SUCCESS in 13m 53s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59cf07f4cec343b5bc9e9502b06adbe9 : SUCCESS in 15m 41s","accounts_in_message":[],"_revision_number":1},{"id":"5b1884c80fe90d08e283284abdb537d957d5aeee","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-16 15:01:32.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a2547bbf06f1633450a1594e59a6582899447e5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 15:01:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"708e72aa4a31a1f428ba84ada0c3643e1ce3f771","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 16:25:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/304676ac350a45c2b57fbb17bf0ab31b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9570297ca1a44de687602db0e0d0358f : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5febf4a0ad0e4ac7b9fa2d1dc720acfe : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/32c65457f8264ef9a0fd21ad8dff5a99 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58fade6244c54f3d9099cc01cdb7eef4 : SUCCESS in 8m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/18bea92677864558856aa6e51b7b39b6 : SUCCESS in 4m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5361104800664f6a9a95b2312e613fa0 : SUCCESS in 12m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ee54e177ac064ceba64c6de12c1e53d4 : SUCCESS in 17m 06s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/74091c2d3f5c4bd680ad121638d64487 : SUCCESS in 58m 33s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11544f51d92c4fcca23a52507372dfe3 : SUCCESS in 56m 12s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2188c2e486a5488aa33123cea7b385b9 : SUCCESS in 1h 01m 54s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a07f71caa570465683065584db65535c : SUCCESS in 18m 39s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7c018b99c2a4271bd47583a8d1f408b : SUCCESS in 15m 50s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cbcbbd5a7e70414ba13bf1129efa0d08 : SUCCESS in 1h 16m 17s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9c50c1d957a640c9ad2de0d600f99003 : SUCCESS in 17m 18s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/794c49cf6dcd4072bf9840f250cabe72 : SUCCESS in 15m 13s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d4e163bcf464dc5be5da6c41872f800 : SUCCESS in 29m 37s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fc932f66f929448bbca9d96189500244 : SUCCESS in 13m 08s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/273aaea375604e3289873751509a6c28 : SUCCESS in 12m 01s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/488d08b4381b441699decc08f77c2bb9 : SUCCESS in 18m 40s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34cea7c9924c4a85b9ec809af0dc4e05 : SUCCESS in 13m 53s","accounts_in_message":[],"_revision_number":1},{"id":"3bd032e2d36b0832dec35ea1016b1aeaad01e2b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 16:25:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"97e3ab04c149cf127b7a523fdfcdf2d8c5fd5576","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 16:26:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f2ef62d0c41423ebb6075250eae915c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eea0123df93c4622a33f4cfbe652916d : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fbf000689559c434157d49b147da60725ba9562a","revisions":{"fbf000689559c434157d49b147da60725ba9562a":{"kind":"REWORK","_number":1,"created":"2022-12-16 10:26:53.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/65/867965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/867965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/867965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/867965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/867965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/867965/1"}}},"commit":{"parents":[{"commit":"3d1ad7d3a99a8bb13b597df8ad28ebb15d998612","subject":"Merge \"Switch to 2023.1 Python3 unit tests and generic template name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d1ad7d3a99a8bb13b597df8ad28ebb15d998612"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-16 10:26:28.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-16 10:26:28.000000000","tz":60},"subject":"Run handlers before configuring veth interfaces","message":"Run handlers before configuring veth interfaces\n\nHandlers from MichaelRigart.interfaces are not run until the end of the\nplay, which results in the veth role trying to create veth interfaces\nagainst a non-existing bridge.\n\nChange-Id: Ie1c9e1ada50774cbcaec528bd64d8df066c21d2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fbf000689559c434157d49b147da60725ba9562a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fbf000689559c434157d49b147da60725ba9562a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[]}
