)]}'
{"id":"openstack%2Fkayobe~868568","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I66b3a85adfe2928b936006780fb20b5fc7cc802f","project":"openstack/kayobe","branch":"stable/xena","topic":"fix-tox-stable/xena","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-12-29 10:00:00.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I66b3a85adfe2928b936006780fb20b5fc7cc802f","subject":"Fix tox.ini for tox4","status":"MERGED","created":"2022-12-28 17:43:22.000000000","updated":"2022-12-29 10:58:37.000000000","submitted":"2022-12-29 10:57:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868568-fix-tox-stable/xena","cherry_pick_of_change":867030,"cherry_pick_of_patch_set":6,"meta_rev_id":"8945f388800ef12770182e560ba814d29190b6eb","_number":868568,"virtual_id_number":868568,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-29 10:57:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-12-29 09:36:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-12-29 09:27:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-12-29 09:36:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-28 17:43:22.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-12-28 18:10:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-29 09:27:30.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"6c36efe10da62393d3d4bdab8a24b6fc43cada0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-28 17:43:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e11d0b1916ac103e4b87303f84266b6e9e9b2aae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 18:10:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fd36e955f704b698ef9bcece8bb0caa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84ea930230ec412e90a67f10d288f71f : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50b0c79b892347d5a9341bfdca92a331 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4375c9802ec049a9bc340572dcfe7a57 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d9df76f88b547aabfe857aaa704dcae : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad1be9d21b54409291e588d4f2b8dabb : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c16b4915f9684284bf1fb057d250b6c2 : SUCCESS in 9m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0392c14291424514976ca44e7b6a0711 : SUCCESS in 3m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0f00f65aefd64225b46c78d5d7f5e110 : SUCCESS in 5m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/005e57ae6cc5405da2ff24c24a933d69 : SUCCESS in 15m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2f05d1ea866945eeba4dbfbc0ff37b5a : SUCCESS in 19m 18s","accounts_in_message":[],"_revision_number":1},{"id":"a4785b40f1dd6886ce52138e00c0ae053cf4c320","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-12-29 09:27:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fed0a79a637a66601aa4b6e85216120e7a4b999a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-29 09:36:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a61a4cac3c245811ee1783049455e9ebe3a953db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 09:36:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"16e3093e10628a34b1fe33a57c6d1e8c4e080844","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 09:58:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a98738c2d67846399c856d7a4e937380\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95fb8043bb884695ba47fee7411aec16 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae3ed683dd9141b1bfb16f302901136f : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/340997812b9d4e9fa088598522df2e81 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd486b032bfa4e1eaf559a302f3f75be : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42f58212e65f45cd922463c78941cce5 : SUCCESS in 3m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/339213a4b01040098f543349cea0d9d4 : POST_FAILURE in 9m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c307d217bc5940ca83e59f20bf16521b : SUCCESS in 11m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3090360583e34fca91bc6e5c0bd415e4 : SUCCESS in 17m 07s","accounts_in_message":[],"_revision_number":1},{"id":"13c79807867c37a286b8dbb4cad5081d8531120a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-29 10:00:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9d69fb2b6f8930b591fe34a3e0718a637254cfd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 10:24:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6673019f4c31483096c62f7e9d17f30d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4605023f8d504cfcb8095d2d600054b1 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b49cc47219b04f73a80ab7888fcbec56 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/100896f923664ea6a3d883c64840cc6d : SUCCESS in 11m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/362c53fcccdf4d4b9850a8a0c25b8fe5 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a32253e207e4627abd3599ae4f5a19b : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c99768388e04149b4a5e448ee3bbbc3 : SUCCESS in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15dce7741e8c4f039faee9bdb08b43d3 : SUCCESS in 3m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7ead6a8723ad44048513d6d8be083f86 : SUCCESS in 6m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bca1912e227b45a6aac11914f4be31ba : SUCCESS in 14m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0d53bad8bbce4317ace976f1f1c3253a : SUCCESS in 17m 33s","accounts_in_message":[],"_revision_number":1},{"id":"320b63283ded84b7e33f05fd4690cd0e3215a987","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 10:24:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"69dbacd4fdf94f0c86248c7a5c59779d243c248e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 10:57:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bd31613ecc1493d8cd4ac19e7fc0e66\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e46f1239dc804b6db737aa75185eacd2 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c59fdcf932b8486f89d3f61a16dabc1b : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df7fde3c1889407e80a67e896a10b46f : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ddecc8723f14aef93430cb651a47e2f : SUCCESS in 9m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f40034d7e71d4d008c96b8c9bb7460d7 : SUCCESS in 3m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a04f531d93fc44fcb60729df73dc0782 : SUCCESS in 5m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d1f7b53c26734c9ca7e6972492c484a4 : SUCCESS in 11m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/255b509c0b1b40869afbce7ddc616880 : SUCCESS in 18m 56s","accounts_in_message":[],"_revision_number":1},{"id":"6d1bca1f1a94f05c72505dbbe596072193c08c08","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 10:57:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8945f388800ef12770182e560ba814d29190b6eb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 10:58:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f5de64d05ab4276bd949f19247becd1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3dce90864484f27b4a842b961114535 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c06f5f20b7b425380071444d70c6fae : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e9a420160afd5fd41fe753691f03f6bf2baee300","revisions":{"e9a420160afd5fd41fe753691f03f6bf2baee300":{"kind":"REWORK","_number":1,"created":"2022-12-28 17:43:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/68/868568/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/68/868568/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/68/868568/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/68/868568/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/68/868568/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/68/868568/1"}}},"commit":{"parents":[{"commit":"e6f224f92bd9048fe856337f8a686e5e44060927","subject":"Do not fail when admin openrc file does not exist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6f224f92bd9048fe856337f8a686e5e44060927"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-14 07:38:05.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-12-28 17:43:22.000000000","tz":0},"subject":"Fix tox.ini for tox4","message":"Fix tox.ini for tox4\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/866943/\n\nChange-Id: I66b3a85adfe2928b936006780fb20b5fc7cc802f\n(cherry picked from commit 5e1e28538451cd4b45bd0b4ed42239a7d622ecf4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9a420160afd5fd41fe753691f03f6bf2baee300"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9a420160afd5fd41fe753691f03f6bf2baee300"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
