)]}'
{"id":"openstack%2Fkayobe~868569","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I66b3a85adfe2928b936006780fb20b5fc7cc802f","project":"openstack/kayobe","branch":"stable/wallaby","topic":"fix-tox-stable/wallaby","hashtags":[],"change_id":"I66b3a85adfe2928b936006780fb20b5fc7cc802f","subject":"Fix tox.ini for tox4","status":"MERGED","created":"2022-12-28 17:43:33.000000000","updated":"2022-12-28 18:38:35.000000000","submitted":"2022-12-28 18:37:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868569-fix-tox-stable/wallaby","cherry_pick_of_change":867030,"cherry_pick_of_patch_set":6,"meta_rev_id":"5dac05eeb57c270abe36bc12e3255a0a5262c46f","_number":868569,"virtual_id_number":868569,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-28 18:37:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-12-28 18:21:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-12-28 18:21:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-28 17:43:33.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-12-28 18:17:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"456e679b8db2bc64d148d569094a157458111067","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-28 17:43:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a932afb5e02aa031a9371cca0f1d2798ea80d6a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 18:17:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caa8d362712e4102b73bdcd00c61d12e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0410a7a10ca40549edec344a36b2d7a : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5965af7c9844690887386d232e6e6bb : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5e52f20222e4c0bbed12889bd95255d : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/330ea3f1ee0a4283985e2d73beceb025 : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/568c3575ce524b07a45671a23af88025 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d21b4599e1e48f0bc675c0b351d4182 : SUCCESS in 8m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e9b0b61e6b945a7a12fe57748939aaf : SUCCESS in 3m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/da7457a6ac424934bb7f34a935f13fce : SUCCESS in 6m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7bce5b90994449aa8ca8fd8d67f77fac : SUCCESS in 10m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2c1eaf5a3a63424bb7661b5eb37bc125 : SUCCESS in 17m 11s","accounts_in_message":[],"_revision_number":1},{"id":"698aefdc78dceea795dfe29a4eeca153e3cfefbe","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-12-28 18:21:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3da09a29062616cfdc79a6bcca7a2c0e4deeb60b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 18:22:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"93962afe772bd530767bbc3c0fe62bd47d5aea06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 18:37:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b0bca0c3c03473dadc1d6478f14569b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89871f6eb4fc4486a9290a6c504fbfd4 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b63493cb25f4a56b93cff138233672c : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ec8758ebddc42949351b9e8a10a7419 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f53a7598e7d04772b5d28f588922b25b : SUCCESS in 7m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c768357127242889aeaf3afcbe9e5f6 : SUCCESS in 2m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/04a69a7306664fe398a51b7a6d4d4c0e : SUCCESS in 5m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/74422833674e44968fbaf23795f070f1 : SUCCESS in 9m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f04c9901c43a4f91a4aa115b4a4475aa : SUCCESS in 15m 07s","accounts_in_message":[],"_revision_number":1},{"id":"bf3ed128b3e0d651b865b27e31d3b46411e5d5d4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 18:37:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5dac05eeb57c270abe36bc12e3255a0a5262c46f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 18:38:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6416affd4c174db3a4d9f537360f491f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00198351428a417bbafe523dd9a82c7c : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e86be0243cb4c799c16d195e8546039 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f232ab555e9f3ad89a0fff5ace588c7f68d0b26c","revisions":{"f232ab555e9f3ad89a0fff5ace588c7f68d0b26c":{"kind":"REWORK","_number":1,"created":"2022-12-28 17:43:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/69/868569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/868569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/868569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/868569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/868569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/868569/1"}}},"commit":{"parents":[{"commit":"03419f8d4e3024438396f6ab148e40431f4bc3f7","subject":"Do not fail when admin openrc file does not exist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/03419f8d4e3024438396f6ab148e40431f4bc3f7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-14 07:38:05.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-12-28 17:43:33.000000000","tz":0},"subject":"Fix tox.ini for tox4","message":"Fix tox.ini for tox4\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/866943/\n\nChange-Id: I66b3a85adfe2928b936006780fb20b5fc7cc802f\n(cherry picked from commit 5e1e28538451cd4b45bd0b4ed42239a7d622ecf4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f232ab555e9f3ad89a0fff5ace588c7f68d0b26c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f232ab555e9f3ad89a0fff5ace588c7f68d0b26c"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
