)]}'
{"id":"openstack%2Fkayobe~869128","triplet_id":"openstack%2Fkayobe~stable%2Fzed~I666d7011bde0050ebc509b427c1d4f5a66b6231a","project":"openstack/kayobe","branch":"stable/zed","topic":"systemd-networkd-vlans-stable/zed","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-01-16 12:42:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I666d7011bde0050ebc509b427c1d4f5a66b6231a","subject":"Support configuring VLANs with systemd-networkd syntax","status":"MERGED","created":"2023-01-04 10:16:27.000000000","updated":"2023-01-16 12:43:17.000000000","submitted":"2023-01-16 12:42:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869128-systemd-networkd-vlans-stable/zed","cherry_pick_of_change":855538,"cherry_pick_of_patch_set":17,"meta_rev_id":"23b98075c0b72005207aac7f0f35f9ab5ccadada","_number":869128,"virtual_id_number":869128,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-16 12:42:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-01-16 11:37:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-01-04 11:50:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-01-16 11:37:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-04 10:16:27.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2023-01-04 11:43:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-16 11:37:36.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"679a65758bcfebad144a0fa5f95402bf5dbf6a5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-01-04 10:16:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2d5732e538571fac0c509b4e65c9efa22f182ce1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 11:43:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff31a4a37ffb4227b337b24156dae7cd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca2d8ccece3446bb9b9a7683302af2bc : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce888353def04606bcd62439bf3826c5 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eaff2e838f9d4bb6bc3131286d1cd9cd : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa3ac3e93942422394eaadda651fae10 : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ffb182a4bb694b469de785ce2920544d : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0bf9b4778984c36a85e2b4fd04960ef : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3444f83f3e30411ea7a73851d467488b : SUCCESS in 3m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/19cfb1e5f6be4738b9708a51406e5356 : SUCCESS in 3m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f66592da68804456934971d9cfb262fd : SUCCESS in 9m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b1f5370b96c44f19a94d711fa28e40ec : SUCCESS in 13m 35s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a3b20081059844729679d83691d81a7a : SUCCESS in 54m 39s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/44ca6886f23d4cbe84fa62e8b97edd05 : SUCCESS in 58m 01s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e59f2456c614c6c890f5976ede9e584 : SUCCESS in 53m 27s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3fd7f4974f6f4d2aa842418ed842538f : SUCCESS in 51m 12s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a5e68c84092846b49fa3a2688f9fb9e9 : SUCCESS in 15m 18s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/dcfbe5d8b6b74c49a6d556807ee7f47a : SUCCESS in 16m 48s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7220fee61174ede83050b7fcd500e9a : SUCCESS in 14m 11s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f7f82e023c14af09004337dfae43ee7 : SUCCESS in 1h 23m 58s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/368914181dc64416a386062dada775d2 : SUCCESS in 17m 01s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42c2885568404eb08a5e922aac959420 : SUCCESS in 13m 04s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c977d68932f64beeb12d09c8b982078c : FAILURE in 22m 35s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddcca7039f3d449994d7d81f69d3ecba : SUCCESS in 29m 04s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d9677c8d2cd5482da84a94ff6487c966 : SUCCESS in 12m 00s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f316c08047254d9092a8b772a5dc3be3 : SUCCESS in 10m 31s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f92d866d94ef4459826e505c0f27139b : SUCCESS in 15m 29s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e43293b88cad444e904ca4dfce0e5267 : SUCCESS in 12m 13s","accounts_in_message":[],"_revision_number":1},{"id":"cbc2dfb582187df6be9012614bc396260a4fe9d4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-01-04 11:50:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d7060841fe474d06315c29746f6919c9d73da99e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-16 11:37:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f84faf890d8a5231ceeb5edabba196c288c423ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 11:38:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a2b6c99afcae33fff3690de90b2e7867abb06157","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 12:42:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e89cfc300da846599cd2baddb8a3ef09\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7be28f452724420299bb29723e1b7e06 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3a7d02618234d4492b4719f39c5e447 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0bd70c0ca70747ceb762e991e05db27f : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e71bb45cc95047759694ec14e4bb226d : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55f9ae3051974c4aae01d64238a3b417 : SUCCESS in 3m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7e6babe0be614ccb98cd54a6f7d822e2 : SUCCESS in 3m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/414b9d514a634042a2b309e1cd028b2a : SUCCESS in 10m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c6d47ed22a0a4354944dc6591f3c3580 : SUCCESS in 14m 17s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5b20073165e54d23bbd92a6594b53812 : SUCCESS in 54m 15s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5baea98fb9cd4b19b724cadd34b374bd : SUCCESS in 52m 25s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8cdf7984fbfe467884947316199c9c34 : SUCCESS in 56m 46s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d10b99a39f8946cf955c27f7d9e40bda : SUCCESS in 13m 40s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7075c10eb77b4ef28ac2ec58e4270b59 : SUCCESS in 17m 00s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05733b7046554cec9589c81056c7374f : SUCCESS in 57m 36s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f836b8ea70cd4bb7bec60ad4f279b719 : SUCCESS in 14m 51s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4f3f14a23ce451f81fcc49981a83f37 : SUCCESS in 11m 47s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/627dd5c50ad740cd9e2e4a278fb38883 : SUCCESS in 23m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6557049bca98427fb03ff2847e4a7505 : SUCCESS in 14m 31s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1bf52fb27b04506bb2367a8e310abb8 : SUCCESS in 12m 49s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a9926e9519134df581c1837d4c745ce7 : SUCCESS in 15m 12s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1f91eb3454240ac9a299da53692c59d : SUCCESS in 15m 22s","accounts_in_message":[],"_revision_number":1},{"id":"eeeed4282f4898232c5d74f3ae54fc54fb05c77c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 12:42:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"23b98075c0b72005207aac7f0f35f9ab5ccadada","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 12:43:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a65214ed5fdb4749ae25894b73a84977\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01e8500a103546b0b1dd7714428e8684 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6141585539134d368568f61cab3a59dc : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"248bacc49044caf707b60c009d6386ff0090deba","revisions":{"248bacc49044caf707b60c009d6386ff0090deba":{"kind":"REWORK","_number":1,"created":"2023-01-04 10:16:27.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/869128/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/28/869128/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/869128/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/869128/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/869128/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/28/869128/1"}}},"commit":{"parents":[{"commit":"8bb11a6503a5a9352d622e64c164e589b09f0967","subject":"Add note about Zed upgrade using Rocky Linux 9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8bb11a6503a5a9352d622e64c164e589b09f0967"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-09-09 15:57:11.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-01-04 10:16:27.000000000","tz":0},"subject":"Support configuring VLANs with systemd-networkd syntax","message":"Support configuring VLANs with systemd-networkd syntax\n\nThis allows operators to configure arbitrarily named VLAN interfaces\nusing systemd-networkd.\n\nStory: 2010266\nTask: 46178\n\nChange-Id: I666d7011bde0050ebc509b427c1d4f5a66b6231a\n(cherry picked from commit 6d7b8812ae82b481519818a00293bb1aec32d058)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/248bacc49044caf707b60c009d6386ff0090deba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/248bacc49044caf707b60c009d6386ff0090deba"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
