)]}'
{"id":"openstack%2Fkayobe~869444","triplet_id":"openstack%2Fkayobe~master~I23c409900db2c37105e7b218bbaedcfacb235787","project":"openstack/kayobe","branch":"master","topic":"sync-kolla-tag","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-01-09 11:21:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I23c409900db2c37105e7b218bbaedcfacb235787","subject":"Sync kolla_tag with kolla-ansible","status":"MERGED","created":"2023-01-06 10:19:49.000000000","updated":"2023-01-09 11:22:13.000000000","submitted":"2023-01-09 11:21:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869444-sync-kolla-tag","meta_rev_id":"dd73a4aeb964da4c06f61ede5111ca393794caf0","_number":869444,"virtual_id_number":869444,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-09 11:21:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2023-01-09 09:45:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2023-01-09 10:07:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2023-01-09 10:07:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-06 10:26:14.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-01-06 12:05:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-09 10:07:24.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"6f5af238f0c9c7ba5ff2768b896c68a42e2e529c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-06 10:19:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9fc8afa75905fb2d9687f2e36380012a21c2132f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-01-06 10:26:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"95f367fc95e0c130a53d5b314760b66a33cbf5b5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-06 10:45:59.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"aebdb6d34ee116e4f3538c9eb5590dd804e48150","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-06 10:53:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"14bd43c76c9441e584fdae687476dffbf70eea41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 12:05:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b2f5105beb44e0783a85b22c5909b70\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11d8736d7aee4aa08f88174685bcfc91 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c787893d521e455caee793995606fe4d : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/005681d422b04d00b08f30d0550a0409 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa9d1d5700e24d9f9f3142dea648f805 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e444d96f44a406b886bce67ee419ef2 : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74374bbc5813452dbd9565c8e8225f15 : SUCCESS in 3m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b62eed19fd1b414194368788a73fec47 : SUCCESS in 4m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6bb9378dbaa34f35b5aaecfef0a6274a : SUCCESS in 9m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cbbc4eac157f4ca3b3b2b8d2cf2eb7dd : SUCCESS in 13m 26s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/df734bc56172447b82456172fb774f1e : SUCCESS in 53m 40s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/967c8f9b2fd2409bb9689673eef4e6c7 : SUCCESS in 56m 24s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5235b9a66e2f4aff858fb27fe2ed47ad : SUCCESS in 50m 44s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8e7e04444f2f4a8c80ff19e23da38ee2 : SUCCESS in 54m 57s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5c55784417ca4ab8b61cd3f0f6fb0fa4 : SUCCESS in 17m 28s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/571966de184e48a0bbbd666ff1da0ed9 : SUCCESS in 16m 11s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7272577aaa7400d85a8d94acc6778af : SUCCESS in 12m 01s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8975b4502404faca53285719fd305a1 : SUCCESS in 1h 07m 02s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/bb533e1498a645acb81537b7583e92ad : SUCCESS in 17m 48s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d04d055212ba4a0ca7182f7c559b85c4 : SUCCESS in 14m 13s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/5dd5eaddfbcf4fddbf25747619667d7d : FAILURE in 22m 40s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8451fd4b8ed4d0f84293b222f2c7415 : SUCCESS in 31m 59s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c1765bab1f1b4507a50780737512f406 : SUCCESS in 12m 58s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68fa84479c90441f94e4b85ac29fea5c : SUCCESS in 10m 55s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6a0a7154437b406bade96945acd905c3 : SUCCESS in 16m 57s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49d0ddb16e8842a3bc840c5ab6946fdc : SUCCESS in 11m 15s","accounts_in_message":[],"_revision_number":2},{"id":"aba0ca1e7de5444a6df90b0495fe11da7b3c7ec3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-01-09 09:45:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cce909db75d4d77a962745c0436201622b1db472","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-09 10:07:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"585694efe5edfe88dadd54f2177bd36a5016a80c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 10:07:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1783253d3f5de14ec9f964706538da309b22390d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 11:21:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cacde45b13474ccdb3081f8e5a4e4e69\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b01c4dbeecd442abf6228599556a83e : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f280e06d26b403d95d83a3f15de9a51 : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86acba7fcd9e499083d2ebd1f6c1713d : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60f292061e4e40fea16f19850dd0e5b5 : SUCCESS in 8m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0971c17a2a1940b79a54983734e3e339 : SUCCESS in 3m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a469b773b50e41e39666e24d6d4e0edf : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8fade2eeea2c4a818a5028bb6cbaf60d : SUCCESS in 8m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/161e8272974c49a987fd2170647a62e1 : SUCCESS in 14m 30s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/629abfe6af974a64abf1322e1e34fec8 : SUCCESS in 1h 06m 55s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8bdad81174240e9a15d154c51dfe59b : SUCCESS in 43m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/133d6f5a5d344df3a70ef6ed5afe6d39 : SUCCESS in 1h 08m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/07dfeff23dc1473bad6d6e189bb3d4a9 : SUCCESS in 16m 40s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef1a0a202a15441c8307f41b8564bbf0 : SUCCESS in 11m 28s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/058e523087464fe8905dab654ffcf08e : SUCCESS in 57m 30s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7bb2d7fedd1c42e1875ba4209591db46 : SUCCESS in 17m 16s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b34aa264159645aeb3140373f66b5bdf : SUCCESS in 12m 16s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c4a987e7b2e4697a2254ffbd603ba5e : SUCCESS in 25m 08s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1bb7d36b502e4af3a64715b2cef5936e : SUCCESS in 12m 50s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc3d155eac7d47dcbec5eadf6483fb18 : SUCCESS in 9m 26s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/81ed578988524d6bb6074e0af7eb2009 : SUCCESS in 16m 37s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1217e75e6d374983b89d94d3be5d5316 : SUCCESS in 11m 38s","accounts_in_message":[],"_revision_number":2},{"id":"ae9d49d797592d090b3debf3b0473798a5ec156e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 11:21:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"dd73a4aeb964da4c06f61ede5111ca393794caf0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 11:22:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75a2bbd22b944699b3c9c3d7c65b7657\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fcda454f05b4e4dbce118f75461532b : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ac2c5807fe148a8a88c494f12db711d : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8640fbd5c0389848290fca9d31c8fbd964e1d63d","revisions":{"117fe65773663027d1d8a3bd5ea88ca0aa9ce826":{"kind":"REWORK","_number":1,"created":"2023-01-06 10:19:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/44/869444/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/869444/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/869444/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/869444/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/869444/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/869444/1"}}},"commit":{"parents":[{"commit":"ee25470b01bb205d2822133305cc1965810bcd9a","subject":"docs: add missing word","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ee25470b01bb205d2822133305cc1965810bcd9a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-01-06 10:13:25.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-01-06 10:13:25.000000000","tz":60},"subject":"Sync kolla_tag with kolla-ansible","message":"Sync kolla_tag with kolla-ansible\n\nWithout this change, Kayobe was tagging container images with `zed` when\nkolla-ansible was expecting something like `zed-ubuntu-jammy`.\n\nChange-Id: I23c409900db2c37105e7b218bbaedcfacb235787\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/117fe65773663027d1d8a3bd5ea88ca0aa9ce826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/117fe65773663027d1d8a3bd5ea88ca0aa9ce826"}]},"branch":"refs/heads/master"},"8640fbd5c0389848290fca9d31c8fbd964e1d63d":{"kind":"REWORK","_number":2,"created":"2023-01-06 10:53:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/44/869444/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/869444/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/869444/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/869444/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/869444/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/869444/2"}}},"commit":{"parents":[{"commit":"ee25470b01bb205d2822133305cc1965810bcd9a","subject":"docs: add missing word","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ee25470b01bb205d2822133305cc1965810bcd9a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-01-06 10:53:29.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-01-06 10:53:29.000000000","tz":60},"subject":"Sync kolla_tag with kolla-ansible","message":"Sync kolla_tag with kolla-ansible\n\nWithout this change, Kayobe was tagging container images with `zed` when\nkolla-ansible was expecting something like `zed-ubuntu-jammy`.\n\nChange-Id: I23c409900db2c37105e7b218bbaedcfacb235787\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8640fbd5c0389848290fca9d31c8fbd964e1d63d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8640fbd5c0389848290fca9d31c8fbd964e1d63d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
