)]}'
{"id":"openstack%2Fkayobe~869632","triplet_id":"openstack%2Fkayobe~master~I2d87017c27139b5e0154edef006ad61d665ede2a","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-01-12 13:44:07.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-01-13 18:09:15.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-01-13 15:23:14.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-01-10 09:56:28.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I2d87017c27139b5e0154edef006ad61d665ede2a","subject":"Revert \"Use OpenStack Zed release\"","status":"MERGED","created":"2023-01-10 09:50:04.000000000","updated":"2023-01-13 18:11:18.000000000","submitted":"2023-01-13 18:09:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"revert_of":867617,"submission_id":"869632","meta_rev_id":"4a95233cb5ef6e5ad81ff6d500281c20bfbf572f","_number":869632,"virtual_id_number":869632,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-13 18:09:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":1,"date":"2023-01-12 10:46:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":2,"date":"2023-01-12 13:44:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-13 13:52:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-13 13:52:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-10 09:50:04.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-01-10 09:50:04.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-01-10 09:50:04.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-01-10 09:50:04.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-10 09:57:42.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"}],"messages":[{"id":"ddd35561b6d80a9888c4d159df87ad21ed3fa992","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-10 09:50:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"100104642cc362c12b48b3b329620b19eb4adf31","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-10 09:56:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b368e5309791c713f76b3ae5745a32e3237031d9","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-01-10 09:57:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"20ca5648dabc1d908e41179fe7ac25615d9b9d55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 10:57:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/50fbb705e18e47409c286ae4cc60f605\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/349fa7773bc7484bb401daed78dcc1be : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eba95e29d84c48bc96a834cc1412f36d : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17fdbd7c74524d53ba2ec5b802df6d9b : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a4cfb934cafe4566a97c20068dbf9567 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d92fce26b56b4657b97167699baa4b0e : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75ba21b4443246d68c8d48d3581f439c : SUCCESS in 3m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/987a3474b1624fcd8348f5afe69d5d8d : SUCCESS in 4m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b12820f6f9ac4e4cbcd583aa38269c07 : SUCCESS in 13m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/654f822eb0bb441fb7a1de814457edc1 : SUCCESS in 12m 06s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/21cb8693fe61409ab274ecfc9eab62fc : SUCCESS in 48m 32s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/bd49490b379b4b49bb67d928f3258824 : SUCCESS in 58m 43s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3f3ebcc7440475c8eb525401670b371 : SUCCESS in 54m 34s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2cefe74e31874882968653d8e3129971 : SUCCESS in 53m 23s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/bf0ec5c6012a4456901115797528124f : SUCCESS in 18m 58s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b52c99a41ddf4976a91b5705e4a6c51e : SUCCESS in 14m 03s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e4ccabc55924b5286eaf741c8620482 : SUCCESS in 16m 04s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47550a8bd3734fb5b3bdfafe8e3f686c : SUCCESS in 1h 04m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ba2bbd67a7794af989813b6cdd3acabe : FAILURE in 15m 35s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f316d8f191047a2ab361ab77405e4a5 : SUCCESS in 13m 37s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/224ba3ea5c5a4b0eb8bb7b6f2ae013e0 : FAILURE in 15m 35s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d3f73a2b57b401abdb8898c2ad1b303 : SUCCESS in 24m 44s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/01721a00ebd7481c9b1365c3d3e0f814 : SUCCESS in 14m 18s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86fc717cc3014da2a92de29f31fdf787 : SUCCESS in 9m 59s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/58b8c847bcea48f2aa7433747369b4f7 : SUCCESS in 13m 48s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e12b32a8f4c469d94e6a389ca286a11 : SUCCESS in 12m 23s","accounts_in_message":[],"_revision_number":1},{"id":"c13a8bba6f4a10597757ccde16d8e11757f066e4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-10 11:42:55.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81bea5a75be6e67bba937df63351d0d06361f816","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-10 13:11:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba909458968d2a0778e7391feb603498bca797cb","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-12 10:13:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"717b638abae2534ba5dde11503c81d6ba79da71d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-12 10:46:49.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"915f72552a6abfacb113a31fa589c424abbffc3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 12:06:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/497d2271201a4aeeb9f4f1b673a791a6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd67d02b71de4865860e14124d9da2d5 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a65643bedbfd486b84562908588fc8f1 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbafe7d439424fe49e79009f75ad1dee : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da45506f26be478ba16535dd790ffd83 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6eab0ffc32546f7a30191805188ec29 : SUCCESS in 8m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b937fda6a9a4dfaab803782f7ddd112 : SUCCESS in 3m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8aa0a8a9bb6a41da98b1d951e5305224 : SUCCESS in 4m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a703a1f0571142f5bd93e6b2a84caf6b : SUCCESS in 11m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8b3c808ad1c046139806ff35a9f06091 : SUCCESS in 15m 16s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/92c09a8eba5a495aaba77e75be8374f9 : SUCCESS in 56m 12s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c49d04a9e02c46fa9555e03f8e503a03 : SUCCESS in 1h 03m 26s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/594ab6b724df46eb9e4501f7311853f2 : SUCCESS in 41m 57s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7763dc7704ee40e19f2e921704d7120f : FAILURE in 46m 55s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/96554350e3ad43388320959438931504 : SUCCESS in 14m 09s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/4d07269020a549dd910612a289eb9530 : SUCCESS in 17m 44s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eda18ee4d10e462ab45624c4e737adcc : SUCCESS in 15m 57s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6fdd43f46e64ba9b1f07bc635304399 : SUCCESS in 1h 18m 35s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7aab858b0b5e4aaa9e3a63fb5b992457 : SUCCESS in 17m 33s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b9185add5104898b6b09654dae690ac : SUCCESS in 13m 16s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/3a35a7daa16048aca69e4e9d29331843 : FAILURE in 27m 28s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d0c84492ded4e478b0db85a9402b639 : SUCCESS in 28m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6ca01de95d6a4d4e80bb0e8ec32420af : SUCCESS in 13m 34s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9d67848699740c88c8aae30e35311c0 : SUCCESS in 14m 03s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3a9e7bd529e248c8b40184ca9fe34c82 : SUCCESS in 15m 57s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88f67d6d068e41d1b335c5f351980b5b : SUCCESS in 16m 05s","accounts_in_message":[],"_revision_number":2},{"id":"2e059b455dd0df8cb9f5520011e11d6283a1b06a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-12 12:08:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"57c921738bf76993f6e7306b94994abdf45dc985","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-12 13:44:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"037e1b69e29421fb1f9e572f3f9beea92dca71a3","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-12 13:46:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6888a794c9a8c8b43dafb4446fede5119daa0f41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 15:18:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65b03f9872d14cebabe12db8d6002b40\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/347a2ea24ef94af880a036e63cd41771 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3abc042414f849c59d66718bc60b6797 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d39694c520934ef0b336f085276f250a : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f4fde6ce51c459dbbb9f25799330673 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2af6393d034f4cb1bd7335bb1e7f2754 : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7268a2344bb45d999531951f7b70005 : SUCCESS in 3m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/107195f9aac24a91bf9379b71d83919f : SUCCESS in 4m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/331ee488ce494720abe6e3d753558d51 : SUCCESS in 12m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/71ee885a83d34f9fa1e7b9aa72ada57d : SUCCESS in 15m 52s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/27a13bcf4e424c53bb98b055356ee78c : SUCCESS in 57m 46s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/5e9fed8357124be6ba873c6837d76ab3 : POST_FAILURE in 3h 03m 29s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b5b3d2dc52074661af0049810a359de2 : SUCCESS in 58m 24s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/86f224235d064b1aaa177b78c768b3a2 : SUCCESS in 57m 30s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8765fad7acd842979acd06733089eb83 : SUCCESS in 17m 32s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d79acf3f3b9647aa8a4335beabcb81c9 : SUCCESS in 18m 29s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed53dc588d6f4437baea8e2748d7f9a9 : SUCCESS in 14m 55s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6b66464ea8545d7842df8e636c3ce4d : SUCCESS in 1h 17m 09s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a762a8143e334c0a82d48d4f448dc6ed : SUCCESS in 14m 20s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26066223c846446cb44ef64795623322 : SUCCESS in 15m 54s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/0596da46a6174706a649e1b02bce4a92 : FAILURE in 24m 22s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abbffe44da2d4ec78549ec8b53d6b7fa : SUCCESS in 31m 42s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/307629e0dd0c469d91f3621e5cf2ff10 : SUCCESS in 11m 13s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eba2bd06d79046eb8a54189ba6ecabfe : SUCCESS in 10m 55s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/67ae66f2dd21436ead3aab5b6711d376 : SUCCESS in 16m 18s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50a85b28676c4877befa24724312de66 : SUCCESS in 11m 05s","accounts_in_message":[],"_revision_number":2},{"id":"b32ad258116c2168aab20af8ec8fc92f0eb2f297","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-13 09:10:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eb20ca7d09364adf0a5c3c81f77632357afb0fe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 10:43:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8fcaf49af7845a782edcf52049ed6b0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/203cfcde4ca14f2780e3418aaca7c140 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3d1f43ce36141159d46e7528fabdab1 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51fedbb862f14ef5a7804d737709f6df : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8d609be8870499590da6a3b2462e6c3 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/196ceb49745241d1afddd78663fd5bda : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a78fb9a9620432381d92bd56979bb51 : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/028245d78057443391b5ce449ee21d5f : SUCCESS in 3m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fe04898445794551a0eff699a0302f27 : SUCCESS in 11m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/edfad45659dd448b98297851062ea64d : SUCCESS in 14m 26s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/368b772f4a454ad1b55a4bbd95526cef : SUCCESS in 57m 33s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/d059b07e6e1a438d85b2cbf36f29b082 : FAILURE in 56m 24s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8f7f8603b834af6b13e820b2b9a2559 : SUCCESS in 1h 01m 39s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/40b33294d1d24a4c9b88cc9be75d4da9 : SUCCESS in 57m 53s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/411c1a36c33d4ecaa816744a48575252 : SUCCESS in 17m 28s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e2db6ee9e9d74f3f8706f34cfff0f2f6 : SUCCESS in 14m 37s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f204ef03cc94c2ea10b31c976d35564 : SUCCESS in 14m 49s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d019b7da4724447b2af8b42991712ec : SUCCESS in 1h 13m 24s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3fa9ed1e70634d1db755945c7d227f92 : SUCCESS in 19m 07s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45be006781f14da182e54606e4dd0359 : SUCCESS in 15m 34s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ee1846b8445b4a41bcc0656c777d5bc9 : FAILURE in 32m 54s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bcb92d011f94a67bc781fd23196d625 : FAILURE in 1h 27m 34s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/300731f882f34140aa37c095f2e2470c : SUCCESS in 12m 32s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc98e28eb0a04bfe8c904b3212c1cf60 : SUCCESS in 11m 17s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/91791ad451da43b19b0933e152dcd6f8 : SUCCESS in 18m 04s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c9925adefa14dafb8f3028bbd627c3e : SUCCESS in 16m 19s","accounts_in_message":[],"_revision_number":2},{"id":"821cd40db29388219dae03b3ea72d6999bc694dd","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-13 12:38:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ed2b28694c69f977e747fa5056c07d0f36a20452","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 13:42:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffd703e7bd154c32825a086c3632e285\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb208ee63fe74d48972f9a82a62e5e70 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cecba94f74b40c2b424121d9b9447c6 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52c6607445ef4e39abdf567be4621854 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d1094a46c4c4fb99e9b464a0c9bd527 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4c63ed18ba2489b97838d9f0cc7eda9 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69c6d4304e3641608f59b5d0ddef7e3b : SUCCESS in 4m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e3bc6cfbaf564678a98663c2059087f0 : SUCCESS in 3m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/61924ecfa45c4d28bfd926b6b7a25497 : SUCCESS in 10m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f6ca91bc0b634a61b387e17b02fd9886 : SUCCESS in 12m 40s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0904fe0b78514e66bf4667c691e642a8 : SUCCESS in 52m 52s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/ae84de33d1614974b2357d81ef668246 : SUCCESS in 51m 01s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07a959b5e4444e89bbf015c065afe0e3 : SUCCESS in 42m 52s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2d7496618ac94c92b22172f29a56d4ba : SUCCESS in 54m 44s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3253bf13fd314d6b8638769fa9b4fe78 : SUCCESS in 16m 44s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/1da72a61aac9417dad726d62cc59ffd3 : SUCCESS in 17m 28s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18d19d195d434a21bd941d8a38b7c3ef : SUCCESS in 12m 54s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00ad042ee74c4aae814a22da760fc5bf : SUCCESS in 1h 01m 29s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/69ab78099a8444979eb6973c3337e952 : SUCCESS in 15m 21s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37bc1eb6018f4676906d31f2435850bb : SUCCESS in 12m 21s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/62492f85e7da4a21aaeacfc0b29ad7eb : FAILURE in 24m 31s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2135b2d954e34aa795f8c3d6e0cee8f7 : SUCCESS in 35m 24s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a5352b04bd5c41a3bc5016604566b1eb : SUCCESS in 11m 43s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ad3659641824fc0a4b647f7703508e9 : SUCCESS in 8m 45s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/01a86dc0de14497caaf7cbf8e1bd48cf : SUCCESS in 14m 47s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36be36e7eb624eb798056c59576aa2d1 : SUCCESS in 10m 32s","accounts_in_message":[],"_revision_number":2},{"id":"a15ce9d63121c778bbc02db6f2fdb7e2229272ec","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-13 13:52:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1bc752d7bb640b6c3ef5b99b1e64d6d3e5b96277","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 13:52:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e5ebecd4f177f9ccd09b2c1383e2f5d9f2a800b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 15:20:23.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5000c3f100549d6927779e56c2a435b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c5e60cdc0cc41af86c66e8f01b12abe : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c6984051aaf4cbb889be811ef7ea51d : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22b6b10016b341fdaf16201091c4afe7 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/726e7e8b9c0948ddb14f41925011ed63 : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/448785daaf0b4417a1bb1fe1a609a1f0 : SUCCESS in 3m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3b484c639d3e40e1902e68f8efbb610f : SUCCESS in 4m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e2ff1bcc9d4144ec8dd43f21c949f936 : SUCCESS in 12m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5c4336e1ee1246e8b4c95283e80a64a5 : SUCCESS in 16m 07s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c9bc00dfa2a547e5970551a853d1460e : FAILURE in 53m 32s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce920a253c4044279391863aab253378 : SUCCESS in 51m 42s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9ba9678a5c634932ae20cef8bc47adb6 : SUCCESS in 43m 45s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/424aafeea4ea4abcaa713ea8305092bb : SUCCESS in 13m 40s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a87088f36fd4b82b64437c721ab4160 : SUCCESS in 15m 28s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9845de474a9c437dbfee67e12b54daf0 : SUCCESS in 1h 13m 43s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9fab863638124ecba59926e098c5b2d0 : SUCCESS in 18m 43s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4aba91f7f33b471eb642ab69fdde7ef9 : SUCCESS in 16m 21s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d6a8ad9ff34404791937e4fecf78737 : FAILURE in 21m 45s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e61e09e671e24a80b5184176858b5338 : SUCCESS in 14m 27s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd46c4515b8f4a23a725336bd539d44c : SUCCESS in 12m 39s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3a5e04bf2fa0408098ea3865b05f026f : SUCCESS in 16m 20s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48f19a08699d4ac5b8f935626d064bd4 : SUCCESS in 10m 31s","accounts_in_message":[],"_revision_number":2},{"id":"88b8a8da22bfb3c866153fb524d91d3aae4fa971","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-13 15:23:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"87c574e095d7aadd37176adfa48a9dd3e82444ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 16:56:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/242a431056b7404c9e4cb93ba3d6d044\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0337100d64bd4037a749d5495b249c4f : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e31772cc7b5d495d88f8bcf789fe4913 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50df8f21f0f344989a5a0b7be27f987d : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69f98ec1aaef4a93932a2647d6739cd6 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb473f905cd24a788a364c7261678508 : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f644343b4f640958151ef405de6ee0a : SUCCESS in 3m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e184fa527a42472ea41c19b3f3730635 : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7d3e91ad87e04554ab6414fefbb604d4 : SUCCESS in 9m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d253394f91ff484581df95b5a697ce1b : SUCCESS in 14m 32s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d25650c8d44741afb3eeb3ef87493a7d : SUCCESS in 52m 04s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f32c03035474419da5ea8827839cc96c : SUCCESS in 1h 01m 35s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df970a36142e49ccb40fd8b62de050e0 : SUCCESS in 48m 29s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3770d262b8d04013aa15f00c53379684 : SUCCESS in 1h 03m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ebc628447cf944c0b219855b85b3484d : SUCCESS in 16m 50s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/c6793fed417d4b169b54e98c3ef0232e : SUCCESS in 20m 38s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/111169c464a540d7bbc62fb34c2ece9e : SUCCESS in 14m 38s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91de251a44e249cd9ddc6402330ab37c : SUCCESS in 1h 07m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/28734debbb4f49e8b42f7dfa3c67e4e4 : SUCCESS in 16m 54s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8497374d66c43db8e14c6f7601d3a2b : SUCCESS in 16m 35s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/85de6484181f49ca8a7da83062ed0931 : FAILURE in 24m 32s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cb64b36f0124c799a0ff213bdfe279c : SUCCESS in 57m 38s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/520cb2e07ab647aeae5ccc964ad8ad45 : SUCCESS in 12m 44s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6208d24c7b1a448c995a4e31763efaf1 : SUCCESS in 11m 13s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e18fd72d5ec04427b6fcff07c3d17f20 : SUCCESS in 14m 21s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fca0663253fc447c8aeff8141e9ece69 : SUCCESS in 13m 58s","accounts_in_message":[],"_revision_number":2},{"id":"e8f1bb5e1f5d70f0523d79b6332ae242eda58c0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 16:56:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"63b16b03cf87609dab75f079500c65b8d487a625","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 18:09:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/990e2141d28046fabc15d69ff5d24473\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42f953c62f864421ab424c03c57b6b0c : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc79a018d295418bb88173f240504a27 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2fd978f29c44057b340bd9f2f188253 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da391a0849b84b5a9f784187d49a8676 : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a30fb70f35b74120a8224602f0c680c5 : SUCCESS in 3m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3464c6e4715a4139a8eee89d2c4446e6 : SUCCESS in 3m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/53990c4d60274a469765dba14a2a31a6 : SUCCESS in 12m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/761b0a48dc864e5cb04233ad1aec476e : SUCCESS in 24m 56s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c468225a36124a8089f6e59395609bb1 : SUCCESS in 56m 32s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eba071445df04bdb852ea1c252a3e3ec : SUCCESS in 53m 50s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3c13e8b59aa9400a95907b070ed3e1cf : SUCCESS in 53m 28s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a0cf74ad687f4f44b9ff8d3589d9146c : SUCCESS in 18m 50s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05c3fa8c4e524d919bb43730a8cbc847 : SUCCESS in 14m 41s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79d027090b4243359ad04a77819f6fe7 : SUCCESS in 1h 10m 59s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/91a3d00fd5d2444f8e5cf5205fc3aec5 : SUCCESS in 15m 20s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/539a80403c8c4b3a9a90a32d5409a3eb : SUCCESS in 15m 33s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7312b505a4934d0698c4c168bc84d13f : SUCCESS in 33m 10s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2cd94c7f61504d31809b9412afec449c : SUCCESS in 12m 44s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59a5c93b25524c7fa9102aac3b7dfce6 : SUCCESS in 12m 36s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/810026fdb6d44b37ac3c1ddf5ce9c2aa : SUCCESS in 14m 28s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e222e49662324cadb271fda5bcdbfd8a : SUCCESS in 13m 47s","accounts_in_message":[],"_revision_number":2},{"id":"c9751aae902038c66e998cb9e247211e11a693ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 18:09:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4a95233cb5ef6e5ad81ff6d500281c20bfbf572f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 18:11:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67375e4bcc8240d7b909b3e739f6961c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba28627e652f418ab42e3d7b5e13d472 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0faf268eb50f4079a49cb2cc2d200bfc : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0c11102afea6d85001945071f27b9f8af7167512","revisions":{"030770376ee59776e0ffc481412c69cc5e93973d":{"kind":"REWORK","_number":1,"created":"2023-01-10 09:50:04.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/32/869632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/869632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/869632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/869632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/869632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/869632/1"}}},"commit":{"parents":[{"commit":"65c59b9f5939f19f5fb914dbcb0fc64cde99771b","subject":"Use OpenStack Zed release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65c59b9f5939f19f5fb914dbcb0fc64cde99771b"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-10 09:50:04.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-10 09:50:04.000000000","tz":0},"subject":"Revert \"Use OpenStack Zed release\"","message":"Revert \"Use OpenStack Zed release\"\n\nThis reverts commit 65c59b9f5939f19f5fb914dbcb0fc64cde99771b.\n\nReason for revert: Zed has been branched and released\n\nChange-Id: I2d87017c27139b5e0154edef006ad61d665ede2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/030770376ee59776e0ffc481412c69cc5e93973d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/030770376ee59776e0ffc481412c69cc5e93973d"}]},"branch":"refs/heads/master"},"0c11102afea6d85001945071f27b9f8af7167512":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-01-12 10:46:49.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/32/869632/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/869632/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/869632/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/869632/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/869632/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/869632/2"}}},"commit":{"parents":[{"commit":"ebd5799fd4b77a138cf7790a8094eb859e8898f3","subject":"Merge \"Replace references to Docker Hub by Quay.io\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ebd5799fd4b77a138cf7790a8094eb859e8898f3"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-10 09:50:04.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-12 10:46:49.000000000","tz":0},"subject":"Revert \"Use OpenStack Zed release\"","message":"Revert \"Use OpenStack Zed release\"\n\nThis reverts commit 65c59b9f5939f19f5fb914dbcb0fc64cde99771b.\n\nReason for revert: Zed has been branched and released\n\nChange-Id: I2d87017c27139b5e0154edef006ad61d665ede2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c11102afea6d85001945071f27b9f8af7167512"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c11102afea6d85001945071f27b9f8af7167512"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
