)]}'
{"id":"openstack%2Fkayobe~869672","triplet_id":"openstack%2Fkayobe~master~Ic0608bc6033025cb47655d601ffaf3744637832f","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-01-16 17:51:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic0608bc6033025cb47655d601ffaf3744637832f","subject":"set previous_release to zed","status":"MERGED","created":"2023-01-10 10:43:45.000000000","updated":"2023-01-16 17:53:44.000000000","submitted":"2023-01-16 17:51:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869672","meta_rev_id":"a6f1cfd8469ca640f2eac89029b36ebe668825e2","_number":869672,"virtual_id_number":869672,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-16 17:51:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-16 08:46:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-16 08:46:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-10 11:47:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-12 13:44:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-01-16 08:46:15.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"31abb21d21ecf9df4176a7738d79351ad1c70671","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-10 10:43:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8909996caeebd1ffb0928b226d17d8387fd5bb9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 11:47:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1aa893d3d1224a63b97d114e63f6890f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/379845b0b84242fb82f9f509730e410f : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18a362633cf64183bbc40655df3b704b : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6fe2bb477b64e3f95da45e1625862a8 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6728c7287518468e9c32f352252289e4 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5844552b08484fff8c633a2784899805 : SUCCESS in 6m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0002c0bef81043a3a8a9e8e452c22890 : SUCCESS in 56m 17s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/a7a89b44d83b44f8a44d340acd412525 : SUCCESS in 56m 10s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa7c51e73e5148839cb4167b7f5f7da0 : SUCCESS in 50m 16s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/75461473a4fc4b4f91b8430cdb79ddcc : SUCCESS in 59m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1934c7c1779e4c4b9d5093637f53898a : SUCCESS in 17m 01s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/50ffe878cfce4759b88d1e9c044a0bfa : SUCCESS in 16m 31s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ed0d5e783dd4468bf5721fb86961caa : SUCCESS in 13m 26s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/796afaa98f6f47a1afa4c79340dafe73 : FAILURE in 39m 35s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/41c2b49bc880480bb05a988061d5ce49 : SUCCESS in 16m 53s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f0f3f3835e0497cac546d74ee69a119 : SUCCESS in 16m 50s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/bc2740f8a27545129f7afba9ee4516e3 : FAILURE in 15m 31s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/caea051d88184b46b69962a812945242 : SUCCESS in 33m 18s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c275d4ce10ed48cda12b88344077134b : SUCCESS in 12m 48s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e57292c1f2544169b41b5ab57db8852 : SUCCESS in 11m 55s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fb93fe63be5240e5a7195232d8969309 : SUCCESS in 16m 35s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ead58f349da44153b14659a98e51d4f4 : SUCCESS in 14m 44s","accounts_in_message":[],"_revision_number":1},{"id":"971a52ad6ea53c5a049f6296946bc9b86edbe8e9","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-12 10:21:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6313413ba8516902d67a006cd1efa3b48eac56a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 11:33:42.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/869632 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"4bcba4de5573e7cd174c7d5c30d276d414de28dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-12 13:22:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"84178b95e6526ee37565df9bca437ae85772383b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-12 13:44:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0d8d2ec8ddb2bc622c958a3cc1795a0dd90fcb42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 14:34:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e7a1900a7825496cb730259072eb5ff9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/213f6ff6c2b64e679c8580c6a74708c5 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc63d1edae97487cb57e57f52f1fe494 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8b29e25a48f49c29c9f7185782912e3 : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d38eb37b54442128c156d7d84b9465d : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27582219021d4c19baf74b4ce193f9f4 : SUCCESS in 7m 47s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/aad9b68173cf4066bff678d8090b53b1 : SUCCESS in 1h 05m 12s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f5b089af29fe487da9f22861451bd36c : SUCCESS in 1h 06m 30s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce3d80ae467d4f9ca6575a5b38999c58 : SUCCESS in 56m 51s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5d62038bf40540f588f46821d9cbeaca : SUCCESS in 1h 00m 09s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3f48ebd501bc4251899ed3abac026bea : SUCCESS in 17m 04s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b9de5b4258e9480aa0c51b43dd4ffd2b : SUCCESS in 20m 36s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c6332e0857245ada46a3d345b3293e0 : SUCCESS in 15m 29s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c8099c083d84d9aa901ae7f258fbde5 : FAILURE in 48m 46s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6444d9aad4f841e686a7b45df2bb47ce : SUCCESS in 18m 22s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7204b5d704624a2f8bbf29ac1c0c0572 : SUCCESS in 15m 58s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/57ef0abb2e704d40a971783fd13b4af3 : FAILURE in 27m 49s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1501153b52b24b6f81f8e9af4b3964c2 : SUCCESS in 31m 53s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3638a4495fc341dcb82c073ef0144419 : SUCCESS in 13m 04s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/193c36fdef6a453f8cbcab485b3d488f : SUCCESS in 10m 41s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c7e52d4a766448fd81e7e8e9246f6b2c : SUCCESS in 17m 10s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e11fce2af38a4f90b7e1c39fdeda4904 : SUCCESS in 13m 25s","accounts_in_message":[],"_revision_number":2},{"id":"fc568a2f286eff0089ad3317779d428af27852b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-12 14:55:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7cdf926873e1158cfbd3b02370364fde31b31fd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 16:03:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d281326bacaf4beeb2749587a5b5f3cc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c39ba2903ad0464eb9a510aae946d2b2 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9f2dd5d480c4386970ffcd8320c785c : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7ce957740f54a618e94b0e7dd1e795c : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d48ab10c8a624140b20c90ddb25de92d : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b09b0e538a940a299417ad8afe6362c : SUCCESS in 6m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/914531e2807b4b469f3cd4b8eef4d7c0 : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ff0032d1556643abbbded0f96199c544 : SUCCESS in 10m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/092feddac57e4300845bf8d369619349 : SUCCESS in 14m 11s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/148a97c8c0a24140a48b1696b36a2265 : SUCCESS in 57m 53s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/a7bc03c2d78a4f059cd87c84398cf9d1 : FAILURE in 49m 07s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3908f62850d1427e80e5929f3f867d74 : SUCCESS in 59m 50s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f153387d68ec4ba4a06f458f32599c26 : SUCCESS in 53m 56s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2bc43dd8828e4ec4bef4de508361f88d : SUCCESS in 14m 47s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/983d509304624a2bb73104a81dfd3ffb : SUCCESS in 17m 35s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce6dc3ab0c62406ba754c83501007b59 : SUCCESS in 14m 35s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d83730df2cbe4615821cceeee8d1ecf0 : FAILURE in 44m 19s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/15102d966d4541038d42356032ca9a6f : SUCCESS in 16m 12s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e07718daffc2456dacd86a45cfb6016c : SUCCESS in 14m 00s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/2d2300e8d3d94fa6a09c55443ffba9d7 : FAILURE in 25m 55s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/550b3b2617fb4e838ee3355df01618e9 : SUCCESS in 27m 04s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/12bf7125d4ea4df4a6c676c9fc2f6843 : SUCCESS in 9m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/222187bb96f348dcbbe8628498d7e2d0 : SUCCESS in 10m 37s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1f0f6885e4c0466d860d1ae81d1d4731 : SUCCESS in 12m 36s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f89e8cc870714ed09090e338b58fb3a1 : SUCCESS in 16m 05s","accounts_in_message":[],"_revision_number":3},{"id":"3e3a2bb0996108ebc0433f362c1c77bf33376687","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-13 13:01:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2d7703b4885adb53ff9f70077da581bee717355c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-13 13:58:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3102bb5c7ccb136fb2d176c35e2f8f0c188e6d14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 15:22:29.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3357518b85484bac9094449949e8ebe7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44616075d12141a681db8386c99a2ad8 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af9295ece1ad41158c31cf3c16c1598b : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c73f0099471475e968c2a07f87d7d42 : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/290ae3b7ee0c4661be63a3b6a09d538a : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/917e83e3ed83441a97dbdd343479c0ca : SUCCESS in 7m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/25e40a03b5a6496da2a846dd6c4fa469 : SUCCESS in 4m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c2c90d8739c441749cfd6d9269983c67 : SUCCESS in 11m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e68a79ae598340ed99374402dd29f50e : SUCCESS in 14m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/80f1532413ae4ac8a149711b50a94c5e : FAILURE in 1h 02m 21s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1c83de6bd73844549b3564bfad75243d : SUCCESS in 1h 00m 48s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b039e746b8a44723b1c82a023a9e213a : SUCCESS in 54m 36s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8e88dc5847c841ad800bce57bde23cfb : SUCCESS in 57m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5e42516b453a44378e5973f251dcb662 : SUCCESS in 15m 26s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7bf915256f574411b3d85cb1dec62cda : SUCCESS in 17m 50s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb1e90b20e824708aeb5fe4fa244690d : SUCCESS in 15m 06s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce8ec31021e940838ba4f29dba7273c4 : SUCCESS in 1h 08m 33s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c5219c72ce0f4db98bc62b6c3be8d792 : SUCCESS in 15m 55s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6029958b551f4975921bc47e220ee7f9 : SUCCESS in 15m 43s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/9dce20b078a64b56b6badd04ca90d94a : FAILURE in 27m 44s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77b2268e33b745e4b357452c1a976e21 : SUCCESS in 48m 33s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/65b924f92e574fa9b7b610b084de2b63 : SUCCESS in 14m 01s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7005fbd3e9be44889be0b9b919af97bc : SUCCESS in 10m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/dd0cd5df8b9c4f3c89f550e65a27689d : SUCCESS in 17m 01s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6aa5a795c6f8467d9b2b846630e9f12f : SUCCESS in 12m 33s","accounts_in_message":[],"_revision_number":5},{"id":"73f2b370363b83e8df7ff129f3a12ebfcabd296d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-13 19:07:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c7163bef3c6bc9bc2f6826295125903b0bd18ded","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 20:26:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/600e6939e2054e98aa25e12f014616be\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c6ae1806e584cc8818b3be9a027ea85 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/523be04d0c4f453586ac45551c5e1f4d : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f32728c8d8c94b38a70610ed523f83e6 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16dc9eefc2634318ad2281ce7158357f : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0918d35f32a2486f89869179ef50faea : SUCCESS in 7m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3df29d5b0f1d4f3f8256a91667a97322 : SUCCESS in 3m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ff87ceb84f7e4ba2bcb280032f4ee194 : SUCCESS in 10m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/94950a68217043a2b1f31986aa9d3072 : SUCCESS in 14m 18s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/884b1c6c443649a0a647203cb364a915 : SUCCESS in 1h 12m 13s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/cab6e5605ef14b72aec588450e8de2a3 : FAILURE in 9m 03s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7a14d7cedcd414288f014454084354f : SUCCESS in 49m 13s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d70a1b85ac5443ebb188282e252ca219 : SUCCESS in 1h 02m 24s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3f37746fbadf4c1cbaccf1b2896c9159 : SUCCESS in 16m 56s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/faeea2fc2f86456cbad320cddec5dc6a : SUCCESS in 20m 14s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb100f1fac4044ca9927dfd979a68993 : SUCCESS in 13m 51s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/659c32fc97ee4e4e8d9259e7b20e0bc7 : SUCCESS in 58m 16s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/04a2a58e00154e0c854e60f9988ae13a : SUCCESS in 17m 32s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a6ab93f72774b8c8a11210fa8b0083f : SUCCESS in 16m 17s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/bb9aae3842024f67901131855924de59 : FAILURE in 26m 30s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5174797ee6534af2b6a0c50ddeceb86d : SUCCESS in 27m 28s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/658aea94a0214a93964a0bd69fc23eb9 : SUCCESS in 12m 42s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1662218e2eb54eef84ea96795dd5acb9 : SUCCESS in 9m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9320a0ddff13425e83f326ce33ad82de : SUCCESS in 15m 44s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af1529f2148147f384f68ef277acea2b : SUCCESS in 11m 45s","accounts_in_message":[],"_revision_number":5},{"id":"30d2cf2132428f3bed538696ad968b21a7e12440","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-13 20:26:13.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"5fd5f67a00a00538b7de1e126ba4d83db0724c38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 21:44:32.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c48b3a34104848cea88cb84ac4abac0c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/788af231505e4b7cb4c99d2219c45440 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc0712c0015b4df1b72dcb4554a9a9d9 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6c58ca4d1d84cf6ae8bfb5efb1c0fea : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e43394e22953422aa7f8d856f455186f : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0d13bf62c52477493d1be96e775dcd8 : SUCCESS in 7m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7a565f07e2ca4a66968fe643c47d15fe : SUCCESS in 3m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4bc6d80f690942c89b6ca7e2e05959ce : SUCCESS in 10m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f1060f9faaa74fe8a7fdaac11ff32846 : SUCCESS in 13m 17s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/79dff18ed6904272b73d9e2907ea2ff3 : SUCCESS in 52m 37s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/bed8accb45ab4cceb1fd6b3d0536c85f : SUCCESS in 1h 07m 47s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d328cf8a4b36449ab7091e5e093143f6 : SUCCESS in 51m 17s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ae015ea8b8b44d20b2c250b8e24ff925 : SUCCESS in 1h 07m 32s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9620e31dbf8048f495414c662788fe5f : SUCCESS in 16m 40s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/70860b6605244fad86bd96f133bd1850 : SUCCESS in 16m 23s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/265d60c6ffb44011878512f0cbe23601 : SUCCESS in 14m 31s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a19db59328e446cf9cca8a474f0b85bf : SUCCESS in 1h 01m 14s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7c838fd12bd547d095f69f111472ed48 : SUCCESS in 13m 55s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d2f24eb529e4fe3b616b451e4e8b1e4 : SUCCESS in 14m 14s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/baaf04d22a344f09accd17f877498b13 : FAILURE in 23m 28s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6002dc198c924c5597566377dd08d7b2 : SUCCESS in 27m 28s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a014d348815840cc814c64be28ebddf2 : SUCCESS in 14m 25s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99baf410d0f749a5bc953a463016027b : SUCCESS in 11m 49s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/253dc904868c43bab1780c1378af2d4f : SUCCESS in 15m 22s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e238df127774583867090fb1bd3cb3f : SUCCESS in 17m 15s","accounts_in_message":[],"_revision_number":6},{"id":"bb378e5c801ef397be3cdb6351e295f6a20e5559","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-16 08:46:15.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"fb61fb770b6cf83d52e43955269a40368ce51c88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 08:46:49.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4f1ecedd1516afae4904b292cd2a328059f29e24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 10:06:26.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a30ad2ae73040d7819bc754888a9fc0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30ca3ae332354577b64b12d26987fc71 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fedc77400a9944a6a03d63860dd1359d : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/99b9ae394cca48b880b69808547a8de9 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/395c62d2b1324b18a3f0914e4e3ad26a : SUCCESS in 7m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bf05704a709c48ebba07fcd74c484c2f : SUCCESS in 3m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1e285d955d3949b19b09faa779440008 : SUCCESS in 9m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ee9b0c4882464ddbb1d6bc04a90a0ff7 : SUCCESS in 12m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3122a050838c4b6ab67f7b9b3432e59f : FAILURE in 58m 51s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30f750c9feec4d7e8e1f1ffb04fcde83 : SUCCESS in 1h 00m 39s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c4ca28c8f2a94e2e824c7bd473343428 : SUCCESS in 58m 48s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0465f9482a674565af26236fbf6abd76 : SUCCESS in 17m 51s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff73cf5405e040df9cb31a867ffef2a4 : SUCCESS in 15m 26s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d42e89634c2447c8601661a92bc5e36 : SUCCESS in 1h 18m 06s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/84edb0591f98450eaf307dfb1985b3c2 : SUCCESS in 19m 45s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8fff6fd5fbb43c3ad8b0bfd1becd7d6 : SUCCESS in 14m 54s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/238c5918b8924a0cbb2f8cdccf2651fc : SUCCESS in 28m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/371fb853663547a492cc6bbbace5dc56 : SUCCESS in 11m 48s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30c8700a73094b2099efde6fb4189615 : SUCCESS in 13m 11s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3d40047f8130409a8bb5598958e01daf : SUCCESS in 13m 51s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27f1fb3368bb4013b546ebd361803740 : SUCCESS in 14m 41s","accounts_in_message":[],"_revision_number":6},{"id":"072c7877bc01c034a37a75965fedc11470990831","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-16 11:34:45.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d4d214d5ef374e564ac16e795ac926baf5eb95b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 13:25:44.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42750bcd3feb467a90ca0874d24ddd21\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/391fcb7552464e248e1e47061c2f9435 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49701450fa434fc4b79024497ef23787 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ff1966e7c5b41b0ae5b442a5b9a2916 : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c424b8549e94b5f96d9cd4bbf127191 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e2fdf216d554e4aa336004f771dc7e6 : SUCCESS in 7m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/79425789202445fabb7eadbf0c504780 : SUCCESS in 3m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/50e07709cc6d41eaabc4ee9f41f501bd : SUCCESS in 11m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5d5d405b33f144f6906795d9aa44561c : SUCCESS in 13m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/fbb377e622404dd381e1bdf82dc90a74 : SUCCESS in 1h 00m 11s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/26b10b42c3154dd6b278b8c16bc9eccf : SUCCESS in 55m 57s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f6b9df4ec82470d9db97a9e0601aac5 : SUCCESS in 50m 00s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/87e7d9dbc0034b59904158a52e8b1209 : SUCCESS in 57m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d91ed62bdb2747e895bb30dc58eab7f4 : SUCCESS in 17m 29s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a296ef71e4524bf5a147287529369b45 : SUCCESS in 18m 17s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/312a2b5923044cf59f923b1f24a1cc39 : SUCCESS in 17m 00s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d014ab0c8a1d4d7082b69a2456783d5a : SUCCESS in 1h 25m 37s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3484ed1cb4a0400e9827a18f9b849b2f : SUCCESS in 17m 59s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42d49992664145c0b82357d3d4b6441d : SUCCESS in 20m 22s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/7a49f147ed364c4b87f57ea6091b7cb5 : FAILURE in 24m 16s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82663974708945ad85568564359d9b34 : SUCCESS in 31m 56s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e08d2c8e7bb54cb584aeec8f2ddf8d92 : SUCCESS in 13m 57s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3740b3399f44b259abefb141389b2ec : SUCCESS in 11m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8af09021a6dd44cd892bbbe5f41dee8a : SUCCESS in 15m 38s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0432bd034ef446a79a55f92724680b19 : SUCCESS in 12m 19s","accounts_in_message":[],"_revision_number":6},{"id":"e749a0a175e0452b78247be88f23ea4d407fd369","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 13:26:32.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"37da9ebcc24ea0153a0d5f0e8c34248ad1709c9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 14:41:02.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/834f1c59512149a8b3250d0bb838109c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5649611c8da74b6287917b309c2f8d40 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9dd983bd6ee4efeb940dd786bea44c4 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd2b970b89924c70af4f86b730cb1a85 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e3bba92bcc54a4999194b8f5c71e66e : SUCCESS in 8m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2e75115e735646279ab0aa728b02e32f : SUCCESS in 5m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5958f9db43c941a3b4d196d3d52db395 : SUCCESS in 12m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6beb50b162674aa88102d495a4a21a04 : SUCCESS in 12m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/22a67d23caf04465baf986ebcf322ccd : FAILURE in 56m 52s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3d62651cd9e48978680f48ba913c08e : SUCCESS in 45m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ea9273cad5384a8db944ed9fc0038a9a : SUCCESS in 46m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0a4b78c6dfdf4185814b228e64364f75 : SUCCESS in 13m 50s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3dc78838404a41cf82f99ea4ff3a0199 : SUCCESS in 12m 50s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c919ea763b249a98f1bf84cf74c0603 : SUCCESS in 56m 16s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/41e9272db4104b6587c3e8819266340f : SUCCESS in 15m 45s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ca87192e5314e7ca129f60c29ddca84 : SUCCESS in 14m 50s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4871250899c8418bbab2c8217d9563b9 : SUCCESS in 30m 38s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0043a70dfd644db4a932172637ef1e84 : SUCCESS in 13m 30s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dcc5fb0505474623ac076557381bde4a : SUCCESS in 12m 59s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0e328ae445004ca986c83e40ec471ff0 : SUCCESS in 16m 08s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/abe3f50682ba47faa7a0add5d572add6 : SUCCESS in 15m 46s","accounts_in_message":[],"_revision_number":6},{"id":"852af5521d4dc13aa128f5a315d8779b8e7774f3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-16 14:56:27.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ff69cd023e0f73e82ec38bf48bdf6253fa3e3a93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 16:18:56.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f321bbbfd7df4f74bdf15771109204eb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b40acca445da41fda0290137cfc6d0a1 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1587a4d9ade4110b3e94ee4f3c0b34d : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fdc06966d7b487291259ed7d0abcd4c : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c021ddfc1244dcda87199ccde6d604e : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd76b5c155344bd08accbd473505af79 : SUCCESS in 6m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/71b792a9e3e546bdb32a6803504a75b3 : SUCCESS in 5m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/97cbf616e3c3476686ea25e479f7bf09 : SUCCESS in 12m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c36ab5f0f0fd4d628531b719e62163dd : SUCCESS in 14m 31s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3a7015acb23e4b9d8b15ddbb5b0a3397 : SUCCESS in 59m 48s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/d219706d7b554343b1f31a8fb52f14bc : SUCCESS in 57m 41s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6ca3658ff344973bb6ab7b4e2a25f86 : SUCCESS in 49m 25s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3e778a0b5d884524ac0a430047d9eda6 : SUCCESS in 59m 18s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8a0382a59447429b9a734c3b4066335d : SUCCESS in 16m 20s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/5553ba9304e9463581bf69902a026c10 : SUCCESS in 17m 07s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/602b77b8c24a44c3b511ea75daa20fc4 : SUCCESS in 17m 00s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4bad01c15c4c4ffca26c69e0494994be : SUCCESS in 1h 05m 21s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3996eaea15c944ec973ce85b1d67ec18 : SUCCESS in 18m 49s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/120ccc8eeca94f8db015841298665f76 : SUCCESS in 16m 52s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/43fa13ff82fa43b4b73e1a9ab2fe8d01 : FAILURE in 28m 33s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75c0a880eac64ef0b7a04394a7bec2b2 : SUCCESS in 28m 14s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f86ea85b4ea546b783c859fcf4b978ec : SUCCESS in 14m 06s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4122d6740a3c4234949d987a3480274b : SUCCESS in 10m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4b9569ac4ff74465911089c12f2f9157 : SUCCESS in 17m 12s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/219c9bcc43fe4c8eb6a825e47178925b : SUCCESS in 14m 25s","accounts_in_message":[],"_revision_number":6},{"id":"70d1bbe07f042cf13bc3653d508db412261d8a49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 16:19:29.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4e909df7112e1c810b66e0918b7b2592fb2eab17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 17:51:46.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f73520a32b574cb48248e9979f390a23\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04d88c77414841b38b2d0a1b3635e470 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4f0d0a5209b437baa56c74f2deb9f3f : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3421c08437b54b75a10d9520fbcafb16 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edbc173acc184946a031294a872b8fbc : SUCCESS in 7m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/54930b97c5af44f385f7598b379632ee : SUCCESS in 4m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/523a907c55e6435d937b58be86b7e36b : SUCCESS in 11m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ca52d0aa846844e5aa15a7f906c9a322 : SUCCESS in 14m 38s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a400205370ac449a81ad6ada79d5985a : SUCCESS in 54m 03s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a271ea19d0ec425fb61d422db161bcb6 : SUCCESS in 48m 23s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4ed6662d46ce4ca0ab0c183272ca1f2d : SUCCESS in 43m 35s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f06fae42d8ca4bffb87b20e1ca22dc36 : SUCCESS in 16m 53s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25a4fa521f70497f9adb431c66c1fb80 : SUCCESS in 10m 43s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a7d9f5f7b004b33b49fab813637f20c : SUCCESS in 1h 32m 03s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/af660d3f16fb4c51972c37344afe1222 : SUCCESS in 13m 15s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d03819ae4a334b18b336fbfe4523e8bb : SUCCESS in 12m 49s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a20dd9c75c04d3498d7982b237b920c : SUCCESS in 22m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c06d2a2f6fe149f5bca09c2e4c94e673 : SUCCESS in 11m 04s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/711b26433e9243c3b6df98d8a9805704 : SUCCESS in 8m 06s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8dd48f98098545b0bc63bd2caef09d0c : SUCCESS in 13m 28s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7b1a54489cc4e2b9464068a44e459c4 : SUCCESS in 11m 39s","accounts_in_message":[],"_revision_number":6},{"id":"51aa73ff7726d509fad1a50146cb514e9c65c232","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 17:51:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"a6f1cfd8469ca640f2eac89029b36ebe668825e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 17:53:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2945bfaf92e04c3482684f42a26a47f6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4e939443bfd495998c8629dc2525458 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"717f4d9fe1bb5798e9e4b0fce24e0131d59120e6","revisions":{"f398c3347257955a133d56bdf8ffbcd5edcea7d6":{"kind":"REWORK","_number":1,"created":"2023-01-10 10:43:45.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/72/869672/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/869672/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/869672/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/869672/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/869672/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/869672/1"}}},"commit":{"parents":[{"commit":"030770376ee59776e0ffc481412c69cc5e93973d","subject":"Revert \"Use OpenStack Zed release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/030770376ee59776e0ffc481412c69cc5e93973d"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-10 10:39:05.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-10 10:43:29.000000000","tz":60},"subject":"set previous_release to zed","message":"set previous_release to zed\n\nChange-Id: Ic0608bc6033025cb47655d601ffaf3744637832f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f398c3347257955a133d56bdf8ffbcd5edcea7d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f398c3347257955a133d56bdf8ffbcd5edcea7d6"}]},"branch":"refs/heads/master"},"3899c2f3660bbdc03b89a90df0e4a16ede8c25e9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-01-12 13:22:04.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/72/869672/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/869672/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/869672/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/869672/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/869672/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/869672/2"}}},"commit":{"parents":[{"commit":"0c11102afea6d85001945071f27b9f8af7167512","subject":"Revert \"Use OpenStack Zed release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c11102afea6d85001945071f27b9f8af7167512"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-10 10:39:05.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-12 13:22:04.000000000","tz":0},"subject":"set previous_release to zed","message":"set previous_release to zed\n\nChange-Id: Ic0608bc6033025cb47655d601ffaf3744637832f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3899c2f3660bbdc03b89a90df0e4a16ede8c25e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3899c2f3660bbdc03b89a90df0e4a16ede8c25e9"}]},"branch":"refs/heads/master","description":"Rebase"},"5fb040e9f422cb5818af642a04e1273f4ccdc7c3":{"kind":"REWORK","_number":3,"created":"2023-01-12 14:55:48.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/72/869672/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/869672/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/869672/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/869672/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/869672/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/869672/3"}}},"commit":{"parents":[{"commit":"0c11102afea6d85001945071f27b9f8af7167512","subject":"Revert \"Use OpenStack Zed release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c11102afea6d85001945071f27b9f8af7167512"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-10 10:39:05.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-12 14:54:44.000000000","tz":60},"subject":"set previous_release to zed","message":"set previous_release to zed\n\nUpgrade CI job needs clouds.yaml to be used from Zed\n\nChange-Id: Ic0608bc6033025cb47655d601ffaf3744637832f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5fb040e9f422cb5818af642a04e1273f4ccdc7c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5fb040e9f422cb5818af642a04e1273f4ccdc7c3"}]},"branch":"refs/heads/master"},"1eac951f45c9daf14c42855b4f922bd73094f483":{"kind":"REWORK","_number":4,"created":"2023-01-13 13:01:29.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/72/869672/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/869672/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/869672/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/869672/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/869672/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/869672/4"}}},"commit":{"parents":[{"commit":"0c11102afea6d85001945071f27b9f8af7167512","subject":"Revert \"Use OpenStack Zed release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c11102afea6d85001945071f27b9f8af7167512"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-10 10:39:05.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-13 13:01:16.000000000","tz":60},"subject":"set previous_release to zed","message":"set previous_release to zed\n\nUpgrade CI job needs clouds.yaml to be used from Zed\n\nChange-Id: Ic0608bc6033025cb47655d601ffaf3744637832f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1eac951f45c9daf14c42855b4f922bd73094f483"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1eac951f45c9daf14c42855b4f922bd73094f483"}]},"branch":"refs/heads/master"},"4ad09400d96bda13ae4b645453192b05d9451f43":{"kind":"REWORK","_number":5,"created":"2023-01-13 13:58:09.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/72/869672/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/869672/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/869672/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/869672/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/869672/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/869672/5"}}},"commit":{"parents":[{"commit":"0c11102afea6d85001945071f27b9f8af7167512","subject":"Revert \"Use OpenStack Zed release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c11102afea6d85001945071f27b9f8af7167512"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-10 10:39:05.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-13 13:57:54.000000000","tz":60},"subject":"set previous_release to zed","message":"set previous_release to zed\n\nUpgrade CI job needs clouds.yaml to be used from Zed\n\nChange-Id: Ic0608bc6033025cb47655d601ffaf3744637832f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4ad09400d96bda13ae4b645453192b05d9451f43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4ad09400d96bda13ae4b645453192b05d9451f43"}]},"branch":"refs/heads/master"},"717f4d9fe1bb5798e9e4b0fce24e0131d59120e6":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2023-01-13 20:26:13.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/72/869672/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/869672/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/869672/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/869672/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/869672/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/869672/6"}}},"commit":{"parents":[{"commit":"0c11102afea6d85001945071f27b9f8af7167512","subject":"Revert \"Use OpenStack Zed release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c11102afea6d85001945071f27b9f8af7167512"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-10 10:39:05.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-13 20:22:43.000000000","tz":60},"subject":"set previous_release to zed","message":"set previous_release to zed\n\nUpgrade CI job needs clouds.yaml to be used from Zed\n\nNewer version of ansible-collections-openstack uses different return\nvalue. [1]\n\n[1] https://review.opendev.org/c/openstack/ansible-collections-openstack/+/841224\n\nChange-Id: Ic0608bc6033025cb47655d601ffaf3744637832f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/717f4d9fe1bb5798e9e4b0fce24e0131d59120e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/717f4d9fe1bb5798e9e4b0fce24e0131d59120e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
