)]}'
{"id":"openstack%2Fkayobe~869746","triplet_id":"openstack%2Fkayobe~master~Ic4dd537bee826ee89c009bf2c4c95781af19cbad","project":"openstack/kayobe","branch":"master","topic":"networkmanager","attention_set":{},"removed_from_attention_set":{"25600":{"account":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"last_update":"2023-04-28 11:12:54.000000000","reason":"\u003cGERRIT_ACCOUNT_25600\u003e replied on the change","reason_account":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-11-01 16:21:53.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-11-01 16:21:53.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic4dd537bee826ee89c009bf2c4c95781af19cbad","subject":"Allow overriding earlier use of NetworkManager","status":"ABANDONED","created":"2023-01-10 16:54:11.000000000","updated":"2023-11-01 16:22:07.000000000","total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"4fac5071ff608a88a3aa11d633e089334a56846a","_number":869746,"virtual_id_number":869746,"owner":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-01-10 18:03:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"disliked":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-01-11 14:45:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":-1,"date":"2023-01-11 09:38:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-04-13 15:32:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"date":"2023-04-27 09:32:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-10 18:03:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-11 09:38:40.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-04-13 15:32:39.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-04-27 09:35:16.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2023-08-27 00:03:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"9109dc377c98c065ef46b24606a62a70ed5ffaf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-01-10 16:54:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"edfa26d0aac0a7c1a496fe16207c0d9ce40264d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 18:03:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c9e7e54c5fd4833b379dc211f50f2e4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90acd7bf1b4749989e49d5aa791f8a02 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aa9492fcf404ee384cfeb419c961a53 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d666dcd724d0464b9dae4bb49959620d : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7f0f738400946e3878d56f0f8872067 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3608629ab644855b2d6d69cdbabfd53 : SUCCESS in 6m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/165e0bca9f4640d6b5f4149c52fad781 : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/277ad2e646b14fa2b89e4a6dd30d7df8 : SUCCESS in 10m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/775d9072d8574944b0f59055f74bd71a : SUCCESS in 12m 47s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/fe74e484760e4a0295d983e8ee6f8738 : SUCCESS in 1h 03m 32s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/69d93ffc12c14740bdffa776654d58b4 : SUCCESS in 46m 20s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/467196e4f3ce4848b23becbf5e4be1c0 : SUCCESS in 57m 01s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/004cc8c27ecf4e53988b6e38d1e33117 : SUCCESS in 53m 52s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b45601c9f3ed4433b9d4ff91b5e36058 : SUCCESS in 17m 00s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/470e09a751294786b09c1aa52aa058a1 : SUCCESS in 15m 33s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a74dcde081f64433b38ed9279c442c58 : SUCCESS in 15m 04s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc35e3c8df7648f9b82176ac64d8c5f0 : SUCCESS in 1h 01m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5d99f17957394971b9a550cd0cef2de2 : SUCCESS in 17m 02s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c60b377861d94685bdcf7277b275dd2f : SUCCESS in 12m 24s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8715c8410ac54402b3dd72946013d158 : FAILURE in 23m 44s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6aa80fb1491d425b9dd3c95f33524beb : SUCCESS in 28m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8e01732131bf485bb180c6e1e9024180 : SUCCESS in 13m 52s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3180d008b8814ec2990d485bffe61719 : SUCCESS in 10m 45s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c36b770d6d294ae99155cd467e3978bb : SUCCESS in 15m 24s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83b440862aa243ce920936609841c893 : SUCCESS in 13m 27s","accounts_in_message":[],"_revision_number":1},{"id":"27f5464708f3199b9662d0902275f2b118a5a2e2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-11 09:38:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e4f301bb761c82b9aa330a91b57023a0a72ad386","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-01-11 11:13:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dce2d065ac3cb7a61631453c5a8a6151ab867d37","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-01-11 14:45:18.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ed8b9bd6a07350d8098dab7985246cbe4745bb35","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-13 15:32:39.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8849addc929cfe019cf5d2b1f4356d062c00ae7f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-27 09:32:19.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7e381e1072d700980bd25ba5acc09e43b1d2c0d0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-27 09:32:27.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"52f799f7ac6de02a5c7f7a011965b27bb62fcb41","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-27 09:32:29.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"a64a21f41da49a2d0a6b8f031a082959269c21a2","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-04-28 11:12:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ac173e09d7b684e7d5cffcd656d89ca71cc216fb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-01 16:21:53.000000000","message":"Abandoned\n\nI\u0027ve cherry-picked this to Yoga since it only makes sense there.","accounts_in_message":[],"_revision_number":1},{"id":"4fac5071ff608a88a3aa11d633e089334a56846a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-01 16:22:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fde392d0b407fca795495c8b90987b15b42f74ca","revisions":{"fde392d0b407fca795495c8b90987b15b42f74ca":{"kind":"REWORK","_number":1,"created":"2023-01-10 16:54:11.000000000","uploader":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"ref":"refs/changes/46/869746/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/46/869746/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/869746/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/869746/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/869746/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/46/869746/1"}}},"commit":{"parents":[{"commit":"ebd5799fd4b77a138cf7790a8094eb859e8898f3","subject":"Merge \"Replace references to Docker Hub by Quay.io\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ebd5799fd4b77a138cf7790a8094eb859e8898f3"}]}],"author":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2023-01-10 16:43:43.000000000","tz":60},"committer":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2023-01-10 16:53:02.000000000","tz":60},"subject":"Allow overriding earlier use of NetworkManager","message":"Allow overriding earlier use of NetworkManager\n\nveths have to be in nmconnection format, as nm-settings-ifcfg-rh has\nno support for reading veth network-scripts, as can be seen by the\nlack of a table at [1] \"Table 37. veth setting\". Hence why we\u0027re not\nusing NetworkManager with old network-scripts.\n\nBut if you set interfaces_use_networkmanager to true, you can already\nstart using NM on rhel8 platforms if you so wish.\n\n[1] https://networkmanager.dev/docs/api/latest/nm-settings-ifcfg-rh.html\n\nChange-Id: Ic4dd537bee826ee89c009bf2c4c95781af19cbad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fde392d0b407fca795495c8b90987b15b42f74ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fde392d0b407fca795495c8b90987b15b42f74ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
