)]}'
{"id":"openstack%2Fkayobe~869931","triplet_id":"openstack%2Fkayobe~stable%2Fzed~Ie1a56ebd6fe3283a2b5b077f658ff1cf3e006784","project":"openstack/kayobe","branch":"stable/zed","topic":"docs-quayio-stable/zed","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-01-16 12:18:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie1a56ebd6fe3283a2b5b077f658ff1cf3e006784","subject":"Replace references to Docker Hub by Quay.io","status":"MERGED","created":"2023-01-12 11:34:49.000000000","updated":"2023-01-16 12:19:56.000000000","submitted":"2023-01-16 12:18:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869931-docs-quayio-stable/zed","cherry_pick_of_change":869442,"cherry_pick_of_patch_set":1,"meta_rev_id":"d3f568d29d6028425d82464f1c135685c0a9cb21","_number":869931,"virtual_id_number":869931,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-16 12:18:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2023-01-16 10:10:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2023-01-16 10:10:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-12 12:50:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-16 10:10:37.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"93c40a1414033862d7bc74e55a612bd93655542b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-12 11:34:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"037d242faaa392f5dc3881e20d11f396fdcaaa8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 12:50:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/25851eb010344d499bc483e4c99fcc82\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ec982be4beb4f2e85fe48c8bbeb28c1 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9edd064d76648ae890997d3c41fecd9 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7465785d38d94ae8a74a8f4797f83631 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/742becf1543e4c1584449a94b575b658 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df0a735ac0994ad4b5a3799a0c3f8f71 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/584582ef622c464eb81b46afe8e7c842 : SUCCESS in 8m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aa8cc56515c64d3f9898f888bf6db0ef : SUCCESS in 3m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d72d3f2db26e46be9bc90df1538ab87c : SUCCESS in 11m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/79dc6c34cb0c461791eae1a220e6ef94 : SUCCESS in 15m 42s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/91f18b3c75e64e3f8fefbda334a63a8e : SUCCESS in 55m 57s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f402cd800a6141568a05905646fe2a38 : FAILURE in 51m 19s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/861ad92950f94ef29908401247819de6 : SUCCESS in 57m 00s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/bc87fa20281249eb810ab81af46cfdfd : FAILURE in 43m 57s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/bccfb80c06654617b98da122bdfbb293 : SUCCESS in 18m 22s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/761d583fb3f5481e9b3aaded6975f25f : SUCCESS in 18m 05s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0edff6b7bddf4f758e247c02aebed7fa : SUCCESS in 22m 01s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3389789ad6534e5e975450c1f345e9b2 : SUCCESS in 1h 13m 06s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ab35dc442ad14515867cfd9624d93080 : SUCCESS in 16m 48s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b78b34c255e74687b4ddb9003989bc68 : SUCCESS in 16m 41s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ef56f7dc39e74cb5a3f5f9dc4caccaf2 : FAILURE in 22m 58s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ab97bce1f4640558527a595c2a7194e : SUCCESS in 32m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8084d259377d4fecb37be09f129c51d8 : SUCCESS in 14m 04s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3917de00adae4239a6fcbf243c873496 : SUCCESS in 13m 18s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/26ec2ba896e641e2ac904f213605c321 : SUCCESS in 14m 51s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34628f8fb5ea450a9f6324e02c2f9441 : SUCCESS in 15m 54s","accounts_in_message":[],"_revision_number":1},{"id":"b738e5875d0c17a5ac207e1b0893abeec0367543","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-13 09:09:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"60c0786e8087990102c0f34d1b6452f07ff0cb51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 10:47:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b72942fa7a140e69f0b1ca204e9496b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/137aa76429c949638aa3a30ebb0b34d6 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec98385720c046cf9252f63c32cd6052 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5094825a0b2f44098acf023c207672cd : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e760dcc5afa84820bb9d8018d35e160e : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a069f8e17314fde8df89fa17853c01c : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05c46b5fec34467aac4359f010a6f2c8 : SUCCESS in 7m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f56050e0f7554217b7ccbe86c785e996 : SUCCESS in 3m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2c56c42cbd7945ac8ef98402c113b666 : SUCCESS in 11m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7ffaaacc1c75400ba00f23278c423752 : SUCCESS in 13m 30s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0c177e071ce8476faca95a574a2e7343 : SUCCESS in 1h 06m 07s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1e94c2246b494ec68bef3a40c9b5514d : SUCCESS in 46m 01s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af7872aabf1245c896c8ef46449a2782 : SUCCESS in 55m 17s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d24c9c9cd50d47479efc1cb8fdc8fd49 : SUCCESS in 51m 12s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/61fcd2d1bfaf495f8f7b34944125d74c : SUCCESS in 18m 51s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/dd0bb5b2343f4f01b8ca3df9a07ed09e : SUCCESS in 17m 53s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48e050dbf39d4fa0836e7188beebb42b : SUCCESS in 13m 37s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/325bb0268f6745d8b0ab1a1b7cd69401 : SUCCESS in 1h 19m 07s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/498ce0c1fa8f4b8c8a99da55526e3a27 : SUCCESS in 16m 29s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e8ce5a4a09b491bac5f5db95088b34e : SUCCESS in 14m 30s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/5a432e3cb5bb460f915ca54a77746104 : FAILURE in 21m 04s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97b68c8b72d14fbeb68b83e9cd42cfc2 : TIMED_OUT in 1h 35m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ffd9647c611340d59af8d1ea01c81dc4 : SUCCESS in 14m 27s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d44eaa4ce4b94e6093d1e4d89afc8c1b : SUCCESS in 11m 52s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8a0d676433e1472e98a0806d8e7a10e9 : SUCCESS in 13m 15s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/abbc76aa219c453daa27a8b2a8817b38 : SUCCESS in 15m 22s","accounts_in_message":[],"_revision_number":1},{"id":"9b75db303a33b75b590c7529c12d83330b2e12b1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-13 11:36:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2366fe0c58276fe1bfd51fbe112dd41123844076","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 13:12:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a5df96599414c3ba88957d50c66ccba\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9974b431be34e808a8182471359a5bf : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fc6edc6a2aa4e92aa69bb6f6ca89f8e : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d056b8b0ac74adbb57ebe2d9ad3c4c1 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/047c93b34f144e35b33246ba03f8d3ed : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0583ce51d612440f891888d6fea0b8d4 : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17d1c45897bd4a76b1182a675edd7702 : SUCCESS in 7m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/283d1f919a8343e4a38357a49af29094 : SUCCESS in 3m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c9d22841f0d2466cb5521b90aa1d5d22 : SUCCESS in 11m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/beed1fc7a4674bffb1aa3071a4e8c3c2 : SUCCESS in 11m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0d73cf81144141f599a1ad0676fb088d : SUCCESS in 1h 02m 14s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f9bb2a6e6a674bf08a0bfaee78de7056 : SUCCESS in 51m 25s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac52d46485ef4a3eb4105df56379e8bb : SUCCESS in 54m 45s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/54158d7f8aa84577a14031cc1e8fe924 : SUCCESS in 51m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ea56f6d1258d4253852fe05555e3aa51 : SUCCESS in 17m 10s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/433d7abcd866456b8bb96d5f199085d9 : SUCCESS in 16m 22s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/983be06efa1443998c68cde78ea0d4ab : SUCCESS in 12m 58s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42a09f8e4f9c4ba7b1ea2b8ff044229a : SUCCESS in 1h 10m 03s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2f0040aae5b64aa6b2fa4c9c3389555e : SUCCESS in 12m 45s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3056fa78b1e14ecd8b8d1ae330f5e9ef : SUCCESS in 14m 58s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c76811a5ce324ec6bce83d7f87165131 : FAILURE in 19m 09s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8225a44faa54536a518ca5c73f5ee50 : SUCCESS in 1h 33m 13s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/34374e5bdbd64b9e84fccd3bacc6268f : SUCCESS in 12m 35s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c76220e994c4f2f8329dd671c88ab98 : SUCCESS in 10m 35s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/93b56e235e454f1090f3323f985a70fd : SUCCESS in 18m 21s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fa2bb578ebb4b9cb1d5505c383ba25b : SUCCESS in 13m 56s","accounts_in_message":[],"_revision_number":1},{"id":"610830c2650be0ee3c5dd740b5ca1dc5d7d7fd68","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-01-16 10:10:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1571e69bd05ba430b687e188f6f265464765005e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 10:11:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a9d2abea725fb297dc14f9363f6c150169bc44e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 12:18:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c31098216b841bf81b0f583853583d8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2d8cb72e50846ea8a68b671df040859 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb1cf0b6b36842798165ee4a52ecb625 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30cb176083f84a30a574e122509baa62 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dddc14f387424c9796788b96f4e38677 : SUCCESS in 6m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fc7f368a680e4766955b0b809efc281f : SUCCESS in 3m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/11aed70a04f243d8b44f30112cd7fa58 : SUCCESS in 9m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3471cb67181446ddb94ef82f10ac89fd : SUCCESS in 15m 17s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/010c2fee1df04caaad594786e2fd575f : SUCCESS in 1h 04m 09s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb63ef4e7f9f45a8a12fe2c8995f829a : SUCCESS in 47m 43s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/03d3e3bccf5344d0a2c34857020274fa : SUCCESS in 55m 53s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0423bfa1dbf449e0bbb6d82f97d17865 : SUCCESS in 17m 52s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3248ddeabda4feba1bbbd29081b4539 : SUCCESS in 14m 58s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9f6865a012440c797e0eb4223a05eb9 : SUCCESS in 1h 09m 47s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/19f71f461ba64e218de5ec8c24fe213b : SUCCESS in 16m 27s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1565bd65113450eb5aea25fa6b14205 : SUCCESS in 16m 15s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3c9114b3fe94080a1a6cec01ee9c2a0 : SUCCESS in 28m 59s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5e5e578b3e024914943e426895e95294 : SUCCESS in 11m 18s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e7c5fae2fb44454815d4a2656faeb4b : SUCCESS in 10m 59s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0ddcfc6a2f8540df9d0e0c09092b70cc : SUCCESS in 15m 43s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23a33f63563242f8983a38eae1435da9 : SUCCESS in 12m 19s","accounts_in_message":[],"_revision_number":1},{"id":"dcf563052eaf85fa936b7e41a72c5c98958cf6ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 12:18:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d3f568d29d6028425d82464f1c135685c0a9cb21","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 12:19:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9744a96e46624d57a6a46352b047b191\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c07b7a96305742dca93e3ecf4322eb61 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d8ecbc88456a1a69084f920df9c74c61ac3ae62f","revisions":{"d8ecbc88456a1a69084f920df9c74c61ac3ae62f":{"kind":"REWORK","_number":1,"created":"2023-01-12 11:34:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/31/869931/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/869931/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/869931/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/869931/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/869931/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/869931/1"}}},"commit":{"parents":[{"commit":"88da20237a0858ad21530d324509a51024d3baf1","subject":"Sync kolla_tag with kolla-ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/88da20237a0858ad21530d324509a51024d3baf1"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-01-06 10:11:46.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-01-12 11:34:49.000000000","tz":0},"subject":"Replace references to Docker Hub by Quay.io","message":"Replace references to Docker Hub by Quay.io\n\nKolla Ansible uses Quay.io as its default registry.\n\nChange-Id: Ie1a56ebd6fe3283a2b5b077f658ff1cf3e006784\n(cherry picked from commit 6e78aa89149add7b15c457266d5712486ef4503d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d8ecbc88456a1a69084f920df9c74c61ac3ae62f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d8ecbc88456a1a69084f920df9c74c61ac3ae62f"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
