)]}'
{"id":"openstack%2Fkayobe~869932","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~Ie1a56ebd6fe3283a2b5b077f658ff1cf3e006784","project":"openstack/kayobe","branch":"stable/yoga","topic":"docs-quayio-stable/yoga","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-01-16 12:18:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie1a56ebd6fe3283a2b5b077f658ff1cf3e006784","subject":"Replace references to Docker Hub by Quay.io","status":"MERGED","created":"2023-01-12 11:34:59.000000000","updated":"2023-01-16 12:19:45.000000000","submitted":"2023-01-16 12:18:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869932-docs-quayio-stable/yoga","cherry_pick_of_change":869442,"cherry_pick_of_patch_set":1,"meta_rev_id":"6ce9091f5882b4aadbc9b38a75ec5ee668dcc040","_number":869932,"virtual_id_number":869932,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-16 12:18:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2023-01-16 10:10:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2023-01-16 10:10:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-12 12:51:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-16 10:10:27.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"dfbc9bee87d45c30d3f920a895c8fe724d508012","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-12 11:34:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7bb4232838c1d39e935f2a792edf7616644b7c60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 12:51:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c466c59ff7ae449ab4b883b3a3070280\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7b3cd0ff7eb446e9951f80537f72651 : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ce225d82bdd4d2d933b026a63158125 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/067410350613411eb122bc08454db515 : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d33b91bea71747419e31d8e6b9c6bea9 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b99fdf4ebfe14758915615e4ff5991d0 : SUCCESS in 9m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8a710ad8e38d4a36affda888b80fe7d2 : SUCCESS in 6m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3727c8f039084b7aab33acb151ce8df3 : SUCCESS in 13m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/08358bd10c524a58b5b8165fff62e182 : SUCCESS in 15m 07s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2de5b7a5f2314385b60eec3965dc5371 : SUCCESS in 56m 37s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca94575be1104728a3c823444d98a4d1 : SUCCESS in 52m 21s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd41293488ed4ea4adf6860d6910d079 : SUCCESS in 13m 57s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69ae6adc31e6423eb16e1ecf64464f4d : SUCCESS in 13m 05s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9642da918454df78a245f12f8aa5bff : SUCCESS in 1h 14m 02s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e847934063b14e17aaa80092848cd76f : SUCCESS in 12m 56s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33604d4b50c244979d1455eb8d421d60 : SUCCESS in 18m 37s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e20566ae5944a8fa5fbaee5c81becf1 : SUCCESS in 13m 19s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a66553054ab64c57adf52dea8412cade : SUCCESS in 12m 22s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc00f4e00973499097e4bbff4a6d5fcd : SUCCESS in 13m 15s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/274b9b87781342d7a18a97b20c350eea : SUCCESS in 12m 56s","accounts_in_message":[],"_revision_number":1},{"id":"1760516adbae6307caae64ef0063178001a92ff3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-01-16 10:10:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"20628a9890198f18ebfadd16121086221adae403","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 10:10:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"98a329c44d1452a0457442a36e70b4dc9308df31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 12:18:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01ba729536a1489a963b89e34f2b5d8f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57d8bcccc4914f9dbd9217e3064fe3df : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c5134e97b5c4515a867177838dbb9ac : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/207bfaa721b64b45894a17e15cf0a180 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b091179a6d046fbbaf4843722d4adc5 : SUCCESS in 10m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aad98f42734c4e958d9b0e0865ecaf3a : SUCCESS in 8m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/007df0cdd6ac41d6a9f71b798c704b91 : SUCCESS in 14m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/adb1c12dd40f4fd9ae5ab5c0bef8a003 : SUCCESS in 17m 43s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6bd3c1f71f64423f8ac45af746cec134 : SUCCESS in 1h 11m 06s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/77bf7ab4cfc64908b460370012f2563c : SUCCESS in 1h 03m 16s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f133e7a4fbf6476bac18c54e36271e8a : SUCCESS in 54m 59s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8f0b51233e347dd8fe464e2c0ed4fc0 : SUCCESS in 1h 01m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1577e4c7ffd342acb0619c3718fd90d1 : SUCCESS in 24m 03s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/b5a64140640a4d199a1b9c5904745528 : SUCCESS in 22m 26s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/639d92afeae34bdb9cd24d78d53ef663 : SUCCESS in 16m 13s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/143dc34e48074aec9506066dc13fd215 : SUCCESS in 15m 20s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/994501335347484a9f5ed01eddb1a8fb : SUCCESS in 1h 12m 20s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bff11b416fde47cc9702b9553fff24db : SUCCESS in 2h 01m 38s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b110ca7346b34581a552f0f1e39720c0 : SUCCESS in 1h 25m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/48c4e78332cd4db2b4af6bc9b791e1ab : SUCCESS in 21m 38s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/995cbeba99cf41128016d3199d068180 : SUCCESS in 20m 15s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c06bcfb6ee9741d89a1936840a33a1fc : SUCCESS in 16m 37s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1037ce266ca448538aadb333c605c613 : SUCCESS in 16m 39s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bd9eaba604c64e069408abddc7bdf43f : SUCCESS in 40m 56s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/514c6a2a2be1454f8c4614162c56e84a : SUCCESS in 31m 24s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/465d3d74b87e4e6e8ecee76f9619005a : SUCCESS in 18m 01s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/06f190bef59a4044bbb5431cb0fe4f8a : SUCCESS in 16m 01s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d25d9c0a0ce4516872634d1785f7e8f : SUCCESS in 12m 45s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc4cad304eb84bad896c181afc4dce8c : SUCCESS in 12m 22s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/46508e29e3a9402e8349816733de4e54 : SUCCESS in 20m 35s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/d999ccd042ab409990dc48c9221191c8 : SUCCESS in 18m 25s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/760ae2e5999944e98fcdb5438996f17e : SUCCESS in 13m 32s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9799474db7349cca44fbea0c7f19bf1 : SUCCESS in 14m 55s","accounts_in_message":[],"_revision_number":1},{"id":"7f685c13772d6ef0ce7a339db81e732be126aafd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 12:18:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6ce9091f5882b4aadbc9b38a75ec5ee668dcc040","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 12:19:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb135a0a36e6403d9c8357a3f84bcfe3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65c1b9d798064f56801a78adfd986d60 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e020633c496abe670c0940c298ff574d1ff15148","revisions":{"e020633c496abe670c0940c298ff574d1ff15148":{"kind":"REWORK","_number":1,"created":"2023-01-12 11:34:59.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/32/869932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/869932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/869932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/869932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/869932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/869932/1"}}},"commit":{"parents":[{"commit":"56cfe1491b0cdf5b1d3e42aa7b0727d151cc59f2","subject":"Add support for Ubuntu Jammy Jellyfish (22.04) LTS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/56cfe1491b0cdf5b1d3e42aa7b0727d151cc59f2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-01-06 10:11:46.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-01-12 11:34:59.000000000","tz":0},"subject":"Replace references to Docker Hub by Quay.io","message":"Replace references to Docker Hub by Quay.io\n\nKolla Ansible uses Quay.io as its default registry.\n\nChange-Id: Ie1a56ebd6fe3283a2b5b077f658ff1cf3e006784\n(cherry picked from commit 6e78aa89149add7b15c457266d5712486ef4503d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e020633c496abe670c0940c298ff574d1ff15148"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e020633c496abe670c0940c298ff574d1ff15148"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
