)]}'
{"id":"openstack%2Fkayobe~869933","triplet_id":"openstack%2Fkayobe~stable%2Fxena~Ie1a56ebd6fe3283a2b5b077f658ff1cf3e006784","project":"openstack/kayobe","branch":"stable/xena","topic":"docs-quayio-stable/xena","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-01-16 12:18:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie1a56ebd6fe3283a2b5b077f658ff1cf3e006784","subject":"Replace references to Docker Hub by Quay.io","status":"MERGED","created":"2023-01-12 11:35:08.000000000","updated":"2023-01-16 12:20:10.000000000","submitted":"2023-01-16 12:18:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869933-docs-quayio-stable/xena","cherry_pick_of_change":869442,"cherry_pick_of_patch_set":1,"meta_rev_id":"4571f4f670950a979c67ec2ee60468fc9d662078","_number":869933,"virtual_id_number":869933,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-16 12:18:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2023-01-16 10:10:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2023-01-16 10:10:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-12 12:59:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-16 10:10:48.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"eefb1bc161c03ed70493878a8ee7faea0bfb0590","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-12 11:35:08.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"914492d000552f95424ac3da9a3fd0e89f3a4d24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 12:59:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/165c57878c8c4258bc018875a5124ecb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/158fa7e625db4aff99af7aeb12b53885 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1542d0839cc4326b70d52c3bdcfc6af : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de82342e737048aebc7e3f9ee9853f65 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/453e33b259cf4d0a97bd4a3ff6356271 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e30fc8310bb246ccabeaefc325032fad : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66b1924c5419429296bd841d359c6ed3 : SUCCESS in 9m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/743a7fa7c8d545f5a9137e0c68456581 : SUCCESS in 6m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9257d219d43b47f2b82ad95bdd27c45d : SUCCESS in 14m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8050fa39446c40f3ba00a3e6dd6bd140 : SUCCESS in 17m 24s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4e64c37a72f64e5eb86cb1b3b7212e0d : SUCCESS in 1h 07m 27s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/64baae02b6b64ab690e6b30343ddccbd : SUCCESS in 1h 09m 08s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e85bc106b434afbba2e531cc0e0a8b2 : SUCCESS in 1h 04m 07s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/bfcc8d6556a842aa880c315b63ca2cc9 : FAILURE in 47m 30s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7a07148c6ae943dc925f06ca85cfd5fb : SUCCESS in 25m 15s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/04d3f93e26b8495a90b457c61ddfbae6 : SUCCESS in 18m 06s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/274272a1242c4d12a651f064da639440 : SUCCESS in 14m 43s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d0f6d24584d641f89953c4873c272680 : FAILURE in 45m 29s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1a7008410f84798bce6a6bf562920d8 : SUCCESS in 1h 17m 26s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ee3d148c9a034f759fd55cab45b98e97 : SUCCESS in 20m 21s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/5983f3e1331747e7878bb25f7882f0bf : SUCCESS in 17m 56s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71d808eb75af4e7d8b6fb70e8bb199cc : SUCCESS in 16m 45s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/6829168bac564335afc8edff0fec4098 : SUCCESS in 43m 19s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3f2996a286774242b8117738ffeffe4d : SUCCESS in 31m 49s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3e2a6026c624335892f0300306a3eb0 : SUCCESS in 20m 44s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d54e2ea2a49a440287f2c06fffb50dcd : SUCCESS in 17m 34s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f7b636930a1440babecf0ff25ceff09f : SUCCESS in 12m 40s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f08677c79564128919f8d585d02ddb8 : SUCCESS in 12m 48s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/b2b8875aa8df48c69ee2b622b5f6413f : SUCCESS in 17m 36s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/12e0077647cd4c16aed8475090a63d1c : SUCCESS in 19m 25s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b9c0118c2be4674b8e09a7ea5f80967 : SUCCESS in 14m 14s","accounts_in_message":[],"_revision_number":1},{"id":"9926653570d1d637084d09b9d1a5459a352835ad","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-12 13:24:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c0df96cb20d0097dde8aa2d468796cc29eac4b17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 16:39:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bebce9f97d544dc92d11c66f94e00e0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5864100b36c46b283128d3891c11df8 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f30ca49e929f49249da59a84e98c0a61 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/827c560f51cf4959b2c959ab7bcb2853 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adb9ab91be634ee6818c2cca87403e9c : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9a75028c9804b219b117a054efef6cc : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9c1ffc226a44acb9406f078d51624bf : SUCCESS in 10m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f794f95af116490a8757a625733b19a2 : SUCCESS in 5m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7ece2e9860e144cea67575dadcfc325c : SUCCESS in 11m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e75d7edf24264f7bb9650d0b25ffc0ab : SUCCESS in 15m 02s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fecea5de97294eaeac035dbafc41d0f7 : SUCCESS in 1h 17m 10s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/5f442e2cb7ba47aa8a0bd1d287275123 : SUCCESS in 1h 12m 14s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1038589c87424e61b6bfacb1442d34cc : SUCCESS in 57m 28s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7649dc7623864236a8ae3d132e0a5438 : SUCCESS in 1h 11m 46s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/40da312c932042c48d985690a430ecbe : SUCCESS in 24m 01s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/ed5f10f868064cb5bc557e01cf728f73 : SUCCESS in 23m 08s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7423607f35a446a99b0817bdb9ad4fd6 : SUCCESS in 16m 20s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/470baa1cabdc4ed6823b4bac6f6cc5f8 : SUCCESS in 1h 53m 18s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89477454405d4f8d9bdb3a542f8bd0cb : SUCCESS in 1h 16m 06s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/35e62592945f4d71851d82f3f0048650 : SUCCESS in 20m 55s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/3dbd997f3aa7479f90fc4ab96b9d0be2 : SUCCESS in 19m 39s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff01591eb4da484eae689af63b9106bc : SUCCESS in 14m 19s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/a1b9b183f7974f3283f7d43ab3a6272c : SUCCESS in 48m 35s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f2d4ffcbb646405db4c655cab655c964 : SUCCESS in 27m 30s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5705e7b47e349c69fc3d97edb54278c : SUCCESS in 24m 24s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ff10e2c0770f4aa3b91be53279672a28 : SUCCESS in 16m 36s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/5e153f4dfe6b4e36865b46626bb88b52 : SUCCESS in 14m 35s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/059175125def47449b96d76e8c146623 : SUCCESS in 10m 47s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/24ff0f08856441d7b5c33a7f64fd45fc : SUCCESS in 19m 50s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/d597843ef56a48fdae89fbf5de905f35 : SUCCESS in 16m 47s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07148db93bbb466ab3d91034fdd87e26 : SUCCESS in 12m 20s","accounts_in_message":[],"_revision_number":1},{"id":"1fc19a714a5a91b35d32cb2fc5ed4acb24eceaa0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-01-16 10:10:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1639ae51f8515c548bd45cfdfbf5508ac0f39ebc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 10:11:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a21770a0866289405e0af53f41ca2dff1b987745","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 12:18:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff05cb8a4ba842579223f1ceddc62df0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47047860f0574e7386bc66145f749f06 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1daafbf16e6744f39d0200f5cbe474bd : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08f2dfdf442344e0b3be1cf0916205a7 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39535cd84f444656b8ba9283f9a1ae1a : SUCCESS in 10m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/da6b6008f0284955bdb5406a2efc0c65 : SUCCESS in 5m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/648eac87ec72425dab2968d500fa31b1 : SUCCESS in 12m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/314fd1d825b04b73995387b099b36b03 : SUCCESS in 16m 48s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3382b182bc9c41439fb1c9925cae0644 : SUCCESS in 1h 04m 23s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/49cfcef25d0f4acda9700639e4dbc6a8 : SUCCESS in 1h 06m 27s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1b195c790ae4e068f6dc4e047ba46b7 : SUCCESS in 59m 54s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5fc48a43de04407db70f6fa83679594d : SUCCESS in 1h 10m 03s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c52197e0b7284a958f6df173b648327f : SUCCESS in 26m 22s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/e43a79974766457e9d6a5792eebb8bb7 : SUCCESS in 20m 52s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d41013b93e0444ea4bd0a2931a2dabb : SUCCESS in 13m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/afc3bdc9e6bf407cbe051cd6ac0d1b06 : SUCCESS in 1h 50m 41s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bf24938fb7e4acda58b58a53e69d675 : SUCCESS in 1h 10m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/62212a7e8972419e8b42ea8647a41ba4 : SUCCESS in 19m 31s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/eb174bd66bac4a2a915928fb6a43823a : SUCCESS in 20m 33s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40e8e5a2658547c0b45c46fc5329685d : SUCCESS in 14m 56s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7978c6cc60284cb0a39e3a5aa72c06ad : SUCCESS in 32m 00s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0fb0bdf71a142788c62d624f7f2c0d7 : SUCCESS in 20m 43s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/07c332e8d76f485baf4f2203cc0256e5 : SUCCESS in 17m 30s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/7765138df95b4441b8652c485ed4fc8a : SUCCESS in 14m 54s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ebc0c18f3b8444f9ae7c99d0b89e3b4 : SUCCESS in 11m 24s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/608c8abd69bd444bb6ef532827de7001 : SUCCESS in 18m 09s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/73c97fc59fcc41a894f13443b2a4acd0 : SUCCESS in 17m 22s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13a52f994ded44ed87e5b20c0d09d543 : SUCCESS in 12m 13s","accounts_in_message":[],"_revision_number":1},{"id":"a797f455ad2cfe7ff84d66bc84b96f64fb911d2c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 12:18:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4571f4f670950a979c67ec2ee60468fc9d662078","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 12:20:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4900debd9759468094a9b1d7b41cf84b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/241485400e9a4477a70c1e01d067a829 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"931c5f594c1cd3336fe083f0a1604b73b9eb0e92","revisions":{"931c5f594c1cd3336fe083f0a1604b73b9eb0e92":{"kind":"REWORK","_number":1,"created":"2023-01-12 11:35:08.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/33/869933/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/869933/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/869933/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/869933/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/869933/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/869933/1"}}},"commit":{"parents":[{"commit":"e9a420160afd5fd41fe753691f03f6bf2baee300","subject":"Fix tox.ini for tox4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9a420160afd5fd41fe753691f03f6bf2baee300"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-01-06 10:11:46.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-01-12 11:35:08.000000000","tz":0},"subject":"Replace references to Docker Hub by Quay.io","message":"Replace references to Docker Hub by Quay.io\n\nKolla Ansible uses Quay.io as its default registry.\n\nChange-Id: Ie1a56ebd6fe3283a2b5b077f658ff1cf3e006784\n(cherry picked from commit 6e78aa89149add7b15c457266d5712486ef4503d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/931c5f594c1cd3336fe083f0a1604b73b9eb0e92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/931c5f594c1cd3336fe083f0a1604b73b9eb0e92"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
