)]}'
{"id":"openstack%2Fkayobe~869963","triplet_id":"openstack%2Fkayobe~master~Id9c4abc285be0240f9bdc21d21537d3a2d9bfd57","project":"openstack/kayobe","branch":"master","topic":"assert","attention_set":{},"removed_from_attention_set":{"25600":{"account":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"last_update":"2023-02-21 17:05:24.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-02-21 17:05:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id9c4abc285be0240f9bdc21d21537d3a2d9bfd57","subject":"Use assert to test/validate condition","status":"MERGED","created":"2023-01-12 11:25:48.000000000","updated":"2023-02-21 17:06:25.000000000","submitted":"2023-02-21 17:05:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869963-assert","meta_rev_id":"0abe542e2e6f22d921cedc4016657b5fd9be508d","_number":869963,"virtual_id_number":869963,"owner":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-21 17:05:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":1,"date":"2023-02-17 06:42:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":2,"date":"2023-02-17 08:01:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-02-21 14:26:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-02-21 14:26:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-12 12:16:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-13 07:56:55.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-02-16 17:33:40.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2023-02-21 14:26:54.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"720efb132bd290136e3056158c2b50b9b56f77b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-01-12 11:25:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c6072711fdab8207bafe7f1f688502f79686cfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 12:16:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d6b65f380eb4600a43a3aa9db5565fe\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a333d6de1be4eda95eb07dc6cf2668f : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/363a2528f97b4fd2a588fa084c137fe7 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6dbfe72b26f4071a801b8e14f284274 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fdc8954edc744ffabf188c5a64357b4d : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ce2e0c694c0450f8c0010b5c487e679 : SUCCESS in 7m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0a5386bf24574499a3a315d052b71c61 : FAILURE in 4m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/31451d7936d04df09fb3136816114507 : SUCCESS in 11m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d351fb31868d4bcfb2e59ad7727b3d9f : SUCCESS in 14m 53s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1fc309fa2de14e6f80bec14faa37fec5 : FAILURE in 9m 46s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/756d754a1bf54deea590d767778d8838 : FAILURE in 10m 49s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a06e761fa56b45aa80fee4023b4499d5 : FAILURE in 8m 48s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0823cc98972a438e80f3b45a55ee932a : FAILURE in 10m 08s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/190d366625f1482ab45d1929b7ddf540 : FAILURE in 10m 22s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/99204eaac69b48bfbfad50dc53951fe8 : FAILURE in 10m 54s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bca1a4725a6a409a88e7e8b27513e65a : FAILURE in 10m 08s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0586d08eeae94467b93db9180cfa490b : FAILURE in 46m 51s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/25c969a325fc4ffeac0ff2b0f42e39fe : FAILURE in 10m 58s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f4a8da9fae744e68b966349b01c2795 : FAILURE in 8m 41s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/0a97c868c2dc4c52b178d0765aaf784c : FAILURE in 13m 29s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d2d04ca585a417a9d24f5a645e4333a : FAILURE in 25m 59s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8bee6d14ae674ea9ba80187a5013e8f4 : FAILURE in 10m 08s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99ce93c7bdef4ad2a63ba65d260a6ef3 : FAILURE in 7m 24s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/479dd35fa9d14ae7aa1a8dd195e04de6 : FAILURE in 8m 49s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d52e9c0096e4b268c8bd59603ffd75b : FAILURE in 7m 16s","accounts_in_message":[],"_revision_number":1},{"id":"c2339aa4bd68498db3bfb6948e087778a74fd3f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-01-12 12:48:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"79754955c9c2c3a8d72b3e37f16e32f6bf74aeb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 13:58:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c885aedb63d4c1f8f9879b377e0dc9f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9aee36a9e69b43a9ab45c81b6d16bd23 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1d01d5e406d42eb88965b4fc771988a : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb49fc6ff60f4abda6a1ed015c58fd49 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cda85d8c54d44f7fa7722d4bd06129d7 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/712a16c02d2641598df0b252a440fc74 : SUCCESS in 6m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c0893075ec7c4b3f9db140244f2b8c54 : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/833d91d37eef4b42917acd713da2f9e6 : SUCCESS in 11m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d61be2b0081c49bcbfa9385f9d5ae011 : SUCCESS in 13m 42s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9ba7d4070e994c92803b8a68cb88c86e : SUCCESS in 57m 36s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/4b85bd478fd84b6bbcc2be3d7a50684c : FAILURE in 35m 13s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04ce18c3e40d429b84db6b1b7552f628 : SUCCESS in 45m 31s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/23ce3f9f704c4884b30180997612aebb : SUCCESS in 1h 03m 32s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e7430b9766eb40dca0370d9ad9df4aa9 : SUCCESS in 16m 38s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/c889be56769943538182dca9cd831f06 : SUCCESS in 17m 32s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/266521f7507b444d8e1e8ee87a7a83a2 : SUCCESS in 14m 24s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29304f4ac4024fef94bbdab87a50ffce : SUCCESS in 1h 05m 53s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9e6dd15fd8ab440692210cc8dc7b86a1 : SUCCESS in 15m 36s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18d28a30e2794c29972686edb350dd62 : SUCCESS in 15m 07s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/78e8b39d11544252b5f37a168e5b6fc3 : FAILURE in 28m 18s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2c90c132cce4f27ae9482085490415c : SUCCESS in 32m 48s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b53507b25039497b895552757c419300 : SUCCESS in 12m 05s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7112eeff9c8b4969b3d4df9a115e7d05 : SUCCESS in 9m 55s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/da79e41490704a52852b075585b829f1 : SUCCESS in 15m 24s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a19119b532a432bab03ece7320a686a : SUCCESS in 13m 28s","accounts_in_message":[],"_revision_number":2},{"id":"1bb4f55e8654f48f2bfcadc30b87a68545b36692","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-01-12 16:23:44.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f65503d0efe4c2c9cc32b0df706585b14b849385","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-13 07:56:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"76ad18319c94e929941c4824e5a07d98aa15c0d5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-16 16:03:09.000000000","message":"Topic assert removed","accounts_in_message":[],"_revision_number":2},{"id":"8541cf01e5631cf33e9f38cdf8e6381238bc71d2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-16 16:03:14.000000000","message":"Topic set to kolla_assert","accounts_in_message":[],"_revision_number":2},{"id":"c04ec2b774afbc67764fc845cb7c4a9f2dfad142","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-02-16 17:33:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"65aa2c1bbd111a2c529fc5bbbe9e0edea948d131","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-02-17 06:40:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3d850bb60cf7bbe09d7fb3683e4bcaf7d354abd2","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-02-17 06:42:20.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"01e207a634e5082199bc368cd344f122c62fac77","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-17 08:01:13.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6e091c3795f3c85e82c6665293435502245bbfc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 08:06:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ade76129e36d4dd89220590342cf4730\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbe5f796f8734da18629f9a99eb1e007 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04d7d58eb25149f19021305e3875dd7f : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cc1e48d4ef143dab39711099e8c5216 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6b249542de94642bb20e216483b46d6 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d90631a1d60a466295302b5eac45c1dc : SUCCESS in 6m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ea48a7c85f894a0d837a44dae431b299 : SUCCESS in 3m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b9cff48972644f7e804c345f46e15543 : SUCCESS in 9m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/468e6ae054db458386fdd0629971b280 : SUCCESS in 14m 33s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/eecbef3be19f40f7b02b5a47b6e619c3 : SUCCESS in 58m 59s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/ae89ffd784794d8497047dc6b11d733a : SUCCESS in 1h 00m 36s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5c02b2c03e14c4fba275810c8218b2c : SUCCESS in 49m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c896d880600a4ac68da4833953d96f15 : SUCCESS in 49m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b6be3cb869194bea8672084d45439a49 : SUCCESS in 17m 45s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/bc77bf949e874f068e7f1cdd9e273eab : SUCCESS in 17m 59s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf64af7bc5424911b90d0d3451db0330 : SUCCESS in 15m 06s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e854b3c5ec4b4448aae49faafdc11c77 : SUCCESS in 1h 17m 54s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a50a75381d4840dca655194ddfec1213 : SUCCESS in 1h 14m 20s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/bf469ec0493542c9a590dc3a4b181fbc : SUCCESS in 19m 58s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aefe5b46a99144c5aa677a2d1bba47a4 : SUCCESS in 18m 11s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1da17af9f3704a41b1230dd4d54ec439 : SUCCESS in 26m 41s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8997402784d543f5a7ec128bb7f0a3a1 : SUCCESS in 30m 49s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00c650ce1d1647c2b44890bd64acc1f1 : SUCCESS in 32m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/312e3160db7f441899dd43d0f6d27533 : SUCCESS in 12m 59s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5beedfd4d9564ed98c2b79891d607730 : SUCCESS in 10m 54s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/455f9c2ac2b0470fba34e0d02a9a67f0 : SUCCESS in 16m 38s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9df3b34ff3c40fa8fe59f642f5cc40e : SUCCESS in 24m 29s","accounts_in_message":[],"_revision_number":3},{"id":"f1dd8bfac50bd6122f948e53673960a978596af3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-21 14:26:54.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c84e9dd83afcfd6b9b1cecfce3d565509627ba60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 14:27:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"41c2e3a06f02150fc00a8c8c6ae9ac766393bfaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 17:05:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2929d75401e4481184bc5235c15e89a0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d2235f40e4a4fc7921b6fa806ab23c8 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbf6cd6533424f04b962efd990db6e62 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7364cfcfe34a42f5b0d36bb21f17e47a : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d43c1b8f0b5643c582d5d0336d3f2d01 : SUCCESS in 6m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0685c7562efd4e5391a52772ed9ff206 : SUCCESS in 6m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c7d1de030f084732b8fb8a985f85e2d9 : SUCCESS in 11m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e5910f6452c34840b23b7308f325de59 : SUCCESS in 14m 03s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d7d8e7132daa4fbeb87f969d0e7c685c : SUCCESS in 49m 14s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1bbd4aefed34e90bc2b5a51aec9b020 : SUCCESS in 44m 15s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/454b48c5e1914b01a16934935824a303 : SUCCESS in 58m 01s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/75c2682b38fb45c59df301cd8c34ac3d : SUCCESS in 18m 14s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b73b13d289eb4fcc9b5a9e8e3224b33b : SUCCESS in 14m 05s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/83639bffc85e44a5a8aa481ae546770e : SUCCESS in 1h 55m 43s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f319921dc98e4cd5841ab27ce8389597 : SUCCESS in 1h 08m 26s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/98f0ddff2c9146b8a40bff463537c639 : SUCCESS in 18m 11s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f73a00cd0ac4741b3d24c0ab5fd9f7b : SUCCESS in 15m 29s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9f467bc81db041b7b245db29061fd625 : SUCCESS in 30m 38s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99c67103cc2e472fb9767e8f5182aa7c : SUCCESS in 26m 42s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1b96eadcee074d5384dcd182148fb499 : SUCCESS in 14m 57s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93cd988aab8c4de5bc1dc44077df3b29 : SUCCESS in 11m 59s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9a063a7980d14d8fb1683927ed27ce18 : SUCCESS in 16m 48s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9bb78b092a654cc385f77ef3d5904ba4 : SUCCESS in 12m 04s","accounts_in_message":[],"_revision_number":3},{"id":"f491deedf096599cf6fee0f08a55b4db0642cc06","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 17:05:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0abe542e2e6f22d921cedc4016657b5fd9be508d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 17:06:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03de602b065e48199a17d9de96ebdcb9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/489286dbcb5848de82b4cc0211336e3a : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e6bd07faf78fb0895504817b398008a339e0d368","revisions":{"8d8666d310b6f6fece912e2069ab115e162b1440":{"kind":"REWORK","_number":1,"created":"2023-01-12 11:25:48.000000000","uploader":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"ref":"refs/changes/63/869963/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/869963/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/869963/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/869963/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/869963/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/869963/1"}}},"commit":{"parents":[{"commit":"ebd5799fd4b77a138cf7790a8094eb859e8898f3","subject":"Merge \"Replace references to Docker Hub by Quay.io\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ebd5799fd4b77a138cf7790a8094eb859e8898f3"}]}],"author":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2023-01-12 11:18:28.000000000","tz":60},"committer":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2023-01-12 11:25:03.000000000","tz":60},"subject":"Use assert to test/validate condition","message":"Use assert to test/validate condition\n\nThis test seems to have a fault, where the variable/list\nkolla_ironic_enabled_console_interfaces could be undefined, the\ntest would not fail.\n\nAnyway, using assert shows what we\u0027re testing, and doesn\u0027t list the\ntask as skipped when we meet the requirements.\n\nChange-Id: Id9c4abc285be0240f9bdc21d21537d3a2d9bfd57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8d8666d310b6f6fece912e2069ab115e162b1440"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8d8666d310b6f6fece912e2069ab115e162b1440"}]},"branch":"refs/heads/master"},"31b53178add5e497a4fce833427d818a2896de79":{"kind":"REWORK","_number":2,"created":"2023-01-12 12:48:16.000000000","uploader":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"ref":"refs/changes/63/869963/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/869963/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/869963/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/869963/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/869963/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/869963/2"}}},"commit":{"parents":[{"commit":"ebd5799fd4b77a138cf7790a8094eb859e8898f3","subject":"Merge \"Replace references to Docker Hub by Quay.io\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ebd5799fd4b77a138cf7790a8094eb859e8898f3"}]}],"author":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2023-01-12 11:18:28.000000000","tz":60},"committer":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2023-01-12 12:45:40.000000000","tz":60},"subject":"Use assert to test/validate condition","message":"Use assert to test/validate condition\n\nThis test seems to have a fault, where the variable/list\nkolla_ironic_enabled_console_interfaces could be undefined, the\ntest would not fail.\n\nAnyway, using assert shows what we\u0027re testing, and doesn\u0027t list the\ntask as skipped when we meet the requirements.\n\nChange-Id: Id9c4abc285be0240f9bdc21d21537d3a2d9bfd57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/31b53178add5e497a4fce833427d818a2896de79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/31b53178add5e497a4fce833427d818a2896de79"}]},"branch":"refs/heads/master"},"e6bd07faf78fb0895504817b398008a339e0d368":{"kind":"REWORK","_number":3,"created":"2023-02-17 06:40:20.000000000","uploader":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"ref":"refs/changes/63/869963/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/869963/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/869963/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/869963/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/869963/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/869963/3"}}},"commit":{"parents":[{"commit":"be0a819623062703fdc266d844d21937edd5f379","subject":"Merge \"Ubuntu: Fix networkd config for multiple VLANs without MTU\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be0a819623062703fdc266d844d21937edd5f379"}]}],"author":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2023-01-12 11:18:28.000000000","tz":60},"committer":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2023-02-17 06:40:17.000000000","tz":60},"subject":"Use assert to test/validate condition","message":"Use assert to test/validate condition\n\nThis test seems to have a fault, where the variable/list\nkolla_ironic_enabled_console_interfaces could be undefined, the\ntest would not fail.\n\nAnyway, using assert shows what we\u0027re testing, and doesn\u0027t list the\ntask as skipped when we meet the requirements.\n\nChange-Id: Id9c4abc285be0240f9bdc21d21537d3a2d9bfd57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6bd07faf78fb0895504817b398008a339e0d368"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6bd07faf78fb0895504817b398008a339e0d368"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
