)]}'
{"id":"openstack%2Fkayobe~869965","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I920385e9a939a97ef30e409975c12fe44c6061f1","project":"openstack/kayobe","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-01-12 14:27:53.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-01-12 14:27:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I920385e9a939a97ef30e409975c12fe44c6061f1","subject":"CI: Readd centos8s/rocky8 jobs after jammy backport","status":"MERGED","created":"2023-01-12 11:42:23.000000000","updated":"2023-01-12 14:29:55.000000000","submitted":"2023-01-12 14:27:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869965","meta_rev_id":"380063a94bc2733e330791dc2345953565bb8a91","_number":869965,"virtual_id_number":869965,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-12 14:27:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":2,"date":"2023-01-12 14:14:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-12 12:17:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":1,"date":"2023-01-12 14:14:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-12 11:43:15.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-01-12 12:06:42.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2023-01-12 12:17:12.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-01-12 14:12:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"08a7d620ea40e8e74f65d0799dbac07e9e773023","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-12 11:42:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c723daadfb7c2f29a50e08d2b1840c601189582","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-12 11:43:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dda46fd68b7403bca4bcfe92f96820a456056dfd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-12 12:06:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c1d7276cb77a490633fedb912dcfd2c604c27f6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-12 12:09:03.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8da4b7893537843f8a39092db9b3388ae7a459a6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-12 12:09:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3383f41827e2e3c826cf9a92a53f8524b79a6e9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-12 12:16:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d7ff31c3bfcc5ab73b4ee261644d8f59a0ff76a8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-12 12:17:12.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"042e35ba22c2dd782f1c129ee88db6536771dd5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 14:12:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17ea7732c18a454382d26c5eb0997e84\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5965588f4a234fef87c4ad5fd9d7dd9a : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e671dc11ee2443987526c91ce1940d0 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4639621b479e4b88ac0702c4fe219de1 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8fd7db25e6649ef88b7989a2fbeabbf : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b03d84383c84bd0836735325ae0b542 : SUCCESS in 8m 54s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/bacce309a562469c8216cae8fa7fa736 : SUCCESS in 1h 04m 30s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/e585db103b99482b97716c0a7370151c : SUCCESS in 1h 05m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2113014f18f94d27b6971b03f5949bf0 : SUCCESS in 22m 59s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/67f886c5605947d3a03f295eba575c88 : SUCCESS in 21m 27s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/476ee9b4e5674a97b42a2e4d8e6fa59c : SUCCESS in 1h 04m 16s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0aa537fc563048f0bfca98d223f7c8cd : SUCCESS in 1h 49m 33s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5706d2d7e19d4444bd1fc6cb486e1e22 : SUCCESS in 17m 11s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/0a290551dbd94c05b7dc1f88b7ff8580 : SUCCESS in 18m 26s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c91cac725aea43a298fbb0989b630cca : SUCCESS in 33m 44s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31ae066ade08450c85856c7f9c9f1ce6 : SUCCESS in 23m 56s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/dd3aab26fdbe4f17b5eeb99280332f49 : SUCCESS in 14m 24s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/7fe153b7c52d41aca9a16d340e4c3020 : SUCCESS in 15m 08s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/34d8f999b0884f16ad1da7d8c36f90c7 : SUCCESS in 19m 43s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2cebfcc54d2e40d7910da750c6cc6b0a : SUCCESS in 17m 18s","accounts_in_message":[],"_revision_number":3},{"id":"b48452cb8ac61239d4e959f4abca5819865384ca","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-12 14:14:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b2756a8baa8632021e6b2d85f1d70e476588b1df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 14:14:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d3d4b2b434f2445858ce3117fd884fd77ce85e60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 14:27:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad06a09925aa438db1ef9d15a980c606\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ad837e5f4294228819e9ce6a09e8e5c : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1edb5a69edd74129af3955e3233a50ad : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0249911c8f1f4d5da8349f8af0938450 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/046e599834ae4857b22951e9d3cd4805 : SUCCESS in 9m 03s","accounts_in_message":[],"_revision_number":3},{"id":"e48d1c4f0169c8c8e718427e244e79787544fc19","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 14:27:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"380063a94bc2733e330791dc2345953565bb8a91","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 14:29:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e437720a03bd4bda83648cc84ae24b39\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e34ed7ef34e649bf9e7efddfc0524f47 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bb6c1a2a2348dbe92a00c5004af485771c8eddba","revisions":{"23949754b5246fc34fb5f9d3ae1bacd692a9cb45":{"kind":"REWORK","_number":1,"created":"2023-01-12 11:42:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/65/869965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/869965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/869965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/869965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/869965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/869965/1"}}},"commit":{"parents":[{"commit":"56cfe1491b0cdf5b1d3e42aa7b0727d151cc59f2","subject":"Add support for Ubuntu Jammy Jellyfish (22.04) LTS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/56cfe1491b0cdf5b1d3e42aa7b0727d151cc59f2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-12 11:41:40.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-12 11:42:21.000000000","tz":60},"subject":"CI: Readd centos8s/rocky8 jobs after jammy backport","message":"CI: Readd centos8s/rocky8 jobs after jammy backport\n\nThese have been removed by mistake.\n\nChange-Id: I920385e9a939a97ef30e409975c12fe44c6061f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/23949754b5246fc34fb5f9d3ae1bacd692a9cb45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/23949754b5246fc34fb5f9d3ae1bacd692a9cb45"}]},"branch":"refs/heads/stable/yoga"},"fc0daa10e3cb0dd0477de59c97b5b97e7cacd020":{"kind":"REWORK","_number":2,"created":"2023-01-12 12:09:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/65/869965/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/869965/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/869965/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/869965/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/869965/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/869965/2"}}},"commit":{"parents":[{"commit":"56cfe1491b0cdf5b1d3e42aa7b0727d151cc59f2","subject":"Add support for Ubuntu Jammy Jellyfish (22.04) LTS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/56cfe1491b0cdf5b1d3e42aa7b0727d151cc59f2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-12 11:41:40.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-12 12:09:03.000000000","tz":0},"subject":"CI: Readd centos8s/rocky8 jobs after jammy backport","message":"CI: Readd centos8s/rocky8 jobs after jammy backport\n\nThese have been removed by mistake.\n\nChange-Id: I920385e9a939a97ef30e409975c12fe44c6061f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc0daa10e3cb0dd0477de59c97b5b97e7cacd020"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc0daa10e3cb0dd0477de59c97b5b97e7cacd020"}]},"branch":"refs/heads/stable/yoga"},"bb6c1a2a2348dbe92a00c5004af485771c8eddba":{"kind":"REWORK","_number":3,"created":"2023-01-12 12:16:19.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/65/869965/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/869965/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/869965/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/869965/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/869965/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/869965/3"}}},"commit":{"parents":[{"commit":"56cfe1491b0cdf5b1d3e42aa7b0727d151cc59f2","subject":"Add support for Ubuntu Jammy Jellyfish (22.04) LTS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/56cfe1491b0cdf5b1d3e42aa7b0727d151cc59f2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-12 11:41:40.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-01-12 12:15:28.000000000","tz":60},"subject":"CI: Readd centos8s/rocky8 jobs after jammy backport","message":"CI: Readd centos8s/rocky8 jobs after jammy backport\n\nThese have been removed by mistake.\n\nChange-Id: I920385e9a939a97ef30e409975c12fe44c6061f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb6c1a2a2348dbe92a00c5004af485771c8eddba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb6c1a2a2348dbe92a00c5004af485771c8eddba"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
