)]}'
{"id":"openstack%2Fkayobe~869977","triplet_id":"openstack%2Fkayobe~master~I9cf265ffc7acce926ce093d1e4f8d53a12be1f26","project":"openstack/kayobe","branch":"master","topic":"networkmanager-test","attention_set":{},"removed_from_attention_set":{"25600":{"account":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"last_update":"2024-11-26 07:58:55.000000000","reason":"Change was abandoned"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-11-26 07:58:55.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I9cf265ffc7acce926ce093d1e4f8d53a12be1f26","subject":"[DNM] ansible-role-interfaces NetworkManager test","status":"ABANDONED","created":"2023-01-12 13:40:29.000000000","updated":"2024-11-26 07:58:55.000000000","total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"67969e2a8df2fc154073b8db8a467228a4e04556","_number":869977,"virtual_id_number":869977,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-04-13 10:11:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-12 13:40:52.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"CC"},{"updated":"2023-01-12 14:54:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f2fac143b823ebf9ff890165f47d2f6860df075d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-12 13:40:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15269006711df6dc5b7d083fb74ed29bcbfc942c","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-12 13:43:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"39aadf43f62fa69aa6709d7294e9e2c1f5addf2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 14:54:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19aadd1583d04b9cadd70b300d760ab5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6412fc6adc4e4d28addda72c2c12d51b : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e33e0a8dc194ed7a3b90e21ef4146da : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bd46fe8584a498187e6f3df5c13ba18 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d9c6f94d68a43868a87cc11640310a1 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efb44335473e4613b8751e90e3eeacde : SUCCESS in 7m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/36cc5c60c8d74e469180e3bb2d17539b : SUCCESS in 4m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8e23bbc4819d449bbd7c283fc91a1648 : SUCCESS in 10m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/826b4af6732840b49cd250d2fbe75d4a : SUCCESS in 13m 03s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/aedea7e719f64255ade9f242d7c4e675 : SUCCESS in 54m 49s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/65be7e779a65464994ca3c0184a76903 : SUCCESS in 56m 52s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8311b0a946748e3a5c21947b938ad85 : SUCCESS in 49m 12s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/34e5625489f54ab78774ac94342e7f33 : SUCCESS in 58m 07s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/55dbb0a7b3a645fcb46abadc60f41a4f : SUCCESS in 16m 05s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/dc91711899b845ed95f3f6893fcb2b81 : SUCCESS in 16m 27s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/236e1275f9564470af91db550b6c7ad6 : SUCCESS in 16m 52s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73340a5d8a234694a8197e2488cc0519 : SUCCESS in 1h 09m 26s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7ed71f866a014f3692b3fb647e12b420 : SUCCESS in 19m 53s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ec55cd7fc964b22869ab947a2bd60bd : SUCCESS in 14m 14s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a91a01e7fa09489d81e9e55d261685f0 : FAILURE in 27m 42s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1029ab92093d45c69c0ea91c26b6bbbb : SUCCESS in 27m 52s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/80a297cfa5f34756bff9875879020e93 : SUCCESS in 15m 16s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c86e4c2b4d1451c8e745d779f6f044e : SUCCESS in 9m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/76066042590a48008a8cca41c9790ad8 : SUCCESS in 16m 40s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e714f9938d24faab75f59cec7b541eb : SUCCESS in 18m 37s","accounts_in_message":[],"_revision_number":1},{"id":"8a9b3ad138fb9f64fcf961854df5c37f16ddfe47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-01-16 22:45:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4ad9b1d29d16402b05bdf59bc16cede6a943b041","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 00:07:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a9ee3d1c214940b0a157102305fbd0f8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1702ec0895934eb7a511080906f1f245 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a98cef3c1ec495f86c83c19bddec197 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8101a3ccb954b7f9469c447e8f6be00 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/175612822c8e46f5bbdf68f5ca3fffb9 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/203a83f60dc84222b8c2f3f5f0549541 : SUCCESS in 6m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/451a1ba7da574f67a2ff94ac414c5cfc : SUCCESS in 3m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/514d90e721144c0a8380cc56bee24919 : SUCCESS in 8m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ae834ecebbea4efc9a8d5156ca8ed6da : SUCCESS in 10m 48s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/045a83e935b2447387a0bf97898a4c60 : SUCCESS in 59m 46s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/7254964681a545fd8bd73b93e515dcd7 : SUCCESS in 1h 04m 17s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3907d80f29047eb8d28bd2d1384897e : SUCCESS in 55m 04s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f337c7f351db49babe5b1781ee6e8722 : SUCCESS in 1h 00m 18s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1372531384cf4ebd911598e6847f43c3 : FAILURE in 10m 08s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ca040220b1f9444f8a3ee7633c5bdcad : FAILURE in 9m 33s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42d5e3fa62e047fca72098007ba45be7 : SUCCESS in 11m 58s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34a6f67f35fc4d7097d1362781d241e4 : SUCCESS in 1h 06m 38s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/77b8b713840a42768a3eb0c9b6a0ae43 : SUCCESS in 19m 23s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82eb7c0e26124338bcd50f3f5adfa6c7 : SUCCESS in 12m 14s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/2188175d206242a1b1d3df4936304a80 : FAILURE in 28m 17s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fc968599ffa4d34ab478dc8b10540a3 : SUCCESS in 24m 49s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/03a92be4c23048c89057bc03f1545a6b : SUCCESS in 12m 24s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46c9310ebd164b5eb745c85ffe496812 : SUCCESS in 9m 06s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/184471c4d87e4966b35e4c35b113a6f4 : SUCCESS in 16m 44s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c78eff129fa24271bac2de7526de6d0f : SUCCESS in 14m 21s","accounts_in_message":[],"_revision_number":2},{"id":"48deb2955c9211d5b258f52a6cb5194d45cf2cfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-01-17 12:52:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"85dd57356b13a36c209635c47e13f289e3a4e8a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 14:08:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/219cd4c6933d462cae8fa8c500b25d51\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/152f8311468642d7bfd99ab7307f81d1 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbf7abe84c454094bb96c836547ade7b : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd0b98e4e02447f7b1b69e639ba401fd : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/257a3a8e71184e7491e9c84ea980967f : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/241f27e1e5854d6e96f8c03dd526f3ea : SUCCESS in 7m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4bc336ab20f14730b202a96c3d0b84f2 : SUCCESS in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e144b1ad307f4a4984630db320209f07 : SUCCESS in 13m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d813960e6b8b4e86a245ad8387cdd848 : SUCCESS in 15m 33s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/735ba9c71f3641ebaee9769b6c1147ef : FAILURE in 33m 28s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f1b3b06fac2043bd908b56b886f3f190 : FAILURE in 34m 39s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ece2ab3dd0d7441686922ea51869e0b8 : FAILURE in 21m 22s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1d825be828c0424cba168f8f6fc8c205 : FAILURE in 35m 34s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/97756e2a523648f2a4f81c6962b17f0c : FAILURE in 12m 04s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/3c68b839159c480f83f4ea79532dca57 : FAILURE in 12m 50s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06978832cf334e77b2276fc1ed1829f1 : SUCCESS in 11m 34s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d2ed985bece404daeff3e84cc57df22 : SUCCESS in 1h 12m 32s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d25ea8aed9d344069b4c8a1f5686d23e : SUCCESS in 14m 36s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e0087b4f3554cf08d73b4244c9c0e94 : SUCCESS in 17m 41s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c51f87321c804899ab9989f3212afb20 : FAILURE in 23m 10s (non-voting)\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4568723ad18d45999958cbfe1b068130 : SUCCESS in 27m 16s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/977a553dbcfe462988cdfc915f19a915 : SUCCESS in 15m 18s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c0451caa2cb4616ac86222613e46cd1 : SUCCESS in 11m 57s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/27d3dcc4610045c8ba609fe7660ed7a2 : SUCCESS in 17m 36s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59c4db0a203146cbacf36379bbf8476b : SUCCESS in 15m 04s","accounts_in_message":[],"_revision_number":3},{"id":"e6a41f8f49aa872e3b94b33492e4c475f364ceaf","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-18 14:14:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d575de6ede01b16df171fb861c1b03b51846bb9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-01-18 14:45:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2a05f54bd7186ac93f964be9a27ab546238ea901","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 17:23:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/76e3a5f8f3154de499203550a8e78f77\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/210f8cac1a2342a0882606f70ac0f3bf : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b38ccb68dbd4a7e8e168c49f125ec64 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e86feced080432990282bacace13dc9 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec664bf287014dfda3ade6dca66865aa : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20cc145ec5be491d83a7b25b1377ac4c : SUCCESS in 6m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b14cfbc5852c4fb1bf8faffd8d9e212d : SUCCESS in 5m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3a923b31cf0e457da25905055005bdfd : SUCCESS in 10m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c331a818e1414935acd74f1cc2255cde : SUCCESS in 16m 18s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4864e8059e2344b9b4d9607b45f8b401 : FAILURE in 9m 07s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/554b956910a34ea5b10137f30b61e6dd : TIMED_OUT in 2h 19m 58s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f09f11126ba744a28b58f23d5e6aa5c8 : SUCCESS in 58m 06s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5d9f07b8ac89467dbb864ddfcfa48ae2 : SUCCESS in 59m 37s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c8049911560a4a90ba5afeb159cbb472 : FAILURE in 11m 48s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/0631773581864f4a9e42ad1fceb57ad9 : FAILURE in 13m 33s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b09af210236b444eb0e6f93ca964b78b : SUCCESS in 16m 22s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fcda4354ede446a986ed2ec7d6d32aaf : SUCCESS in 1h 14m 13s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4cc0376447ae468d8b15900a948d1ae6 : SUCCESS in 17m 20s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8b88c6aa10a4696b41d532da7028118 : SUCCESS in 16m 45s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/3ead5b09550843aabd12af5a33a99f05 : FAILURE in 29m 38s (non-voting)\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18d41316dee94828b8b0fd05f0eeb19c : SUCCESS in 29m 51s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6aa94f79de9740f0ad31acace2b86d5f : SUCCESS in 13m 43s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd78282bf225403da1a0737541dcf914 : SUCCESS in 8m 42s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c435513496c045949e3edbde0b0b8c36 : SUCCESS in 17m 55s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1aa16fba884463c93b0fcbbd8658f2b : SUCCESS in 13m 27s","accounts_in_message":[],"_revision_number":4},{"id":"398b1ea99ef9cfb1e86fa61e4637a3c8e6057a65","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-01-18 17:30:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"aa8e31f0018d9c9df09deda8d14758ed19f7eb97","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-01-18 18:42:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9ff19ec58a681f74ac983e526d235cb8d331e2b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 19:01:09.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a236a9ffad834969b500f0d87ec38a2e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ba7a93cc13c4ad68c8ccedcc74886f1 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d7a2f1eff994e93b8ac855b61a129b0 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6988f00b9aff40f29945355647cd463c : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d76c0759854477e8ba81cd2b5697bd0 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf81c3304ac042a08c7fed242985bd36 : SUCCESS in 9m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f443113c183f46dcb3ddf06ec184d3bf : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/edc8df7c91a54a91952f8fd213430784 : SUCCESS in 10m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/69c618de991343b48765cd0969a300fb : SUCCESS in 13m 13s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d43671284c5a4e429b9d688b5f2d8055 : SUCCESS in 55m 34s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/231e21328f564fd2a9befbede8c70f56 : SUCCESS in 58m 43s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f890cd82fbe8421ea0d6d4e157586fa6 : SUCCESS in 49m 18s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/81deb2d98e5449fe96fb6078e460a2f9 : FAILURE in 54m 32s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7e41212cbe7d4d29be23e7fb237cbc17 : FAILURE in 10m 26s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/4416193c2e5f44e8aa1bd300b831cd61 : FAILURE in 12m 43s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1a3570c43af49d7a0cccb1e0d74647c : SUCCESS in 19m 06s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b68e784edd634d0aa72464b47f188b5d : SUCCESS in 1h 27m 41s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/58892acbdfc04de8bbf75bb2314229a3 : SUCCESS in 18m 27s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b95a634c06fe4da789fb15310b1232d6 : SUCCESS in 16m 40s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e04fd96d72b846c6a55b00be7e47e18a : FAILURE in 27m 11s (non-voting)\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e42710a3055432eb047b08c6bf416b6 : SUCCESS in 33m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/858e2b374dd0418f9776fa706d226a88 : SUCCESS in 14m 23s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57f51b4d8f034df3b2b3972053da71b7 : SUCCESS in 11m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2f6a3db22a1447d8967b79f83e87d967 : SUCCESS in 15m 38s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f60fa64c3c27420ca71cfb9eb873c898 : SUCCESS in 13m 24s","accounts_in_message":[],"_revision_number":4},{"id":"e483c1b1e26e318b63a1f5c734cfecb6a7a36925","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-19 08:41:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"00c59e27b8315d282e0dcc9c463cce673a5069ef","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-01-19 09:22:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"273f0645c8687ecd5942ae93076820c506404ab5","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-01-20 09:12:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2988cab2948544c2945076117bc492db0aa1c55d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-03 11:09:34.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"b1d2bbb72c6a270d08907c22687e32e9607fa180","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 13:02:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/018a4a78e882496d879b8ab78c44ad13\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/155e8503de154563b022fa51fee887e2 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6d5b9efd4c9473f9ae1d01e6130bcf4 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69469b1af5034f618f3a3ff85a8f97c5 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f092401deb141349ac37fcf29884de5 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d096481f56c4cbca9ccfda269a13e7c : SUCCESS in 6m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c76d9b20d9d34d8d8aa4409d0fe0472c : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/afda500a80eb4f4a9481843275752c81 : SUCCESS in 10m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/19713b8bc3f44dcab729134eb450a5d6 : SUCCESS in 15m 00s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/16d50e0c00d34ea097402c6abf3c7c80 : SUCCESS in 51m 12s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/74b21240787b4d228dc1f48b205ac256 : SUCCESS in 54m 45s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8ed448a7d084e8f91052b2c5af7850f : SUCCESS in 1h 26m 04s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7cfab6e3c9d348459acf4639b05a9d41 : SUCCESS in 52m 16s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/018fdf82d83e4ed19e3375d552c984a6 : FAILURE in 9m 24s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/de6da9db718f473398ba6848af806888 : FAILURE in 17m 15s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89d3f2a158974fb8ba32f9cc58958de2 : SUCCESS in 20m 00s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/01bd13efa69c4592b13d145521e4b411 : SUCCESS in 1h 42m 52s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce7903a62e89407385b22220bb884626 : FAILURE in 7m 00s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1d74a9741f2a4b75b1ddcd7b08b8fa25 : SUCCESS in 16m 19s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c715108c71b34b8b97b9a84955b39bd1 : SUCCESS in 20m 30s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a1955152957243fc8bdc723f8845643e : SUCCESS in 27m 32s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2cd712383bf74728ae462a342d23d1aa : SUCCESS in 31m 31s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50932da8fa804478847e957af468890b : FAILURE in 16m 48s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/07c18e625404418f837c5767fe449daf : SUCCESS in 13m 14s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ed779749b704c92bcf269db3d1b3b19 : SUCCESS in 11m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7194f369d5fb47da97e993d44f127d5e : SUCCESS in 17m 43s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4642bc3d003a42d5891a2580a2f6b74c : SUCCESS in 11m 56s","accounts_in_message":[],"_revision_number":5},{"id":"f0c4c50e29044ae51c67c5f641430a2faae9a1fe","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-03-11 08:54:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1650d381af37a23ee844d66ad1dfecb7859ca0a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-30 09:42:16.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"b751c61280d0091677d68c00b3e87756d31c8802","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 11:11:43.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/50bbebbf5afd44b2897f4436c3b93f20\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c80f186c1f44547ab01509a4def8dd9 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f443296db50b4ddb899037f56d1d52e2 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/252ec128ce8b451f94105818f2dd9528 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0891820225e347a1b31b17c793af15e1 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd33a7a3a2304e7f985cac35903bb9cd : SUCCESS in 8m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/066a9b28d4a3432690144e541ac7c662 : SUCCESS in 4m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3c8b561b13ea4a8783dd8ef84e3697b7 : SUCCESS in 10m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1c456f5631674740bfef4faf3ff0f3f3 : SUCCESS in 14m 43s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3f458375c903431d8488c774b5dd44c6 : SUCCESS in 47m 15s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/ef6b4668ea5e41b2b48ee0d637214a0b : FAILURE in 9m 31s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/552b31ed739a46dcbea237b3d534b3e9 : SUCCESS in 51m 07s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1c2ab3f08e8d481a83ce658e60ea2fc0 : SUCCESS in 51m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0f19676b2e954978a8752d985c8cbf2e : FAILURE in 9m 09s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/c8956b27b2f84c3bbcad0a1969580868 : FAILURE in 10m 06s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70f3f1f8537a4c2f9eca5aba1ab76e8c : SUCCESS in 13m 41s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/42b1580353294e4ea417423648c76bdf : SUCCESS in 1h 23m 43s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d56735f89174005b791369ad00cba6e : SUCCESS in 1h 17m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8c09353b6c574420b795f81c0c4248b5 : SUCCESS in 18m 29s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed3d168fb9694b4a8f989454dbf022fa : SUCCESS in 11m 54s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/34d35c42f2e7468eaaf6f9ed6d01695f : SUCCESS in 30m 29s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6d2a1c66591b480292ba358eedfadc1b : SUCCESS in 27m 36s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41478b1579a54926bc6e3b042ceabb73 : SUCCESS in 24m 33s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/537db27611104b109949e9cf6cd3d58f : SUCCESS in 12m 59s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7235151e6c73406da9246c1a5feb1c9a : SUCCESS in 11m 11s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3cfd72ae56b74bb181a7eef8b646e033 : SUCCESS in 17m 02s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/811198d3a6cc4215a42572f9766baeb1 : SUCCESS in 13m 28s","accounts_in_message":[],"_revision_number":6},{"id":"5dc6d80b71d21bf6b86386482f22d0673538457a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-30 11:36:58.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"65e8ae9305c0f14a445f616e71d632d417b5d6c7","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-30 11:37:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c44f100eccbfa41e8629680643c12d6d3667c12d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-30 12:33:24.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6d6c5fff475ae6452b96e32d1364472478bfb014","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-30 14:50:20.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4d0a4d1671a71a973e3ee58b9b4db416c132cc84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 16:36:13.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a35e9ecc21440fd90c2bef7c6c17a29\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34eee6e3592d450d94283c93ce63b2d5 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d05abd3f45a145da9c2f5171b8e964bd : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f90a1e7ad59e46fa84c8383c8b4f1430 : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a97aa1cb478444ba7ffde7c275bada9 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f0cb07fb8bd4c758232bfff8bdc4bf1 : SUCCESS in 6m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/be32ba736b484d45b509e4d055f34012 : SUCCESS in 5m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8cfd130861c84876ace2487df703dbb2 : SUCCESS in 8m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ae9cb2c1172148dd9189f8cf156eed30 : SUCCESS in 14m 36s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/55889420129840e6b13c3bb7b47299fb : SUCCESS in 51m 28s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f91363229dc74bcf836bfcad5c7c5eca : SUCCESS in 43m 41s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5cc1d6937eb749e4a745a1035cb0f066 : SUCCESS in 49m 00s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/709979d0019a48f590edd30107449172 : SUCCESS in 1h 06m 07s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/516e1aae1c304f6b8553eee5e8df67ad : SUCCESS in 17m 56s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/5cb6ed56e5da40438cad60a6de05521a : SUCCESS in 18m 06s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a06b7aa574e4e1082ee93ae190cfdbe : SUCCESS in 14m 22s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8c8dafead6f446ab9e4cde3b45800bb7 : SUCCESS in 1h 40m 42s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2dd8ddb402b74c98a3d7fa7e645b97f6 : SUCCESS in 51m 14s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/83cc717540a940b68a3919f8c1fe7e92 : SUCCESS in 19m 36s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a8c2a6b3fe148628c0166028acec008 : SUCCESS in 14m 38s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/165821d2ba384a009a2d9ca070012132 : SUCCESS in 31m 15s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d2909a3148b74dfab4666cfb59f14cdd : SUCCESS in 26m 17s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9fa4d67d98b4b208acce4a78823521c : SUCCESS in 22m 47s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c19c365035aa4dc99b833b4095a28dc0 : SUCCESS in 12m 57s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e57b6623b4a49f4a9f7ca49d7681d72 : SUCCESS in 10m 41s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ea150fb2634b4788a62476aaf9ffdc1e : SUCCESS in 15m 58s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27cd1b5ef4db4e8ba0f787b15b5d4bb4 : SUCCESS in 17m 07s","accounts_in_message":[],"_revision_number":7},{"id":"443a2c89f981336a6fd1d60e35471b37a25ba220","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-31 08:38:08.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1d3c046a7dee4efbb13c30ba12a56c13a2c6cc66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-04-05 13:23:02.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b6eb5b6ffe820b94b4f755417e74ef18d754be83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-04-05 13:29:09.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"cd463f3db073cee505996ef2904c5a6a3f6d436d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-04-05 13:29:50.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"ebd46197c4473cc5e46d20ae0a87dd4adf467cc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-04-05 13:31:30.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"51d0658ebc4f43fed566d3db8d34fcdae0f51587","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 15:16:26.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b18fd765d649497ca3b94f848363a199\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c55680c30ef4d5c8da3bd1c5ebd2ef8 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/891e01c0c2e34c11a6446317394448f7 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4661bd908294b669cb4abe5f378ba2c : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2c8aaeedd044584a97db0bcd691af04 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf5966a1411844d89c090f50e6755f96 : SUCCESS in 8m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e7572dcf24154063b7cfea371a535c8b : SUCCESS in 3m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ae5256b3468f49e493be9a20f7343b6a : SUCCESS in 10m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/37fee295a3e74d7492825d005a4efb13 : SUCCESS in 19m 03s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3a9cdb064fd047ad8a3e3a3f7a258108 : FAILURE in 41m 52s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/b11eef75333a42278ca135e2110c7de6 : SUCCESS in 57m 25s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d51d92d6411d4287b33184753fa12516 : SUCCESS in 47m 59s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3a6ff27ec77e44338696aacef64a56ae : SUCCESS in 1h 00m 49s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5bd699fa7ecb4d31a501d7542b92dd4e : SUCCESS in 19m 08s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/37338be8bcf8466f8ac214b81c1f9d10 : SUCCESS in 18m 16s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7d97541e8ef4731b7d62a18083d35e3 : SUCCESS in 14m 30s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e7ace81872314cc993792e5d8bd1a994 : FAILURE in 1h 37m 56s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f61d7aa77d94ad89b30976ccaa0b4f0 : FAILURE in 1h 06m 05s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/80bf1f00ca694c99816a931e326c8bf8 : SUCCESS in 17m 45s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7330d62f1d1b4de0b0066cd27c63ccc0 : SUCCESS in 12m 17s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/3d746be9f5f6491c8023622b386da77b : FAILURE in 20m 20s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/595456dfd9874fae9b215282608643de : SUCCESS in 27m 45s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/daddcc64ee8b435194483062fd555646 : SUCCESS in 21m 43s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/43d0b513a68845a5b312e7f3b28bf674 : SUCCESS in 12m 29s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/047f62f7eb8f4b6788fedbdfb102c1af : SUCCESS in 8m 25s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ff31e8dc9eef4ade9ae57766dde2bb56 : SUCCESS in 16m 17s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eced8a263b15486880bbc659978b87d4 : SUCCESS in 15m 08s","accounts_in_message":[],"_revision_number":11},{"id":"7697129dc314f20668a8a81bc117321d9ac5c4c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-04-06 08:01:09.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"9ea80103cc63fa5acca1ebe0606dc3be4db3d531","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 09:36:00.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/acf4458f52ef4bb1ae0391366a61f65c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ae68be15a22416d8536f940bbef09e3 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08021683581946138e6c3958ea05dfc1 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d776f92a63045caa1c0c69e6cd7fe40 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81b05eb6b8ea45ccb2b005036e4c2fa4 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83d1263657fc45bcba874bf9d7bb0069 : SUCCESS in 6m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c53a8fedd42b407cbf9c62c21bbdb828 : SUCCESS in 4m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/46f42ba2a9794dc0b7900918bd6bc729 : SUCCESS in 11m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ba3487f1f0ba486eb7d585dc8b05da48 : SUCCESS in 13m 29s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0213b79151b1424e96eb3dee285d3571 : SUCCESS in 53m 34s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/284f8b583d024ec5abcdca3a23749d53 : SUCCESS in 54m 38s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/13632f9cc0194a28bb8de43477de9038 : SUCCESS in 46m 36s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c60c2f7c58024b4aa5cc5864689c2c8b : SUCCESS in 1h 01m 21s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/bdb0e8b8a65446da9e525c0d1a23cb54 : SUCCESS in 18m 55s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d76c26b5aaf14ca3afa6a578cbfda0e6 : SUCCESS in 20m 10s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03bfb5387d4b4a17abe9d9bd51b766dd : SUCCESS in 12m 19s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/89a055c44c4846d7bd838907b8e02e64 : FAILURE in 1h 29m 32s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38082344a75e4483aaea91aacf438996 : FAILURE in 1h 04m 08s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/0bf7d086cbbe44b5a9e453586b81e822 : SUCCESS in 18m 22s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55bf53d3d0f545c88a69f087de701e28 : SUCCESS in 14m 14s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/7f4290734c614d17adcf55c7cac6f880 : SUCCESS in 30m 10s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5d6bb7fa3be244e08b143b4bc69ed022 : SUCCESS in 22m 15s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eca5317963cc4fe79a650c7889e46c51 : SUCCESS in 22m 16s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9a9a053a0496449cb876471edc2c0e4a : SUCCESS in 13m 39s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d079af4c0d8e4493ab17e438c8945ae2 : SUCCESS in 10m 15s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3d8170308a4b4501a8ba24a3283bf64d : SUCCESS in 16m 32s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f6306f014ba4870b17f254a517df48d : SUCCESS in 13m 37s","accounts_in_message":[],"_revision_number":12},{"id":"82177788b7d935f54d3a9638b005d335c42735ff","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-04-06 11:23:14.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"d3903d449dcf2960c31135c2d8a519c789ef5fe5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 13:02:17.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/33f733774647477098eed1b40847e468\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf0c7107722a464f870fb989a8eb38bc : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b15306d0f50d472eb1aa0636dc46e7b6 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46bc16f2a80f47b4aee8863b204cfe53 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01e7558770ec4dbd872a79febea9263a : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ff6f9f191aa4085b4a3257505fe7256 : SUCCESS in 8m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/db9aead71d5142c0a2b6f14ee385cbce : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7f4a8f4ca7b545f09e4d9ba5443c0813 : SUCCESS in 10m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/78c25c1ed43641a88598a576c69810d1 : SUCCESS in 15m 38s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e5280d9df7654789a84c945019307726 : SUCCESS in 57m 35s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/cf69af2aed9d4bcb9fd66ca57cee4757 : SUCCESS in 54m 01s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7a4c957c8dc401390b312f2025ff344 : SUCCESS in 51m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f21e06ff98b04460b6fc34e8bbbac68e : SUCCESS in 54m 03s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a130f865750444abbe693e7e8b909467 : SUCCESS in 19m 22s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ee1cce196e854e959e0d4b0c8d4f3d65 : SUCCESS in 18m 30s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0102a7d0c90409086bef1390cb698b9 : SUCCESS in 16m 05s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5a4e32a955dd48fc94845012e3de07cb : FAILURE in 1h 32m 08s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd7bf04a54c54e8c927d3ef2763a4ff9 : FAILURE in 1h 08m 59s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/403c15f425944f6fbc5e29d028e31875 : SUCCESS in 16m 01s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2670735d38704dcd8cc8181c5f2b38d2 : SUCCESS in 14m 42s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/204136e2ac004a23af7ded68ad82bd5e : SUCCESS in 30m 29s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/df13dfa442db4db99f3e83b75a1281be : SUCCESS in 24m 28s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc895ba1ae60476c8c52530e26a382aa : SUCCESS in 20m 14s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3a8d45dbf4124c46864a21c164d92639 : SUCCESS in 13m 13s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d37bd211a6ba4a66a4fc0ee6e9bb0653 : SUCCESS in 10m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/70aeb7828bda45e6a81d8a4192df5cd7 : SUCCESS in 16m 12s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80910941ea854088a38c1dcdee8b0f53 : SUCCESS in 13m 02s","accounts_in_message":[],"_revision_number":12},{"id":"e001d0972a72ffbfb6807ca23257fe2714d42360","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-04-07 09:47:05.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"changekind:NO_CHANGE\")\n","accounts_in_message":[],"_revision_number":13},{"id":"b8bca9b8d576b2bdfa71b85f70ee864d26f9839c","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-04-07 09:47:40.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"c1ff0dbd3e836890a3526e4bab286c592652921d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 11:24:15.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c73ec2cd0ec42ae886aba2bfad16a6c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2371d3f2bd7d4d69a018b1f82b53e1a2 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/434da06181f44d0fab957f75f930fa67 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4bd6e8ea01b4207ac244cbb4432b36c : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f563fa698c884f07a3b0891bef8d3f2a : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d314b1cb326491ea74fba6415530582 : SUCCESS in 7m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2cb42aac6bc647339c1686d6dae942c7 : SUCCESS in 4m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f38b7a4c80974c98a418f1d879729c2f : SUCCESS in 10m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bada175fdf204a53960d8f6508f81688 : SUCCESS in 15m 06s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4f05a6218abb41759d0bc21773b77b48 : SUCCESS in 53m 49s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/e200becf3dd94034a696c6f88c2a8821 : SUCCESS in 1h 02m 20s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32c72f1dd6ab42ada389518174adeeb3 : POST_FAILURE in 39m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/05768b0e9fec46b981e9d2c9706f8c14 : SUCCESS in 1h 03m 37s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/82a5b3ce58314bf7a98d54f3d768442c : SUCCESS in 16m 47s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e410d437b9fc49ef941cc1c4a194aa5d : SUCCESS in 13m 12s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0102ee8e6bf48889d108da9f7a56568 : SUCCESS in 14m 57s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e0f8bb2bf64d405fbe9fcaa72a4322a7 : FAILURE in 1h 32m 07s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7945685b8cc4b7196ecf63990058f08 : SUCCESS in 1h 07m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/64eb0cea762645c499ffeca8e7789254 : SUCCESS in 16m 52s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/418afc878eba41c8b823d181ffd07542 : SUCCESS in 13m 45s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6b9fb1f6bc37437d82de47b7489ef105 : SUCCESS in 35m 27s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/dfcca0fd47564359b7878431635c257b : SUCCESS in 27m 00s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2492dd5c3824acca2f0d866f79c89e3 : SUCCESS in 21m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6414c6b4da304a84a38cabe00e22d90c : SUCCESS in 12m 42s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b983b3f2dbf400e8af8b3c3be2b63f7 : SUCCESS in 10m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/59f9881834b944979ab5587754aca9ba : SUCCESS in 15m 49s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d478c01c00cb461e8568cfaad1f35604 : SUCCESS in 12m 39s","accounts_in_message":[],"_revision_number":13},{"id":"71ddb33792a21f730a49797be416666f67916603","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-04-13 08:50:26.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"6fe044ef6c1d4f2be3e48f14edeb7c36cb190b04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 10:11:23.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67440530e186453b8147dc1f6bd7d18b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7bc7b5e102c4d6094c36a979f884e70 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b16f162bb6eb455987db80aab6acbc46 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c237d9f0b4f148fd96d6ec82de847fe0 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ada59ff48f346a7b2a3b5ad31f7c0cd : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd3c87287bbb4e2fb9d8ca0407ea2b18 : SUCCESS in 6m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2f6f1f5563014de19f4b1e9a957c4d06 : SUCCESS in 3m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0e8dec60d36b4519af69dcf5ad580c50 : SUCCESS in 8m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b036c2f74555475dac76aabafe12dc78 : SUCCESS in 12m 18s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d8909a2bb7f64f1181229de8599582ea : SUCCESS in 44m 17s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/a381e02c407340258a80c2263e5bf0c4 : SUCCESS in 52m 02s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07b22500a13041ea8e5a4bd4690ace32 : SUCCESS in 39m 51s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e7d1e639751345ccb94adb0f104afdf8 : SUCCESS in 46m 51s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9104833b6685435c9e52d6c62b75e489 : SUCCESS in 18m 01s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/aa06841dc46241d1a1c44edf2246d676 : SUCCESS in 16m 15s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/faacd9482cea445db7c5ff012b81dcf6 : SUCCESS in 13m 04s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0119a58500dc4f3295d6965fd7b2363b : SUCCESS in 1h 14m 41s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1809804df96240969fa820e850c8ade8 : SUCCESS in 1h 00m 21s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4085165051014f38a00e52e5c5f11ef0 : SUCCESS in 14m 51s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b974a8aa8684eaeb8ecd88e46546c5d : SUCCESS in 13m 17s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/0cacaab8f0f747f8ab3d00425c357858 : SUCCESS in 25m 57s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/91819b8127fc41749586cd7d0d8fef92 : SUCCESS in 22m 24s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8b7d68cdcde4d78a025608a1f9bd45e : SUCCESS in 18m 59s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6423a8f6771d4f8c969a4aaad755ff3e : SUCCESS in 11m 02s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac831be3ae6147019a3db0f2f57418a6 : SUCCESS in 10m 18s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b2da737814f04598b83485adb7ab5ceb : SUCCESS in 15m 29s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4e51c638a0f4ea5b9cbc0c0a750fecf : SUCCESS in 12m 14s","accounts_in_message":[],"_revision_number":14},{"id":"67969e2a8df2fc154073b8db8a467228a4e04556","tag":"autogenerated:gerrit:abandon","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-11-26 07:58:55.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"a2abbd023f01b4cb499b01f72d5f3dcaf78c5f35","revisions":{"7d80ad8d0db9216aee53388a45fe6ba5ad1672a0":{"kind":"REWORK","_number":1,"created":"2023-01-12 13:40:29.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/77/869977/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/869977/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/869977/1"}}},"commit":{"parents":[{"commit":"ebd5799fd4b77a138cf7790a8094eb859e8898f3","subject":"Merge \"Replace references to Docker Hub by Quay.io\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ebd5799fd4b77a138cf7790a8094eb859e8898f3"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-12 13:39:36.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-12 13:40:28.000000000","tz":60},"subject":"[DNM] ansible-role-interfaces Network Manager test","message":"[DNM] ansible-role-interfaces Network Manager test\n\nChange-Id: I9cf265ffc7acce926ce093d1e4f8d53a12be1f26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7d80ad8d0db9216aee53388a45fe6ba5ad1672a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7d80ad8d0db9216aee53388a45fe6ba5ad1672a0"}]},"branch":"refs/heads/master"},"bbae32f8ce95033308865cfecbcd9dce553f9031":{"kind":"REWORK","_number":2,"created":"2023-01-16 22:45:41.000000000","uploader":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"ref":"refs/changes/77/869977/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/869977/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/869977/2"}}},"commit":{"parents":[{"commit":"7a78dda176e7798f677c740a10cb9203342d50ea","subject":"Merge \"Apply TODOs for Antelope cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7a78dda176e7798f677c740a10cb9203342d50ea"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-12 13:39:36.000000000","tz":60},"committer":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2023-01-16 22:45:02.000000000","tz":60},"subject":"[DNM] ansible-role-interfaces Network Manager test","message":"[DNM] ansible-role-interfaces Network Manager test\n\nChange-Id: I9cf265ffc7acce926ce093d1e4f8d53a12be1f26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bbae32f8ce95033308865cfecbcd9dce553f9031"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bbae32f8ce95033308865cfecbcd9dce553f9031"}]},"branch":"refs/heads/master"},"7161a4b1b21f5dd88b85dead8671959522f21222":{"kind":"REWORK","_number":3,"created":"2023-01-17 12:52:24.000000000","uploader":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"ref":"refs/changes/77/869977/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/869977/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/869977/3"}}},"commit":{"parents":[{"commit":"7a78dda176e7798f677c740a10cb9203342d50ea","subject":"Merge \"Apply TODOs for Antelope cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7a78dda176e7798f677c740a10cb9203342d50ea"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-12 13:39:36.000000000","tz":60},"committer":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2023-01-17 11:13:01.000000000","tz":60},"subject":"[DNM] ansible-role-interfaces NetworkManager test","message":"[DNM] ansible-role-interfaces NetworkManager test\n\nThe ansible-role-interfaces is keeping backwards-compatible rules\nas strings, but been updated to handle rules as dicts for EL-family.\n\nnmconnections seem to want a table_id rather than name for it\u0027s\nrules, and the dictionary is the preferred way, where the templates\ncan perform a lookup of the id for a given name.\n\nHowever, if using strings for nmconnections the strings must specify\nthe table id, instead of the name.\n\nChange-Id: I9cf265ffc7acce926ce093d1e4f8d53a12be1f26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7161a4b1b21f5dd88b85dead8671959522f21222"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7161a4b1b21f5dd88b85dead8671959522f21222"}]},"branch":"refs/heads/master"},"135c9f047daaee1491533c7212adaafee87232f4":{"kind":"REWORK","_number":4,"created":"2023-01-18 14:45:36.000000000","uploader":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"ref":"refs/changes/77/869977/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/869977/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/869977/4"}}},"commit":{"parents":[{"commit":"7a78dda176e7798f677c740a10cb9203342d50ea","subject":"Merge \"Apply TODOs for Antelope cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7a78dda176e7798f677c740a10cb9203342d50ea"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-12 13:39:36.000000000","tz":60},"committer":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2023-01-17 20:47:32.000000000","tz":60},"subject":"[DNM] ansible-role-interfaces NetworkManager test","message":"[DNM] ansible-role-interfaces NetworkManager test\n\nThe ansible-role-interfaces is keeping backwards-compatible rules\nas strings, but been updated to handle rules as dicts for EL-family.\n\nnmconnections seem to want a table_id rather than name for it\u0027s\nrules, and the dictionary is the preferred way, where the templates\ncan perform a lookup of the id for a given name.\n\nHowever, if using strings for nmconnections the strings must specify\nthe table id, instead of the name.\n\nChange-Id: I9cf265ffc7acce926ce093d1e4f8d53a12be1f26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/135c9f047daaee1491533c7212adaafee87232f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/135c9f047daaee1491533c7212adaafee87232f4"}]},"branch":"refs/heads/master"},"c179a7d4b60e0b54fc0361122f69734db4657694":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-03-03 11:09:34.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/77/869977/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/869977/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/869977/5"}}},"commit":{"parents":[{"commit":"db9ce40e81af0e20fb89b96ec968293843a3fbe2","subject":"Merge \"docs: add information about enabling TLS to automated guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/db9ce40e81af0e20fb89b96ec968293843a3fbe2"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-12 13:39:36.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-03 11:09:34.000000000","tz":0},"subject":"[DNM] ansible-role-interfaces NetworkManager test","message":"[DNM] ansible-role-interfaces NetworkManager test\n\nThe ansible-role-interfaces is keeping backwards-compatible rules\nas strings, but been updated to handle rules as dicts for EL-family.\n\nnmconnections seem to want a table_id rather than name for it\u0027s\nrules, and the dictionary is the preferred way, where the templates\ncan perform a lookup of the id for a given name.\n\nHowever, if using strings for nmconnections the strings must specify\nthe table id, instead of the name.\n\nChange-Id: I9cf265ffc7acce926ce093d1e4f8d53a12be1f26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c179a7d4b60e0b54fc0361122f69734db4657694"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c179a7d4b60e0b54fc0361122f69734db4657694"}]},"branch":"refs/heads/master","description":"Rebase"},"4fd4f741819df1935ea7b38e35aa0987b39e84df":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-03-30 09:42:16.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/77/869977/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/869977/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/869977/6"}}},"commit":{"parents":[{"commit":"137e64b57c86fde852b49f9975ffbca26da79de1","subject":"Fix opensearch container builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/137e64b57c86fde852b49f9975ffbca26da79de1"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-12 13:39:36.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-30 09:42:16.000000000","tz":0},"subject":"[DNM] ansible-role-interfaces NetworkManager test","message":"[DNM] ansible-role-interfaces NetworkManager test\n\nThe ansible-role-interfaces is keeping backwards-compatible rules\nas strings, but been updated to handle rules as dicts for EL-family.\n\nnmconnections seem to want a table_id rather than name for it\u0027s\nrules, and the dictionary is the preferred way, where the templates\ncan perform a lookup of the id for a given name.\n\nHowever, if using strings for nmconnections the strings must specify\nthe table id, instead of the name.\n\nChange-Id: I9cf265ffc7acce926ce093d1e4f8d53a12be1f26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4fd4f741819df1935ea7b38e35aa0987b39e84df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4fd4f741819df1935ea7b38e35aa0987b39e84df"}]},"branch":"refs/heads/master","description":"Rebase"},"01753259f3b5f7626724de8e634865171293f9f3":{"kind":"REWORK","_number":7,"created":"2023-03-30 14:50:20.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/77/869977/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/869977/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/869977/7"}}},"commit":{"parents":[{"commit":"137e64b57c86fde852b49f9975ffbca26da79de1","subject":"Fix opensearch container builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/137e64b57c86fde852b49f9975ffbca26da79de1"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-12 13:39:36.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-30 14:50:15.000000000","tz":120},"subject":"[DNM] ansible-role-interfaces NetworkManager test","message":"[DNM] ansible-role-interfaces NetworkManager test\n\nThe ansible-role-interfaces is keeping backwards-compatible rules\nas strings, but been updated to handle rules as dicts for EL-family.\n\nnmconnections seem to want a table_id rather than name for it\u0027s\nrules, and the dictionary is the preferred way, where the templates\ncan perform a lookup of the id for a given name.\n\nHowever, if using strings for nmconnections the strings must specify\nthe table id, instead of the name.\n\nChange-Id: I9cf265ffc7acce926ce093d1e4f8d53a12be1f26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/01753259f3b5f7626724de8e634865171293f9f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/01753259f3b5f7626724de8e634865171293f9f3"}]},"branch":"refs/heads/master"},"dbffd6e0f2018541c977ba1de3585fa70b49478b":{"kind":"REWORK","_number":8,"created":"2023-04-05 13:23:02.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/77/869977/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/869977/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/869977/8"}}},"commit":{"parents":[{"commit":"137e64b57c86fde852b49f9975ffbca26da79de1","subject":"Fix opensearch container builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/137e64b57c86fde852b49f9975ffbca26da79de1"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-12 13:39:36.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-04-05 13:23:00.000000000","tz":0},"subject":"[DNM] ansible-role-interfaces NetworkManager test","message":"[DNM] ansible-role-interfaces NetworkManager test\n\nThe ansible-role-interfaces is keeping backwards-compatible rules\nas strings, but been updated to handle rules as dicts for EL-family.\n\nnmconnections seem to want a table_id rather than name for it\u0027s\nrules, and the dictionary is the preferred way, where the templates\ncan perform a lookup of the id for a given name.\n\nHowever, if using strings for nmconnections the strings must specify\nthe table id, instead of the name.\n\nChange-Id: I9cf265ffc7acce926ce093d1e4f8d53a12be1f26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dbffd6e0f2018541c977ba1de3585fa70b49478b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dbffd6e0f2018541c977ba1de3585fa70b49478b"}]},"branch":"refs/heads/master"},"cc6e333335713eccfe2a53694dc0bb3418b9f49c":{"kind":"REWORK","_number":9,"created":"2023-04-05 13:29:09.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/77/869977/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/869977/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/869977/9"}}},"commit":{"parents":[{"commit":"137e64b57c86fde852b49f9975ffbca26da79de1","subject":"Fix opensearch container builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/137e64b57c86fde852b49f9975ffbca26da79de1"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-12 13:39:36.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-04-05 13:29:02.000000000","tz":0},"subject":"[DNM] ansible-role-interfaces NetworkManager test","message":"[DNM] ansible-role-interfaces NetworkManager test\n\nThe ansible-role-interfaces is keeping backwards-compatible rules\nas strings, but been updated to handle rules as dicts for EL-family.\n\nnmconnections seem to want a table_id rather than name for it\u0027s\nrules, and the dictionary is the preferred way, where the templates\ncan perform a lookup of the id for a given name.\n\nHowever, if using strings for nmconnections the strings must specify\nthe table id, instead of the name.\n\nChange-Id: I9cf265ffc7acce926ce093d1e4f8d53a12be1f26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc6e333335713eccfe2a53694dc0bb3418b9f49c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc6e333335713eccfe2a53694dc0bb3418b9f49c"}]},"branch":"refs/heads/master"},"d305c39b6072562470021d721e5a4b70a7886e1e":{"kind":"REWORK","_number":10,"created":"2023-04-05 13:29:50.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/77/869977/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/869977/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/869977/10"}}},"commit":{"parents":[{"commit":"137e64b57c86fde852b49f9975ffbca26da79de1","subject":"Fix opensearch container builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/137e64b57c86fde852b49f9975ffbca26da79de1"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-12 13:39:36.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-04-05 13:29:43.000000000","tz":0},"subject":"[DNM] ansible-role-interfaces NetworkManager test","message":"[DNM] ansible-role-interfaces NetworkManager test\n\nThe ansible-role-interfaces is keeping backwards-compatible rules\nas strings, but been updated to handle rules as dicts for EL-family.\n\nnmconnections seem to want a table_id rather than name for it\u0027s\nrules, and the dictionary is the preferred way, where the templates\ncan perform a lookup of the id for a given name.\n\nHowever, if using strings for nmconnections the strings must specify\nthe table id, instead of the name.\n\nChange-Id: I9cf265ffc7acce926ce093d1e4f8d53a12be1f26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d305c39b6072562470021d721e5a4b70a7886e1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d305c39b6072562470021d721e5a4b70a7886e1e"}]},"branch":"refs/heads/master"},"07263768386aea6bc0bb18b68311dcf54eff2485":{"kind":"REWORK","_number":11,"created":"2023-04-05 13:31:30.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/77/869977/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/869977/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/869977/11"}}},"commit":{"parents":[{"commit":"137e64b57c86fde852b49f9975ffbca26da79de1","subject":"Fix opensearch container builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/137e64b57c86fde852b49f9975ffbca26da79de1"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-12 13:39:36.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-04-05 13:31:22.000000000","tz":0},"subject":"[DNM] ansible-role-interfaces NetworkManager test","message":"[DNM] ansible-role-interfaces NetworkManager test\n\nThe ansible-role-interfaces is keeping backwards-compatible rules\nas strings, but been updated to handle rules as dicts for EL-family.\n\nnmconnections seem to want a table_id rather than name for it\u0027s\nrules, and the dictionary is the preferred way, where the templates\ncan perform a lookup of the id for a given name.\n\nHowever, if using strings for nmconnections the strings must specify\nthe table id, instead of the name.\n\nChange-Id: I9cf265ffc7acce926ce093d1e4f8d53a12be1f26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/07263768386aea6bc0bb18b68311dcf54eff2485"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/07263768386aea6bc0bb18b68311dcf54eff2485"}]},"branch":"refs/heads/master"},"550d57a22936f38b9895b9d821bd67c334ca0b01":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2023-04-06 08:01:09.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/77/869977/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/869977/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/869977/12"}}},"commit":{"parents":[{"commit":"e6432ce671dc65b3380f7f298e5c87bed937df0d","subject":"Merge \"CI: Add custom routing table output to logs.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6432ce671dc65b3380f7f298e5c87bed937df0d"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-12 13:39:36.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-04-06 08:01:09.000000000","tz":0},"subject":"[DNM] ansible-role-interfaces NetworkManager test","message":"[DNM] ansible-role-interfaces NetworkManager test\n\nThe ansible-role-interfaces is keeping backwards-compatible rules\nas strings, but been updated to handle rules as dicts for EL-family.\n\nnmconnections seem to want a table_id rather than name for it\u0027s\nrules, and the dictionary is the preferred way, where the templates\ncan perform a lookup of the id for a given name.\n\nHowever, if using strings for nmconnections the strings must specify\nthe table id, instead of the name.\n\nChange-Id: I9cf265ffc7acce926ce093d1e4f8d53a12be1f26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/550d57a22936f38b9895b9d821bd67c334ca0b01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/550d57a22936f38b9895b9d821bd67c334ca0b01"}]},"branch":"refs/heads/master","description":"Rebase"},"9eaa0457acd5034bbf82cc55e2511a9ccf2cb4a0":{"kind":"REWORK","_number":13,"created":"2023-04-07 09:47:05.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/77/869977/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/869977/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/869977/13"}}},"commit":{"parents":[{"commit":"e6432ce671dc65b3380f7f298e5c87bed937df0d","subject":"Merge \"CI: Add custom routing table output to logs.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6432ce671dc65b3380f7f298e5c87bed937df0d"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-12 13:39:36.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-04-07 09:47:01.000000000","tz":120},"subject":"[DNM] ansible-role-interfaces NetworkManager test","message":"[DNM] ansible-role-interfaces NetworkManager test\n\nThe ansible-role-interfaces is keeping backwards-compatible rules\nas strings, but been updated to handle rules as dicts for EL-family.\n\nnmconnections seem to want a table_id rather than name for it\u0027s\nrules, and the dictionary is the preferred way, where the templates\ncan perform a lookup of the id for a given name.\n\nHowever, if using strings for nmconnections the strings must specify\nthe table id, instead of the name.\n\nChange-Id: I9cf265ffc7acce926ce093d1e4f8d53a12be1f26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9eaa0457acd5034bbf82cc55e2511a9ccf2cb4a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9eaa0457acd5034bbf82cc55e2511a9ccf2cb4a0"}]},"branch":"refs/heads/master"},"a2abbd023f01b4cb499b01f72d5f3dcaf78c5f35":{"kind":"REWORK","_number":14,"created":"2023-04-13 08:50:26.000000000","uploader":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"ref":"refs/changes/77/869977/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/869977/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/869977/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/869977/14"}}},"commit":{"parents":[{"commit":"e6432ce671dc65b3380f7f298e5c87bed937df0d","subject":"Merge \"CI: Add custom routing table output to logs.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6432ce671dc65b3380f7f298e5c87bed937df0d"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-12 13:39:36.000000000","tz":60},"committer":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2023-04-13 08:49:42.000000000","tz":120},"subject":"[DNM] ansible-role-interfaces NetworkManager test","message":"[DNM] ansible-role-interfaces NetworkManager test\n\nThe ansible-role-interfaces is keeping backwards-compatible rules\nas strings, but been updated to handle rules as dicts for EL-family.\n\nnmconnections seem to want a table_id rather than name for it\u0027s\nrules, and the dictionary is the preferred way, where the templates\ncan perform a lookup of the id for a given name.\n\nHowever, if using strings for nmconnections the strings must specify\nthe table id, instead of the name.\n\nChange-Id: I9cf265ffc7acce926ce093d1e4f8d53a12be1f26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a2abbd023f01b4cb499b01f72d5f3dcaf78c5f35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a2abbd023f01b4cb499b01f72d5f3dcaf78c5f35"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
