)]}'
{"id":"openstack%2Fkayobe~870522","triplet_id":"openstack%2Fkayobe~master~Iefd2d0b7a3a3e07f5c112d58e2ec0b3da0a747d3","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-04-17 13:11:10.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-04-17 12:31:31.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-04-17 15:10:29.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2023-04-17 15:10:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iefd2d0b7a3a3e07f5c112d58e2ec0b3da0a747d3","subject":"Build Ubuntu images with IPA when on Ubuntu","status":"MERGED","created":"2023-01-16 10:19:55.000000000","updated":"2023-04-17 15:11:31.000000000","submitted":"2023-04-17 15:10:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":26,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870522","meta_rev_id":"3bc0c0ea2726848e65cdf5896dbeed5034fba703","_number":870522,"virtual_id_number":870522,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-17 15:10:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-04-17 13:11:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-04-17 12:31:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-04-17 13:14:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-04-17 13:14:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-16 10:22:27.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-01-16 11:31:11.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2023-01-16 13:39:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-13 17:08:38.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-04-17 13:14:41.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"f49825570c051440f2022abc8e2d64f4bd6c1dbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-01-16 10:19:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36625e1a9a1f74d45402b62b9ad5b2cded20ac7c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-16 10:22:27.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"05091023a947184d08ed91f057c0eedfc4f853dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-01-16 10:31:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4ca1ebe250767b82814f95c9d8a9991ea58b3bba","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-01-16 10:47:42.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d2bc8d1bc2c9783d0e84df69322daf3f5364d4ac","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-16 11:31:11.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4dc1a2d0f1296ed3eb6a83b1880729c03a627b19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-01-16 11:34:06.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"5c3dbb2cdc8284db7870158166bd40d59a5977ff","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-01-16 11:34:19.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"36afcc662090c1865dde4de6f17b5d35173e9720","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-16 11:35:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"35acdc3ce38eaa43e0157419b84425fa7ff1e7d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-01-16 12:20:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"900d3f2ae639fa50d073a457f83b6255fac6adc6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-16 13:38:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9c5d3e4e000517eb54fb8b192d143e6a1e8dbdb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 13:39:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ad1219d98924e148ef7369c13ef66bf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f46f73fd30d74a4180dd15ca0e993294 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/383565829b254199a8fa8d91ba51eb46 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/136b61cdcdf84d21ad1f9fe4f47ac66e : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2cc0936440a424f984a7006d34afd79 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5687dc31eee24bdf898dfefd136326f0 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/643a98fc03fb43fdb61998e4404bcce2 : SUCCESS in 2m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4b152354ad3c4f1b832ba059e6cad8b1 : SUCCESS in 3m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/84863f3d62db43ffbe77ee4b0be2ba09 : SUCCESS in 10m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2d0b91329fba4445980e4d7d3f8197c3 : SUCCESS in 14m 38s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e348927e710c43109a873d0857a643dd : SUCCESS in 58m 43s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/43a25bd4e06d4c31bdb5d329457442b3 : SUCCESS in 56m 52s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e408194ef4e543a6bad4d6717475b23a : SUCCESS in 47m 57s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/427a7c8ee7db48bfaf225018c141ad9c : SUCCESS in 56m 56s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b0b064d5be024d6aa74e1e51554d9e0a : SUCCESS in 17m 28s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d8bc235417a6422282060f2dfd7698e2 : SUCCESS in 21m 24s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/310be53c40f8448192749eb6ac698853 : SUCCESS in 15m 23s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2294fe935e124faf885c7d42ee23f5d4 : SUCCESS in 1h 11m 08s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/88327954fc244adf96bc161bfbef1055 : SUCCESS in 18m 33s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36c56a8cfeee4507857ab8953c0e3fdb : SUCCESS in 28m 03s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c2ebb331648a488086c325a37e3de581 : FAILURE in 26m 39s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aab4abfc72074cc5bda5b93e305fdde2 : SUCCESS in 27m 22s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0788ca23e1604cf1a283bd469f84c6ab : SUCCESS in 12m 55s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1126a66770884d33ac417e20d782ce6d : SUCCESS in 12m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6f1de711b8c1422c808f46c7d1d352eb : SUCCESS in 16m 44s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6aaf969cc0144eb6873a53e83241ac60 : SUCCESS in 24m 35s","accounts_in_message":[],"_revision_number":4},{"id":"b0216f2ce983498632bc05ca0732dea5a30a21fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-01-16 13:48:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"44e6befbd8b2afd3ede3d89a010c4a3baf7d2254","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-01-16 13:50:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"eed9b807176d9193d5fd136351068ee13f10b474","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 15:10:42.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/de1b404b70ba44baaa0ef6770ebd9f21\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b6ebeb731454feeafad98acdc562086 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4349f9e6ea24e5192baa04233ae89c0 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2dc95e77c6c4c119df6af788ae2acee : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe3bd15a416146359a1257c14a0b0641 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/798e5981ee924d8d97e3e6144a1f7588 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03db95855d5243aa9fbda0ecae80a4ee : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f366ff5f6d8c4d2895af4f5605f508cd : SUCCESS in 5m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e7eaa4bacf5c4691a49197e40f3609f9 : SUCCESS in 11m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f6d8a5f49dc94c81b129d7c188707c40 : SUCCESS in 12m 38s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a522c38eb5c34fc3883bcef7f94fa7ce : FAILURE in 52m 33s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c13d28ef3fde4ebe87f40253dbd73950 : FAILURE in 54m 54s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c74e0b5745d47d58e850db80ef1bcb3 : SUCCESS in 51m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/87d01a3115ad4d4dabe6ea1aa2d68869 : SUCCESS in 57m 02s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ddd7e0f02a1e4b1d9c392d76d4b8ea29 : SUCCESS in 17m 56s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/9ebb0c1820f14771a64d2413a71144aa : SUCCESS in 20m 17s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1ded3b684434c7cb3e05751be877a23 : SUCCESS in 16m 04s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69bb8d602d9e4447893f0c088ef2a3e6 : SUCCESS in 1h 17m 41s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3ebc637addfa4fb3a83904290a36cc4f : SUCCESS in 19m 31s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f64820556ea041e3a0d456b9ccae3db5 : SUCCESS in 16m 36s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1637675ad7f34dcdbec3f6135654efdc : FAILURE in 27m 38s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f640ce059de46809c4ec8538e96a413 : SUCCESS in 29m 58s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3f434914d7dc4a4bb698b03170e909a2 : SUCCESS in 12m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a7841f29d984e909fed71ea9fee44a5 : SUCCESS in 9m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/904d6ac2177d45828f66f5e3bfd0d422 : SUCCESS in 16m 11s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de0fc09b24bf486a97f805efa4029247 : SUCCESS in 12m 47s","accounts_in_message":[],"_revision_number":5},{"id":"60d15c29ae6a8c97ce2e26c72cd819db2e70786c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-19 14:34:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5a51b30426b12d4628c16f655e09f679d3351c10","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-19 14:34:37.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"43ebee41384d97fddef53c88ffa1bb018b32846a","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 15:14:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c16709add7714b52b8f566d1283e2a74\n\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/d5fde05d93cd47dcb970e4c653e961f6 : SUCCESS in 35m 21s (non-voting)\n- kayobe-seed-images-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a260d64dbef45e2b82f8f40ffffee33 : FAILURE in 20m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e3acd63a90c0a2dee77be271156dbb9b3d6e1ae8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-19 15:21:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b2db1573fb0e9963644851baa6abfcaaee3c93ce","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-19 15:28:38.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a3fccaff1e3a29f89ff4a3b933e4619385057efc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 16:07:07.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f84f9878edb41e892d3e46f4a183548\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/994e25a84efe4b41a87508557ddee070 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1dcca872aaa418d9930601615a10877 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a998348f3583433c82269677614d7e7b : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d4f8896131241428a58f8950f454581 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f939f5c6ff94505986e296368956590 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36cbc3944f394d6fb28d89767fd97c44 : SUCCESS in 3m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/52028c77f9c44ed69340a027144b3f93 : SUCCESS in 4m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f5a0a3537eaa410a9e0dea66de2aa67b : SUCCESS in 12m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/019b97789ae1454ba0da49043cf9c0fd : SUCCESS in 18m 14s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5e4a737f43ac43eab9cb781ee10321ab : FAILURE in 50m 50s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/69a08b28f12249df8d30e335a31d1cb0 : SUCCESS in 1h 10m 29s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29ccbfc292c14890ba8b73a822e4bb95 : SUCCESS in 48m 05s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/75043b86ffc148d98ecf7911f7668b1f : SUCCESS in 59m 02s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/78d7e066a183449c9dbaae429d008e4b : SUCCESS in 19m 34s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/c843629e2352488db4ae84a7bc50ddd7 : SUCCESS in 21m 49s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d3d98a5f07846de85cdfc6d2728c1cd : SUCCESS in 16m 15s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33709dac08864f4f910608dfdd96c62a : SUCCESS in 1h 07m 49s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/664f713860ed4a2eb50a1720e3b61dd1 : SUCCESS in 18m 14s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3cfccdaaa1e04ec69e8153e02f84a1f7 : SUCCESS in 17m 21s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a2933e1be16d4a499131e0078355fca2 : FAILURE in 30m 06s (non-voting)\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5dbea8b624d48c895728cf8461ed114 : SUCCESS in 33m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c20b05cc4ccc41a997af1fdd5deb2655 : SUCCESS in 15m 09s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33396dd9eeb34541969012934bd60709 : SUCCESS in 13m 11s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/853bae900d0c4c619ca7fed174843ac2 : SUCCESS in 17m 15s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75f03764fdae4137b44ea4774e203012 : SUCCESS in 15m 06s","accounts_in_message":[],"_revision_number":5},{"id":"ae89e6f7062cb884b68b7bb2f9b4a79aa9dcff69","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 16:20:40.000000000","message":"Patch Set 5:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7777ba7d8e74495ad4017a722e864b4\n\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/4dbb02ced0594c54aba88b00586ae46d : SUCCESS in 31m 58s (non-voting)\n- kayobe-seed-images-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/241d95879487459fbc039ce4057cae0f : FAILURE in 19m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ea39332869e9808da51045c85e5b0c9260bb7fde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-03-30 14:52:49.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"5caa8678265c3e364d043d4563e0fcef6de45099","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 16:31:28.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85542bd16cd04c0c891d9a21289e1f1d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0adf4ba7673448c5b34f1b02fc65e096 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b9a8f1539dd41fea5c4e2b156fbaa35 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c3f15aefc154637960a7abd514ea45a : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c383447ccc1e4a7596cdf0853f2ee4b6 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c420d0b433c64a658fb43be9846273b7 : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c8a1be3cdae48eda30bf7bc357459fa : SUCCESS in 3m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c77ebde558fb418fa0ce63b7932b699b : SUCCESS in 3m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a8ad6d0553574bde84fab0f98934dbd9 : SUCCESS in 9m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a70f95e9f8a44ef3a9b8dfc454620a00 : SUCCESS in 17m 28s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/550585203b1942a885050be749674549 : SUCCESS in 41m 33s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/3c3ea22d254a4fc8a26181c5f301b14f : SUCCESS in 47m 07s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e36992c33d5042d9b8f4b3800f9ba980 : SUCCESS in 39m 22s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7f25b75c2544455697da4cbd388f3cbd : SUCCESS in 1h 06m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7c9b6a1e84144121a3cda5acbfd6df7a : SUCCESS in 13m 18s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/09bfda59f5d1412d98d5ecd7d994889f : SUCCESS in 20m 58s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e71fd5c230645efae553544e62a6a7d : SUCCESS in 18m 18s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7fc83cde64bb405695c9eaacf319cfa8 : SUCCESS in 1h 32m 43s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2b4d239392a4018972062272b46ef6c : SUCCESS in 1h 15m 48s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8295d882c07048e299e7f4f5657fb2e2 : SUCCESS in 17m 18s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8226a32806df49bf932ae017d183185e : SUCCESS in 16m 44s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/06cea52b1f3c4570b2b6b5eebdeb71b5 : SUCCESS in 29m 48s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0d5b86488c234c919fb71cd72a397bce : SUCCESS in 25m 25s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a5569747917494db6edfcd35d5547ee : SUCCESS in 23m 12s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0347751b277e4114a18e66313fd87060 : SUCCESS in 13m 55s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/135da849c1ad4907879ba73de1bce610 : SUCCESS in 9m 56s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9fc33dad25cc4778af9aa49495d6df99 : SUCCESS in 17m 05s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67cbe05be91247529c95182c887c79bd : SUCCESS in 13m 23s","accounts_in_message":[],"_revision_number":6},{"id":"aa8190dd694a82f7007069a41498eb8666a36bc7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-03-30 16:32:58.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ed1ed0fb4c7da5f2040499c616a50363101a61bf","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 17:13:56.000000000","message":"Patch Set 6:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49c89e4bfbd8466a85dced61d728285a\n\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/b7691dadd1e7483298cec3ea4e824f50 : SUCCESS in 35m 20s (non-voting)\n- kayobe-seed-images-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7fca6b9c91d242d884882c3073081350 : FAILURE in 14m 17s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9236e0b470953bb836c26aac4d8da7817e33ff61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-03-30 19:21:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"989dc04c8b14ada9f823c13f0d4a147c87306d38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 21:01:12.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f27b4ba4fdcd479e86b546fdc1c0bacc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4df341c2304e41c2b28a673f94f42234 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c3e528cc58e43d58715e0cc2c3308ed : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e211a0d15c8b49d7ad1fd6d6e762cae1 : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/acb040beddca49e5a0738d573024a23b : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f34754c81f8c46b4be08ca3bfa09462f : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29d2bb84933a4ff4a350569521d5ade4 : SUCCESS in 3m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e716a58edfac463bb5fd939f032472ca : SUCCESS in 3m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d7cb1ab6fa6a42048eab349f9e7f5988 : SUCCESS in 9m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9dbe9320fc4541699ec3731fbd845387 : SUCCESS in 12m 14s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1405606179cc47368dbd5d266aac6f71 : SUCCESS in 55m 26s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/96b2a4e24fc64f3aacc3c8b0f5069e3d : SUCCESS in 59m 46s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14ff193fc8ec4d40bfed7f911a55bb4b : SUCCESS in 51m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/52eba75dd6924433af3093a3278287f7 : SUCCESS in 50m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1a0ba7ce69fe480ca7b39729382065d0 : SUCCESS in 17m 38s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/98dc6e49c5f84a91aed582cdd22e6387 : SUCCESS in 18m 03s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/abe7d0e529444a9098a55e5da0c81203 : SUCCESS in 13m 04s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e177b7ff832a452182858d3178584bc7 : SUCCESS in 1h 35m 09s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/568ea636c60d493b8c910f9c3ceff6d8 : SUCCESS in 59m 26s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/dbda75ca0e1c47c994a4e7e6dc646bf6 : RETRY_LIMIT in 4m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75eb40cea3a44c048d891190461162fe : RETRY_LIMIT in 3m 09s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ddb729aa9c85483db9be18ba7621c328 : RETRY_LIMIT in 3m 53s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4150365894e44162b4fcfea746d78c5c : SUCCESS in 25m 02s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31d3e4a9275b4dc7bcdca4c21745f1b2 : SUCCESS in 20m 02s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f8fc12829b2b451bb7697d81ce272034 : SUCCESS in 10m 43s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7812007987d54494a5b50f359d4b00e6 : SUCCESS in 10m 03s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/17ffa7b40688437c9afb30095296b820 : SUCCESS in 19m 53s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e279e46ceab8416689230b94aa5f0570 : SUCCESS in 12m 42s","accounts_in_message":[],"_revision_number":7},{"id":"85878e1777e7e6cf6fe535e96c5be19b9c1aa838","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-03-31 08:06:18.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5e31ecc6e213ad684c317594fd16679c4fad085c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 09:47:22.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ade6fd0b9e943d9867d51453a8facf3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd38357f45dd40b28681d7511183adac : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9de7ca5fdd4c4c4fae29b6225c5e1425 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5ba51c2f1e14fcbb34e803841291cf7 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/655f249a49154784b7c66e6363d1048b : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/307a88fc72854f2c80a746864a5cb5b1 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cebf9c1fbc641f4b58ea5d9f7f3117a : SUCCESS in 3m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b12808dc76c9470da27b8a884cfc1ae3 : SUCCESS in 4m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/35258fd579534f0f80f44b774bd5cc87 : SUCCESS in 10m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/10b44419597e4af793a337b2328a078c : SUCCESS in 13m 51s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7f05e35db9dd4c2890aff2f4195bb996 : FAILURE in 36m 05s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/51ed833b573248deb4fd9726bc0b462d : FAILURE in 41m 12s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f159308dbb04c03a3929f0348d4903a : FAILURE in 29m 40s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c3d4c6f086e54b84a808c7ff9584e862 : FAILURE in 39m 12s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/60cb57236b1745caaf5944e1b35bb0c0 : SUCCESS in 16m 59s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a9e1f00f446f4ea3ba1fbcfc67d21bd3 : SUCCESS in 17m 39s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/026285e0c4984970b143980b46629c22 : SUCCESS in 13m 27s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a8614d5dba8441c391f88aecb4fecdfc : FAILURE in 1h 33m 58s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e2dc16749a64799aa7a0c15e6cd613d : FAILURE in 1h 10m 16s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d680593f4439480db7756a298a9e238e : FAILURE in 9m 35s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/679af8d840824ccfa80861c84daaf684 : FAILURE in 6m 37s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/72d0327aab354e438204110ce27144a8 : FAILURE in 9m 27s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/72890462cfeb4d43a93e7dde84c4a858 : SUCCESS in 28m 46s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c0d4d0096744a75abdf215d727cdc1d : SUCCESS in 20m 27s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3e0be2723e3a43a5b004854dad923b7e : SUCCESS in 10m 44s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26b46fc7de114ae9961197801856e5fb : SUCCESS in 12m 31s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cc156df36d2340f9903c6b91a11c2cfb : SUCCESS in 17m 30s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f24f2aca9524a53ab94c4956f701cb0 : SUCCESS in 13m 10s","accounts_in_message":[],"_revision_number":8},{"id":"baceb5feef9fd971e350e6ff4379f589eee33f3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-04-03 08:32:05.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"77c8a7528738d6f0a8bb8ff18550f7efb41aa7cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-04-03 08:58:58.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"677914b79ae1572ed9125bce9f2f86576d28ef0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-04-03 09:28:02.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"1a53b08e3fc8de9e0378fcf2c2d86fcdba41b409","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-04-03 10:09:58.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"d09efb4cdf8b78cc44b0700aac0683eb85f0b3cd","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 10:42:45.000000000","message":"Patch Set 11:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11f6565b632d4f6699cc88c38a935ac5\n\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/ac1f2bfaca824f0fa67803367e49f2cf : SUCCESS in 29m 49s (non-voting)\n- kayobe-seed-images-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d7f4be588774e26b4cba706c7033113 : FAILURE in 14m 33s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"1026c2baf6c80a77877184176d43bb4bcc38e462","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-04-03 11:13:26.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"cb1359e71747c0380ca8ad4556c4a44076528276","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 11:21:23.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae1f7b3ae5c845109b5aa7b4e8b7df11\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eef33b5473b24eb981d66899d3bff07a : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b75118de6cf4a798ef23a01d7ce5ec3 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0d4371b27de4b33b09880bb862afe4a : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef1f72d0545d488faf6db3357ede6ebe : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6938a208c3548ab97c02aee98ba9a82 : SUCCESS in 8m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de04218d4ef447639a33c931d398e585 : SUCCESS in 3m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/74206c905bcf485da041d1ade4dce7dc : SUCCESS in 3m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ee159c28580a4b96942dc68c0636b430 : SUCCESS in 8m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b0c726b469b14276bdad90e115773a0b : SUCCESS in 11m 57s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/fa5b7ceb8a1a429cb9d43ba95235efec : SUCCESS in 55m 38s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/54eab4f075c94fc098997d879507631a : SUCCESS in 1h 03m 20s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7fba6d6a3524b16943cd3bfee13984c : SUCCESS in 42m 49s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/537b6298efe843499a8571ee7ffc2173 : SUCCESS in 1h 03m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/977d4d3c21744e69a0575b28148f1ed6 : SUCCESS in 16m 45s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/fa33db44977b4a38a157a9bfd9f026be : SUCCESS in 15m 16s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb0aa0a2dadb4f8db4b3b282d555625e : SUCCESS in 14m 17s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f99878ae393f4e3eb9d23b264d11a60b : SUCCESS in 1h 46m 26s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32ab1a1f18ac42debe83b4d4b9ac7677 : SUCCESS in 1h 16m 22s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/75bc1ab1bc364924a46ea07622153ce8 : SUCCESS in 18m 22s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03bf603b6c6c4057af978fd1923e1f49 : SUCCESS in 11m 49s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e7efeedc83034e53b386e42b517fc336 : SUCCESS in 32m 05s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/94b5ee475f504aa2bd1baf93af684560 : SUCCESS in 23m 43s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1bb85753d0054faaace61c549bddb3ff : SUCCESS in 19m 47s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5d40fa6d6aa043b18034ddf0692b1cce : SUCCESS in 13m 05s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e708e19a4844a82994c6ea69b9a6879 : SUCCESS in 9m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f257261d3d2c44a1a3fff2ddd92fd156 : SUCCESS in 14m 56s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92d8ad294acd44efb99fa982612f61a9 : SUCCESS in 10m 39s","accounts_in_message":[],"_revision_number":11},{"id":"d0332e9756f13dbc3506330b271fae2598b3fb11","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-13 17:08:38.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"f44529c0df9661d3fd0bb5beaa99604141c90a01","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-04-17 04:20:00.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"4655b98728247a50d11cbf4dd93267f8a58f0c2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-04-17 04:29:09.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"2eccbed7079df434e280a7fbf71932a16a43e089","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-04-17 04:32:24.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"bb24044691de3056b70ede1faa91ed0c385eac7f","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 05:11:00.000000000","message":"Patch Set 12:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c925292bd874452b9ba9eb28e81a1cef\n\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/abee867e7c24478daedcb866f887314b : SUCCESS in 37m 12s (non-voting)\n- kayobe-seed-images-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3d8411087484c8fa9f08017dbd1c0e0 : SUCCESS in 31m 47s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"8b80dbe1391a1e6d7815bfbc0db5e1f2b25d2728","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 06:17:11.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7380ecf3dab2420a8224c2d0679b3b36\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b29041fd6fd458084e23364ee11b885 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/809cdefb13fd4a86a9634fe160734432 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c47873e41049471984d41f400227b804 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e90b6123cd994c9e8b27991a925b295a : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8218065b88348bfaac2e6bfeb8d0aea : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f463acbc4b0b4453a993162ea40e0083 : SUCCESS in 3m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1c5760f687434403a13388c9a04f1fcf : SUCCESS in 3m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/88c4d098f703480a8137531bee400656 : SUCCESS in 10m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/44ec01eca4524bab97a5a235fe5e9863 : SUCCESS in 12m 42s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2bf4d14c97a5473e862182ef1b9ef111 : SUCCESS in 56m 23s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/89f3da321d8f458bb676073d4d1891a5 : SUCCESS in 48m 20s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41bb2275ea6e454d85e6e9e2fb75a87e : SUCCESS in 44m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6bb15b8e4646498eaad8b0bac859a089 : SUCCESS in 1h 04m 59s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a168f9969d77495d9da6c843d436d99b : SUCCESS in 16m 06s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/6898c841631f4a6a84bc5dbaab36e17d : SUCCESS in 15m 28s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/352c092f4eb442e8ace75c083c9b234a : SUCCESS in 14m 20s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b4ebd3655c094ae1bed959611f859262 : SUCCESS in 1h 42m 52s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/252effc10f9547a8be3d01f248367e8b : SUCCESS in 1h 00m 14s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/670d72cb9ff544058223853659a756ea : SUCCESS in 17m 13s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/336bb563c1784c28a120314da49b9bc0 : SUCCESS in 13m 43s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8531183619ea47f7bb6edf56fe1ce256 : SUCCESS in 33m 37s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bc2369cbae1144f0a9efb6888310456f : SUCCESS in 26m 48s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e4248afdb2444a2b317603f2a9bd00d : FAILURE in 15m 05s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b71f4be3b11c4505ab04e51e07f0e6f2 : SUCCESS in 13m 20s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07d150d392064b0d8cfacd1517df5764 : SUCCESS in 9m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/be83e56741a64df582989d8472beec1f : SUCCESS in 16m 54s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75ae7bb80df64956a5e8741d755b49e2 : SUCCESS in 12m 40s","accounts_in_message":[],"_revision_number":12},{"id":"a3d42499f132ff9a51b91c60874975daebdd6e12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-04-17 07:39:24.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"eb370a449bc902ecdb47c97ea6e49bc63e16f2b4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-04-17 09:05:52.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"bb8a7c9d99cc6d852dd10917fd13d490d590a615","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 09:46:44.000000000","message":"Patch Set 13:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e654703c68a48079814fdab6ea9eddf\n\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/3d4679b9f9734f8094da61a5f6360ec2 : SUCCESS in 26m 56s (non-voting)\n- kayobe-seed-images-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65aa691c5a354aad848b262969d46d4a : SUCCESS in 34m 12s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"f8a07a7f496c206bd5f5cbdfbf920894c91bc925","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-17 10:07:30.000000000","message":"Patch Set 13: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"50a489d0c958975b163a9d2f2ac874adf4b7d819","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-04-17 10:21:18.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":14},{"id":"b8fa034319b7157ccd8472b6d972f3e157580745","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-04-17 10:21:42.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"1892ff7116b6a893eb4a5e349d68ad8bfcee376f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-04-17 10:21:54.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"2ea2527945500a07645322710bd25d984fecc00e","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 12:01:31.000000000","message":"Patch Set 14:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce4574c87f9d4fb285b3b19f8d76a489\n\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/d4b41b95679c49df984805126cf88289 : SUCCESS in 54m 59s (non-voting)\n- kayobe-seed-images-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85a839fdf0594a7eb28e67ef399c9dc6 : TIMED_OUT in 1h 34m 27s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"530c1f8568c150c8ef28173e5a7b93bb891512d0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-04-17 12:02:55.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"6b6edbfc648ba06a72e3ff686c45412696a06e1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 12:16:06.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/841dcb5c37aa4ab18c532c8a21858db9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24cbec8b6294460eb18366611e79e27e : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e6c6832738c4969a6cf90a560d80190 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d60019af3ac54e81b26ba4b5185566ba : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a67bbeebf964298a4b5bb6289624aed : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d70c3378e8642bbacf0c57934994e16 : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a640d5bcada4a2caa40a95622b91809 : SUCCESS in 3m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d64c1701b43e4e8a82154fa0cebe7d99 : SUCCESS in 3m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/013c9aca22a34baf9d173d4d016955b0 : SUCCESS in 11m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1fadee2ffdd541789865634759ccff03 : SUCCESS in 14m 51s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9e79c65c1fcd49ddbe885db8ecec67a9 : SUCCESS in 1h 08m 49s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/deab632d69984a418e9d91d39c0ca527 : SUCCESS in 1h 07m 49s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d9d94d89e0344d89b567f5171baf9fd : SUCCESS in 1h 04m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1df7213e657b4849b0fb96480fa97026 : SUCCESS in 1h 05m 45s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/57bcad04212f4f0ab53c4465b4c215be : SUCCESS in 16m 16s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b5c5431cd5d640ed88a6e030cfc4f634 : SUCCESS in 16m 25s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/133a027f7103450a9b8a150f5ac5afe1 : SUCCESS in 15m 04s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b326282ac3a14d55940f88d62ef14816 : SUCCESS in 1h 48m 52s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79b165cc4c1245fe925c52f75f5575f9 : SUCCESS in 1h 24m 21s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/add7c1db435a4a7fb47f0cffdff0f81c : SUCCESS in 17m 33s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ffa49a6e4e9948759ab33e4dc3170c85 : SUCCESS in 16m 38s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/33c50d19121b43f99be3037c6fca5c12 : SUCCESS in 34m 52s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ff7aed7cebc2435aabf99d29a1800279 : SUCCESS in 27m 40s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/221449e29f894bd589b40bbaf9db492f : SUCCESS in 22m 14s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a6b29c34f1cb418e9946817e6ff0b328 : SUCCESS in 12m 44s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ceb9d8963ea9471a977d3a61bd5a0057 : SUCCESS in 10m 45s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0a52d83764e3448f8ff4cb5a920df2fc : SUCCESS in 17m 34s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e79c457209f4809b47242d59e6959d0 : SUCCESS in 14m 35s","accounts_in_message":[],"_revision_number":14},{"id":"605a324f3d47b40001b14e7b5ae1bcf8eb8bdcc4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-17 12:31:31.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"7302e8bc5ed920e7e3e3c625c2fa67efaddee6f5","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 12:51:33.000000000","message":"Patch Set 14:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca7a925208b741968605951a9b8cce04\n\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/5ab4a757d48f45559aba1382b4fca6d3 : SUCCESS in 27m 43s (non-voting)\n- kayobe-seed-images-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/467282d3058f485f9dc362be963ee9aa : SUCCESS in 40m 21s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"986fcf0a16ae6c0bb3bb4387266f849adab0f3d3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-17 13:11:10.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"05da5930a8025676688b07b81d78f5abb3b4035f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-17 13:14:41.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"cd8b39e3bf14332c78e2fc4b9c0bf7c5931062c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 13:15:12.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"16022320e2b93bc527c0374eaf33780b030c4624","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 15:10:29.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6010816a0d2c4b80b2f7e579769429e9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0a36f3d40d24117897557b6f1f7eda7 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef535e8ac3324889917cff04e4eab03f : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d1f59ddeb56447d95770230103d8a00 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e92f7e77751c4ff8a80d7e7f5dfeae55 : SUCCESS in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18e7f0df79e348728c65d0aa0caabe50 : SUCCESS in 2m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ad0a7ae46a1f4d5587794cdfdceb2bea : SUCCESS in 4m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4c9da474b27246f19280c72441ef93fc : SUCCESS in 10m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0fcc7ccbb5c846b989a841123ccf1bb7 : SUCCESS in 16m 00s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1b202b3bf27c42e295301dcc871ad62d : SUCCESS in 55m 46s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc6859720ade40769bf4bd47c9c89fbe : SUCCESS in 59m 45s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5bbcc4de66fe4cbeb02400729136a85e : SUCCESS in 1h 01m 29s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/de8d4719daec477cb2dc56ee3f8362f6 : SUCCESS in 21m 04s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be5f8ee899af4e2585723a7169ffa0e6 : SUCCESS in 14m 52s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4e69d62180e944d7aaef7b92345fcb04 : SUCCESS in 1h 48m 51s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b00188ce84b42e7a8863b6d19da01dd : SUCCESS in 1h 09m 26s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/aabf904985074592a05c4f0656332156 : SUCCESS in 21m 31s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c98aa91e887f4d32ad82c2d3f34217a6 : SUCCESS in 14m 08s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/963478086efe404898dc54b4ae2a636b : SUCCESS in 26m 49s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/267a7613ac594d089b6fdad04a19c831 : SUCCESS in 21m 58s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e44de406461545fabb6ee0b2ad410d16 : SUCCESS in 13m 11s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5969bc93a60b41898b5f1c0267305cac : SUCCESS in 10m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1a387f982145419eac8fa83114550d40 : SUCCESS in 16m 45s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbc8603f7bc846d98fc5b3d428ba614c : SUCCESS in 12m 56s","accounts_in_message":[],"_revision_number":14},{"id":"ceacbf929ac480bfb2ea7a50ef0546af08ef2f38","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 15:10:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"3bc0c0ea2726848e65cdf5896dbeed5034fba703","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 15:11:31.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a792a8ff7fdf45fdaa0a0fe06621d22f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2109a779f617475cb30e40dfd48e2a7d : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aabe21efff414e5785da0627594796c3 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"bdaeed184bf41163b9b410995ad021cee93fb39b","revisions":{"2724171765ea800944b19f22c3cf95125675813b":{"kind":"REWORK","_number":1,"created":"2023-01-16 10:19:55.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/22/870522/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/870522/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/870522/1"}}},"commit":{"parents":[{"commit":"aed2461eddca7b55b8452da116e437b0d38c6504","subject":"Merge \"Revert \"Use OpenStack Zed release\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aed2461eddca7b55b8452da116e437b0d38c6504"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-01-16 10:10:24.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-01-16 10:19:21.000000000","tz":0},"subject":"Build Ubuntu images with IPA when on Ubuntu","message":"Build Ubuntu images with IPA when on Ubuntu\n\nChange ``ipa_build_dib_elements_default`` and\n``ipa_build_dib_env_default`` to use ``os_distribution`` and\n``os_release``by default. This allows for Ubuntu images to be built when\nrunning on Ubuntu.\n\nRocky will still build CentOS images.\n\nChange-Id: Iefd2d0b7a3a3e07f5c112d58e2ec0b3da0a747d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2724171765ea800944b19f22c3cf95125675813b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2724171765ea800944b19f22c3cf95125675813b"}]},"branch":"refs/heads/master"},"8f9fc0ad0854aa60a932e5c34abcd6d0c060f526":{"kind":"REWORK","_number":2,"created":"2023-01-16 10:31:33.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/22/870522/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/870522/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/870522/2"}}},"commit":{"parents":[{"commit":"aed2461eddca7b55b8452da116e437b0d38c6504","subject":"Merge \"Revert \"Use OpenStack Zed release\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aed2461eddca7b55b8452da116e437b0d38c6504"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-01-16 10:10:24.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-01-16 10:30:04.000000000","tz":0},"subject":"Build Ubuntu images with IPA when on Ubuntu","message":"Build Ubuntu images with IPA when on Ubuntu\n\nChange ``ipa_build_dib_elements_default`` and\n``ipa_build_dib_env_default`` to use ``os_distribution`` and\n``os_release``by default. This allows for Ubuntu images to be built when\nrunning on Ubuntu.\n\nRocky will still build CentOS images, as they still need to be tested\nbeing built with IPA.\n\nChange-Id: Iefd2d0b7a3a3e07f5c112d58e2ec0b3da0a747d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f9fc0ad0854aa60a932e5c34abcd6d0c060f526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f9fc0ad0854aa60a932e5c34abcd6d0c060f526"}]},"branch":"refs/heads/master"},"940d984be7e6bb8f4fde9cd39e41cd3fe2c56925":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-01-16 11:34:06.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/22/870522/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/870522/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/870522/3"}}},"commit":{"parents":[{"commit":"aed2461eddca7b55b8452da116e437b0d38c6504","subject":"Merge \"Revert \"Use OpenStack Zed release\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aed2461eddca7b55b8452da116e437b0d38c6504"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-01-16 10:10:24.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-01-16 11:32:57.000000000","tz":0},"subject":"Build Ubuntu images with IPA when on Ubuntu","message":"Build Ubuntu images with IPA when on Ubuntu\n\nChange ``ipa_build_dib_elements_default`` and\n``ipa_build_dib_env_default`` to use ``os_distribution`` and\n``os_release`` by default. This allows for Ubuntu images to be built\nwhen running on Ubuntu.\n\nRocky will still build CentOS images, as Rocky IPA images have not been\ntested yet.\n\nChange-Id: Iefd2d0b7a3a3e07f5c112d58e2ec0b3da0a747d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/940d984be7e6bb8f4fde9cd39e41cd3fe2c56925"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/940d984be7e6bb8f4fde9cd39e41cd3fe2c56925"}]},"branch":"refs/heads/master"},"be007c42afe0d40a006c56715292af34f8c51221":{"kind":"REWORK","_number":4,"created":"2023-01-16 12:20:28.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/22/870522/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/870522/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/870522/4"}}},"commit":{"parents":[{"commit":"aed2461eddca7b55b8452da116e437b0d38c6504","subject":"Merge \"Revert \"Use OpenStack Zed release\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aed2461eddca7b55b8452da116e437b0d38c6504"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-01-16 10:10:24.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-01-16 12:20:14.000000000","tz":0},"subject":"Build Ubuntu images with IPA when on Ubuntu","message":"Build Ubuntu images with IPA when on Ubuntu\n\nChange ``ipa_build_dib_elements_default`` and\n``ipa_build_dib_env_default`` to use ``os_distribution`` and\n``os_release`` by default. This allows for Ubuntu images to be built\nwhen running on Ubuntu.\n\nRocky will still build CentOS images, as Rocky IPA images have not been\ntested yet.\n\nChange-Id: Iefd2d0b7a3a3e07f5c112d58e2ec0b3da0a747d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be007c42afe0d40a006c56715292af34f8c51221"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be007c42afe0d40a006c56715292af34f8c51221"}]},"branch":"refs/heads/master"},"fb92b7bc396217cdeab3bfac9c08490cf1f99845":{"kind":"REWORK","_number":5,"created":"2023-01-16 13:48:08.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/22/870522/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/870522/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/870522/5"}}},"commit":{"parents":[{"commit":"aed2461eddca7b55b8452da116e437b0d38c6504","subject":"Merge \"Revert \"Use OpenStack Zed release\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aed2461eddca7b55b8452da116e437b0d38c6504"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-01-16 10:10:24.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-01-16 13:47:54.000000000","tz":0},"subject":"Build Ubuntu images with IPA when on Ubuntu","message":"Build Ubuntu images with IPA when on Ubuntu\n\nChange ``ipa_build_dib_elements_default`` and\n``ipa_build_dib_env_default`` to use ``os_distribution`` and\n``os_release`` by default. This allows for Ubuntu images to be built\nwhen running on Ubuntu.\n\nRocky will still build CentOS images, as Rocky IPA images have not been\ntested yet.\n\nChange-Id: Iefd2d0b7a3a3e07f5c112d58e2ec0b3da0a747d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb92b7bc396217cdeab3bfac9c08490cf1f99845"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb92b7bc396217cdeab3bfac9c08490cf1f99845"}]},"branch":"refs/heads/master"},"ca7e215cc0ef99fd906c28c2f9c90a91bac97d07":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-03-30 14:52:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/22/870522/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/870522/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/870522/6"}}},"commit":{"parents":[{"commit":"137e64b57c86fde852b49f9975ffbca26da79de1","subject":"Fix opensearch container builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/137e64b57c86fde852b49f9975ffbca26da79de1"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-01-16 10:10:24.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-03-30 14:52:49.000000000","tz":0},"subject":"Build Ubuntu images with IPA when on Ubuntu","message":"Build Ubuntu images with IPA when on Ubuntu\n\nChange ``ipa_build_dib_elements_default`` and\n``ipa_build_dib_env_default`` to use ``os_distribution`` and\n``os_release`` by default. This allows for Ubuntu images to be built\nwhen running on Ubuntu.\n\nRocky will still build CentOS images, as Rocky IPA images have not been\ntested yet.\n\nChange-Id: Iefd2d0b7a3a3e07f5c112d58e2ec0b3da0a747d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca7e215cc0ef99fd906c28c2f9c90a91bac97d07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca7e215cc0ef99fd906c28c2f9c90a91bac97d07"}]},"branch":"refs/heads/master","description":"Rebase"},"de9f7eb80b7f6056e7fe8ba7299a824616698dd2":{"kind":"REWORK","_number":7,"created":"2023-03-30 19:21:24.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/22/870522/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/870522/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/870522/7"}}},"commit":{"parents":[{"commit":"137e64b57c86fde852b49f9975ffbca26da79de1","subject":"Fix opensearch container builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/137e64b57c86fde852b49f9975ffbca26da79de1"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-03-30 19:21:08.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-03-30 19:21:08.000000000","tz":120},"subject":"Build Ubuntu images with IPA when on Ubuntu","message":"Build Ubuntu images with IPA when on Ubuntu\n\nChange ``ipa_build_dib_elements_default`` and\n``ipa_build_dib_env_default`` to use ``os_distribution`` and\n``os_release`` by default. This allows for Ubuntu images to be built\nwhen running on Ubuntu.\n\nRocky will still build CentOS images, as Rocky IPA images have not been\ntested yet.\n\nChange-Id: Iefd2d0b7a3a3e07f5c112d58e2ec0b3da0a747d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/de9f7eb80b7f6056e7fe8ba7299a824616698dd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/de9f7eb80b7f6056e7fe8ba7299a824616698dd2"}]},"branch":"refs/heads/master"},"2ada117d7ced8d6d53328ed003748a056f3c3aec":{"kind":"REWORK","_number":8,"created":"2023-03-31 08:06:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/22/870522/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/870522/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/870522/8"}}},"commit":{"parents":[{"commit":"137e64b57c86fde852b49f9975ffbca26da79de1","subject":"Fix opensearch container builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/137e64b57c86fde852b49f9975ffbca26da79de1"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-03-30 19:20:27.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-03-31 08:06:12.000000000","tz":120},"subject":"Build Ubuntu images with IPA when on Ubuntu","message":"Build Ubuntu images with IPA when on Ubuntu\n\nChange ``ipa_build_dib_elements_default`` and\n``ipa_build_dib_env_default`` to use ``os_distribution`` and\n``os_release`` by default. This allows for Ubuntu images to be built\nwhen running on Ubuntu.\n\nRocky will still build CentOS images, as Rocky IPA images have not been\ntested yet.\n\nChange-Id: Iefd2d0b7a3a3e07f5c112d58e2ec0b3da0a747d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2ada117d7ced8d6d53328ed003748a056f3c3aec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2ada117d7ced8d6d53328ed003748a056f3c3aec"}]},"branch":"refs/heads/master"},"38fcd8b058dccabf79a46c9bb3afe9168d541395":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-04-03 08:32:05.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/22/870522/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/870522/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/870522/9"}}},"commit":{"parents":[{"commit":"e5815017b5df5d7cb7e747a09da9979f50f2f024","subject":"Move to Launchpad for project management","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e5815017b5df5d7cb7e747a09da9979f50f2f024"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-01-16 10:10:24.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-04-03 08:31:47.000000000","tz":120},"subject":"Build Ubuntu images with IPA when on Ubuntu","message":"Build Ubuntu images with IPA when on Ubuntu\n\nChange ``ipa_build_dib_elements_default`` and\n``ipa_build_dib_env_default`` to use ``os_distribution`` and\n``os_release`` by default. This allows for Ubuntu images to be built\nwhen running on Ubuntu.\n\nRocky will still build CentOS images, as Rocky IPA images have not been\ntested yet.\n\nChange-Id: Iefd2d0b7a3a3e07f5c112d58e2ec0b3da0a747d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38fcd8b058dccabf79a46c9bb3afe9168d541395"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38fcd8b058dccabf79a46c9bb3afe9168d541395"}]},"branch":"refs/heads/master"},"9b46dba5ef06c9c041beec0ca042a09321cdddec":{"kind":"REWORK","_number":10,"created":"2023-04-03 08:58:58.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/22/870522/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/870522/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/870522/10"}}},"commit":{"parents":[{"commit":"e5815017b5df5d7cb7e747a09da9979f50f2f024","subject":"Move to Launchpad for project management","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e5815017b5df5d7cb7e747a09da9979f50f2f024"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-01-16 10:10:24.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-04-03 08:58:53.000000000","tz":120},"subject":"Build Ubuntu images with IPA when on Ubuntu","message":"Build Ubuntu images with IPA when on Ubuntu\n\nChange ``ipa_build_dib_elements_default`` and\n``ipa_build_dib_env_default`` to use ``os_distribution`` and\n``os_release`` by default. This allows for Ubuntu images to be built\nwhen running on Ubuntu.\n\nRocky will still build CentOS images, as Rocky IPA images have not been\ntested yet.\n\nChange-Id: Iefd2d0b7a3a3e07f5c112d58e2ec0b3da0a747d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b46dba5ef06c9c041beec0ca042a09321cdddec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b46dba5ef06c9c041beec0ca042a09321cdddec"}]},"branch":"refs/heads/master"},"373d4e01dab598cdfd3b268fc4bbc1a80b4bd155":{"kind":"REWORK","_number":11,"created":"2023-04-03 09:28:02.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/22/870522/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/870522/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/870522/11"}}},"commit":{"parents":[{"commit":"e5815017b5df5d7cb7e747a09da9979f50f2f024","subject":"Move to Launchpad for project management","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e5815017b5df5d7cb7e747a09da9979f50f2f024"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-01-16 10:10:24.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-04-03 09:27:55.000000000","tz":120},"subject":"Build Ubuntu images with IPA when on Ubuntu","message":"Build Ubuntu images with IPA when on Ubuntu\n\nChange ``ipa_build_dib_elements_default`` and\n``ipa_build_dib_env_default`` to use ``os_distribution`` and\n``os_release`` by default. This allows for Ubuntu images to be built\nwhen running on Ubuntu.\n\nRocky will still build CentOS images, as Rocky IPA images have not been\ntested yet.\n\nChange-Id: Iefd2d0b7a3a3e07f5c112d58e2ec0b3da0a747d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/373d4e01dab598cdfd3b268fc4bbc1a80b4bd155"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/373d4e01dab598cdfd3b268fc4bbc1a80b4bd155"}]},"branch":"refs/heads/master"},"50487d5d11ddc904d057be7e5c3e3e3f7d393295":{"kind":"REWORK","_number":12,"created":"2023-04-17 04:29:09.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/22/870522/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/870522/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/870522/12"}}},"commit":{"parents":[{"commit":"51a9019c6b8a35c3576b28d9d755341768697c66","subject":"Merge \"Fix merge action plugins verbose output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51a9019c6b8a35c3576b28d9d755341768697c66"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-01-16 10:10:24.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-04-17 04:28:53.000000000","tz":120},"subject":"Build Ubuntu images with IPA when on Ubuntu","message":"Build Ubuntu images with IPA when on Ubuntu\n\nChange ``ipa_build_dib_elements_default`` and\n``ipa_build_dib_env_default`` to use ``os_distribution`` and\n``os_release`` by default. This allows for Ubuntu images to be built\nwhen running on Ubuntu.\n\nRocky will still build CentOS images, as Rocky IPA images have not been\ntested yet.\n\nChange-Id: Iefd2d0b7a3a3e07f5c112d58e2ec0b3da0a747d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/50487d5d11ddc904d057be7e5c3e3e3f7d393295"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/50487d5d11ddc904d057be7e5c3e3e3f7d393295"}]},"branch":"refs/heads/master"},"41f87b29a461f4e615752f94b7d0f29858c6b6cd":{"kind":"REWORK","_number":13,"created":"2023-04-17 07:39:24.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/22/870522/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/870522/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/870522/13"}}},"commit":{"parents":[{"commit":"51a9019c6b8a35c3576b28d9d755341768697c66","subject":"Merge \"Fix merge action plugins verbose output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51a9019c6b8a35c3576b28d9d755341768697c66"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-01-16 10:10:24.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-04-17 07:39:18.000000000","tz":120},"subject":"Build Ubuntu images with IPA when on Ubuntu","message":"Build Ubuntu images with IPA when on Ubuntu\n\nChange ``ipa_build_dib_elements_default`` and\n``ipa_build_dib_env_default`` to use ``os_distribution`` and\n``os_release`` by default. This allows for Ubuntu images to be built\nwhen running on Ubuntu.\n\nRocky will still build CentOS images, as Rocky IPA images have not been\ntested yet.\n\nChange-Id: Iefd2d0b7a3a3e07f5c112d58e2ec0b3da0a747d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41f87b29a461f4e615752f94b7d0f29858c6b6cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41f87b29a461f4e615752f94b7d0f29858c6b6cd"}]},"branch":"refs/heads/master"},"bdaeed184bf41163b9b410995ad021cee93fb39b":{"kind":"REWORK","_number":14,"created":"2023-04-17 10:21:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/22/870522/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/870522/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/870522/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/870522/14"}}},"commit":{"parents":[{"commit":"51a9019c6b8a35c3576b28d9d755341768697c66","subject":"Merge \"Fix merge action plugins verbose output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51a9019c6b8a35c3576b28d9d755341768697c66"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-01-16 10:10:24.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-04-17 10:20:58.000000000","tz":120},"subject":"Build Ubuntu images with IPA when on Ubuntu","message":"Build Ubuntu images with IPA when on Ubuntu\n\nChange ``ipa_build_dib_elements_default`` and\n``ipa_build_dib_env_default`` to use ``os_distribution`` and\n``os_release`` by default. This allows for Ubuntu images to be built\nwhen running on Ubuntu.\n\nRocky will still build CentOS images, as Rocky IPA images have not been\ntested yet.\n\nChange-Id: Iefd2d0b7a3a3e07f5c112d58e2ec0b3da0a747d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bdaeed184bf41163b9b410995ad021cee93fb39b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bdaeed184bf41163b9b410995ad021cee93fb39b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
