)]}'
{"id":"openstack%2Fkayobe~870546","triplet_id":"openstack%2Fkayobe~master~I640207dd9e987cb6d5743e22cae6039a65d10566","project":"openstack/kayobe","branch":"master","topic":"kolla_assert","attention_set":{},"removed_from_attention_set":{"25600":{"account":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"last_update":"2023-02-21 16:39:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I640207dd9e987cb6d5743e22cae6039a65d10566","subject":"vips: assert shows what we\u0027re testing and when","status":"MERGED","created":"2023-01-16 14:16:05.000000000","updated":"2023-02-21 16:40:09.000000000","submitted":"2023-02-21 16:39:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870546-kolla_assert","meta_rev_id":"22658b8571de98f9894e2884ecccaf2ed13b0c98","_number":870546,"virtual_id_number":870546,"owner":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-21 16:39:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":1,"date":"2023-02-01 21:57:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":2,"date":"2023-02-13 07:53:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-02-21 14:25:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-02-21 14:25:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-16 15:17:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-13 07:53:36.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-02-21 14:25:22.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"445151b40b0799069ee3874897cf05cfc7f97e99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-01-16 14:16:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da3a6cac9bb6c2335adae1225e83590ebaa30ba2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 15:17:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e43199350aee409cb3d2c93469255f69\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57f05f24ccf941e58013deda758427ff : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e583a0e26a464726adcbbba06f3e7f21 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fadfa9d495444bdb6ba0a4f60902170 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f2f716bef75422cb30ee795991ec961 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef2b69752caa43d785a490cfcb0c1d67 : SUCCESS in 7m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3da2c871f642492fbbd2a4966709c606 : FAILURE in 4m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4931d908abc6436fad8d60ca94e197d0 : SUCCESS in 15m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ce426a30129c490488dd9f4de365f44d : SUCCESS in 13m 46s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9bb3e4dc3642407c81c50ae11dbc3e28 : FAILURE in 10m 24s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/66141de45d9544ebbec9b06d86c862df : FAILURE in 12m 24s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4dab7a3e31f84476917c57d3d7cf4320 : FAILURE in 8m 42s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e4bc8860caa140f69c8ef3b29de4ff81 : FAILURE in 9m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/bd8b074d153c471da27289a26d87369c : FAILURE in 10m 28s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/c23d9d7e15c141ac96b7d8be4532d4de : FAILURE in 10m 32s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efa2d923ff024c15ba8f109a5a56dea9 : FAILURE in 9m 14s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8badb76bf1bf4da6a77233f823fc5205 : FAILURE in 57m 59s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/dc141d4cf1ed40d69648b6b599cd2797 : FAILURE in 9m 58s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b9ccfadcec7414fbbfdee0d2a8e3e66 : FAILURE in 9m 59s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8e65ea61c5c64c1293a3c3c38035628a : FAILURE in 9m 35s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33d9d45fbd0f48e4b0235c5dcd715179 : FAILURE in 26m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d4f7613eee1f4d6b94ef8cf50489c32b : FAILURE in 10m 08s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3507c264b8c54173b0fd9ef983d15db3 : FAILURE in 8m 04s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a3c4e73ec8b54f1c8de1f2db457ff286 : FAILURE in 9m 46s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/746ed448ac0b4c789e115c6844c4b06e : FAILURE in 8m 22s","accounts_in_message":[],"_revision_number":1},{"id":"a49d8e39b5c872020a423f47f09ff0200a4329c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-01-16 15:32:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c72d9c20cd2ae616fc21d654d1bb5f5e6ef6fe90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 17:10:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/891b100ece7d455282c3249f63c6b0e5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f27be91aaae040ae8c5870bcca2c33d3 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5045ef70a61f4a8db0952bbc18679bd4 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/851b0a875c4f497687117dc9a4c00aa6 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8d51a9e1d9194f4c8bdf80d4dbbda18f : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30f513b6441d496c865a51d5c7c3412e : SUCCESS in 7m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b12c47c6c3554503a67f2118794fa33a : SUCCESS in 4m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cc1be8c63c5945638456914299238333 : SUCCESS in 11m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9888fa7079d640bbabd00f645e2028ee : SUCCESS in 15m 14s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a64c8dce4ec1484ab0b50b14f5d1d138 : SUCCESS in 57m 30s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/2f96dd2e01e2493583b1b50c65129330 : SUCCESS in 1h 06m 50s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5fcad17869884992b131ddaaad027507 : SUCCESS in 44m 39s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/214c789d75274291bbb3a9641779e572 : SUCCESS in 1h 00m 26s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6b85b6fb64314d6698dbd5b886ce049d : SUCCESS in 16m 57s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d313906828be4270b1328820cf941626 : SUCCESS in 17m 52s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93e0211bfd114f7b87a9f73327a2a67a : SUCCESS in 13m 24s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6ad8a0b4a574ecd9198e4b99d81fba3 : SUCCESS in 1h 15m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c0328e5faba54c50a3a622ac31279bb5 : SUCCESS in 17m 03s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/177314a3f6c44fec8b7d1934cced0950 : SUCCESS in 14m 22s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/5c20d793015545bab036729bd5a86fbb : FAILURE in 22m 35s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e01452fcbd65401a8efdccf7d3824263 : SUCCESS in 30m 45s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/111fb2bbbef14565a91834297e6c35eb : SUCCESS in 16m 06s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/271a5f933aef47468700e1ffbf9fa151 : SUCCESS in 10m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0b611cf9fb394740b58aaa36019c55b2 : SUCCESS in 18m 31s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4be16c7e0fe3471586e3c0ac1d710557 : SUCCESS in 12m 34s","accounts_in_message":[],"_revision_number":2},{"id":"15fda7953f29bfb5b04c9581206e6300b550b60f","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-02-01 21:57:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7b5b454ea732a45c7c0119059443589bf7da760a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-13 07:53:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"97dc6786b64f087c57c549b09cf73e3e00cff3ec","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-16 16:02:18.000000000","message":"Topic assert removed","accounts_in_message":[],"_revision_number":2},{"id":"3a83927682158458e063190ea435c85e584ce547","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-16 16:02:24.000000000","message":"Topic set to kolla_assert","accounts_in_message":[],"_revision_number":2},{"id":"4a238b65991310427de72ce7a2f3ed5a1e18a945","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-21 14:25:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ddc4a44575bbbeba4f22bc2d75e2254a1e4b6fc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 14:25:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"90ae5f79060acafe50e00dbda32e96bea4807d84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 16:39:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba1e6ae02f2c4ae589395de50c016b80\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6f16694d91a4103a13112960a041643 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc422c879dc4497fb1308dcc2a30ef1f : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7862ed0a8c6490b92013666227eaec5 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/727dfe0422b74df69a581191bb8d53f6 : SUCCESS in 7m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f728f9e5b41740aaa4541d09742f99b0 : SUCCESS in 3m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/917918bcd4b84048b0971560bb3f4ab2 : SUCCESS in 10m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6bf44149da3f49a08a79c8e02eb18801 : SUCCESS in 14m 17s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3541eb0e12fd40a79318fffb537b15bf : SUCCESS in 52m 20s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a70468babf8f454596cc2405669181a6 : SUCCESS in 47m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/894e527a102e47a5874ef6df2c65477c : SUCCESS in 1h 03m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/bc2e640bcfe243a2aecf897e40a24f50 : SUCCESS in 14m 40s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4526a2913d984fa8ac2cbebb89d7243b : SUCCESS in 15m 40s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0f63fac1f7a143f7bb56602c4d961350 : SUCCESS in 1h 35m 01s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9c31b2a3af6418eaba9f8362b0cddb5 : SUCCESS in 1h 06m 52s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ad4592682d15434ba7d934bbb3a08070 : SUCCESS in 15m 29s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b807aced651344acb6829be2b3cbc901 : SUCCESS in 14m 31s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/792a77ddddc846678cfb6de22ac72f79 : SUCCESS in 35m 10s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b8213c67eb241fd89e9df11bf759c8a : SUCCESS in 29m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b5659338769a45ceaad6487f6b883f7d : SUCCESS in 12m 44s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0372e397cbb2411db0b767277c6d77cc : SUCCESS in 10m 35s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/77f2362310e74cb4b923f037a825868f : SUCCESS in 17m 05s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d506721701f840c9a8af6b99f54e1e8e : SUCCESS in 15m 17s","accounts_in_message":[],"_revision_number":2},{"id":"657e278b1292bf7b204bbb5d136f2209c3c30fe8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 16:39:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"22658b8571de98f9894e2884ecccaf2ed13b0c98","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 16:40:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6df6378793664f8fbc25407a8965a979\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88d18ea322c147ebaf739a10ec3f12d5 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b7c8c30bb3c8d492354f809a088316d0c257c8c6","revisions":{"2355384a9e89425d10873c48a395fcd927175004":{"kind":"REWORK","_number":1,"created":"2023-01-16 14:16:05.000000000","uploader":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"ref":"refs/changes/46/870546/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/46/870546/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/870546/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/870546/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/870546/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/46/870546/1"}}},"commit":{"parents":[{"commit":"aed2461eddca7b55b8452da116e437b0d38c6504","subject":"Merge \"Revert \"Use OpenStack Zed release\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aed2461eddca7b55b8452da116e437b0d38c6504"}]}],"author":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2023-01-16 14:11:53.000000000","tz":60},"committer":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2023-01-16 14:14:47.000000000","tz":60},"subject":"vips: assert shows what we\u0027re testing and when","message":"vips: assert shows what we\u0027re testing and when\n\nassert will also fail when we\u0027re not meeting the conditions, makes\nclear what we\u0027re actually testing, and isn\u0027t listed as a skipped task\nwhen the condition is ok.\n\nChange-Id: I640207dd9e987cb6d5743e22cae6039a65d10566\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2355384a9e89425d10873c48a395fcd927175004"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2355384a9e89425d10873c48a395fcd927175004"}]},"branch":"refs/heads/master"},"b7c8c30bb3c8d492354f809a088316d0c257c8c6":{"kind":"REWORK","_number":2,"created":"2023-01-16 15:32:55.000000000","uploader":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"ref":"refs/changes/46/870546/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/46/870546/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/870546/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/870546/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/870546/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/46/870546/2"}}},"commit":{"parents":[{"commit":"aed2461eddca7b55b8452da116e437b0d38c6504","subject":"Merge \"Revert \"Use OpenStack Zed release\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aed2461eddca7b55b8452da116e437b0d38c6504"}]}],"author":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2023-01-16 14:11:53.000000000","tz":60},"committer":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2023-01-16 15:32:17.000000000","tz":60},"subject":"vips: assert shows what we\u0027re testing and when","message":"vips: assert shows what we\u0027re testing and when\n\nassert will also fail when we\u0027re not meeting the conditions, makes\nclear what we\u0027re actually testing, and isn\u0027t listed as a skipped task\nwhen the condition is ok.\n\nChange-Id: I640207dd9e987cb6d5743e22cae6039a65d10566\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b7c8c30bb3c8d492354f809a088316d0c257c8c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b7c8c30bb3c8d492354f809a088316d0c257c8c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
