)]}'
{"id":"openstack%2Fkayobe~872638","triplet_id":"openstack%2Fkayobe~master~I09076ff9daf3e45f2c91f57319c661278aa2f9ea","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-02-08 09:51:39.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I09076ff9daf3e45f2c91f57319c661278aa2f9ea","subject":"CI: lower memory for baremetal compute test","status":"ABANDONED","created":"2023-02-03 10:09:29.000000000","updated":"2023-02-08 09:51:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d5daec438b5335c6c537582ccca9cf51d3f3cf54","_number":872638,"virtual_id_number":872638,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-02-07 16:08:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-03 12:36:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fadfe198267208f59581f3988394744ffc6b82e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-03 10:09:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"421c4e803eccde260b8dfe44d62646e13c037dca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-03 10:16:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ac02d98ac5afd7b7d9b9d6b2b926d60dce15b67b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 12:36:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/77f3f69217b243aebc85c8edf7b30c87\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36cf38feaa454769b1640615cd74a4b3 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e20ab1135e447e39749c646755524e8 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6383453767494eb297ed2d3082dad330 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/51dc5a70c2f84d5a87004aa73e59fa02 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2469b9d793642e18dd83cce32d05e8d : SUCCESS in 6m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6df6f898b781433b84a3c426ba6c1c64 : SUCCESS in 4m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/780dec1e9d9d4a3dbc7cefa859fac120 : SUCCESS in 12m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ee16d0145cbb49f7b99343f1b9985e59 : SUCCESS in 13m 39s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/971b859bd6134fcf97ecf2eb04bf2bdb : FAILURE in 1h 50m 24s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/851dcd58bbd4472da7168e9e2d7507bf : FAILURE in 1h 57m 09s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/860fec7dd798482896d4a97b3ca48d0e : FAILURE in 16m 40s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/74da953dd90d4546bed1bbdafb3e385f : FAILURE in 1h 58m 18s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0a1c7edcd8614efe94a7d86e07299ba8 : SUCCESS in 20m 13s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/6cfefb8ac6764e0abe44bc65b7414239 : SUCCESS in 20m 24s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/507e4157e6694584a8dd41671ec4fb83 : FAILURE in 13m 17s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0fe152ea75ec42fba33e9ec6c9aba9a5 : FAILURE in 2h 02m 43s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/869173b7e78849859bad0b4280dde5cb : FAILURE in 18m 55s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/dee89f2042f54d9280aff41ff273afe5 : SUCCESS in 15m 19s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ebe6180bcddd40be8a40e32b4ea24a77 : SUCCESS in 20m 09s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/09157515d9b346078f637a7a522cfed4 : SUCCESS in 33m 57s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5d2c38a8b45f4b8981cc05ee2940468b : SUCCESS in 28m 10s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c175e09336fa4810a75aa924c30511f0 : SUCCESS in 52m 04s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2e2e5b62aca84b13a13c85f41fb9098b : SUCCESS in 14m 51s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e81dc4b0e69a4c52bd1895846831b89a : SUCCESS in 10m 14s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3f836f0052634642886d9532249efc4c : SUCCESS in 17m 42s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb5eaf5cb9cb4524834681d2fe3d717a : SUCCESS in 13m 27s","accounts_in_message":[],"_revision_number":2},{"id":"de9fd286b05ebce02afc5aff0d80a2fc0cd35354","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-07 10:48:43.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"e42fa9d83e787a7f253b47da3bb6dbf2aeb1fd38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 11:12:33.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/872647 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"a445672464402fde3836e0d11963671c2ed2b2b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-07 11:38:49.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"5abdbab7854ee7f75a9452660bff64ddc7caf361","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 13:35:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5b065152c5142fba3293c6e927e37fc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8305ed0154d413c809ee5395e740597 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86ddf7847d0f4911aa3087f7a21e2216 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cde2f4918c7469eb14f71fd95fe0d8c : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/077f107356e346a58f86983a88487919 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61f7eb2a5ed443ed8765dc6369df8cce : SUCCESS in 7m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0d91b893b6d34d45998cc0f935b624fd : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1e6067a58d7c402daa86cc7df736df42 : SUCCESS in 10m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4ee09221104f4ef2a6937fb5158166ca : SUCCESS in 13m 49s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ec2d7d0ae8504b11bb9d1f28ec5cd9b4 : FAILURE in 1h 54m 25s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/bbe8cebe86cd4d27858df22315628571 : FAILURE in 1h 49m 46s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c121c9dc6cb43beb3fea84cabd74efb : FAILURE in 14m 22s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7c809e45de234c26a0743446017cd344 : FAILURE in 1h 49m 08s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/135e788d225642d3b610aa90e92d170b : SUCCESS in 17m 00s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/6a9c1b909d6f4686a89e6623eadd402d : SUCCESS in 18m 08s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d457023da44419081e72de84176a86d : FAILURE in 12m 48s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8c89fde29f0d4b09986cbfaa8165d348 : FAILURE in 56m 05s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21bf490353e64e67a3193cb10827ca9f : FAILURE in 14m 11s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5d8d3a3ac8ca4e96be63197ff3160bf2 : SUCCESS in 18m 15s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c3311995c6e468982185cd541c0dc19 : SUCCESS in 16m 04s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6e1c7b1a29334e0cb543a236fa633f38 : SUCCESS in 33m 06s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6ad3170fa261475cb5525107f5ac642c : SUCCESS in 33m 31s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64a113bff5684de9897d3e8f72c27f77 : SUCCESS in 30m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a90bbe3fb9df402d9b17c3bb148efeb7 : SUCCESS in 12m 43s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef4d727d9d6f41249f18d10035e83c81 : SUCCESS in 10m 45s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2f571a60db5f474d873d4f53458a6655 : SUCCESS in 16m 57s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a543a38895c444bea7d550bb9acb2537 : FAILURE in 15m 17s","accounts_in_message":[],"_revision_number":4},{"id":"644af6b56c7fd9d53106e9e6dd6be3f3ecbaf60e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-07 14:45:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"27a81059c09cb9030a90094f97b5a60b7d8fa25b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 16:08:18.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d48282e79c04e8186fdb690f762a0dd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abee6f165554421ba4c187ae9d8b97b9 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd1d0ef43f2b44cf87cd308ad028e50e : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a2eaecc01d5401a9c6ad237682cd676 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b11bddc418a34251afd9b6f36e891adc : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf9c5eb27100475aa4a295bfacc92d12 : SUCCESS in 9m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/df9c713442104c47a36952cf037a85ae : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dad75865a44f4b99b3306a6b85a8cdb6 : SUCCESS in 14m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d526ef2a075142898bb386e66d767c15 : SUCCESS in 14m 21s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c1b19a2d0b474c8b91d434d9857c4072 : FAILURE in 50m 32s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/96034d30a01e4d5ca1968bb4e40d5e1c : FAILURE in 59m 01s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02135f6c5e4542018a0bff18c89118a1 : SUCCESS in 53m 15s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/eda85469d9d54bb69253545fdec84585 : SUCCESS in 1h 13m 09s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/01f951c2e3b94adaa5031cd111228554 : SUCCESS in 18m 18s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/dfedb25c90d64a96bbb51033dcd53dc4 : SUCCESS in 18m 33s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee90598b7dcd4ba8824dbe91e43ad415 : SUCCESS in 13m 51s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/532275be8ea24d89a615c41636611e04 : SUCCESS in 1h 20m 40s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef44e3d97d514aaaa359b1adc036b531 : FAILURE in 24m 09s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d011f01c1ddf44a99b4154758721a6e9 : SUCCESS in 18m 38s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c9995b4ed07464b89d53d765da36b57 : SUCCESS in 12m 55s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/3d3d3ad12248435ab1fceaa0d0a09a29 : SUCCESS in 40m 47s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7855c30f0e2d4b5ca3c8e4b2676af92f : SUCCESS in 33m 50s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1911e0c459d43c4b36e2f73940017af : SUCCESS in 37m 13s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ef0608c79635416e9399e3710674ffff : SUCCESS in 13m 49s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/023d43ad8f4d4c979af20ca5ef901946 : SUCCESS in 11m 59s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3ca00afa24524496a5fb747d02b51084 : SUCCESS in 16m 21s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/600cd53a61064981b0fe47288b2eb0f8 : SUCCESS in 16m 04s","accounts_in_message":[],"_revision_number":5},{"id":"d5daec438b5335c6c537582ccca9cf51d3f3cf54","tag":"autogenerated:gerrit:abandon","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-08 09:51:39.000000000","message":"Abandoned\n\nOOM issues fixed with https://review.opendev.org/c/openstack/kayobe/+/872981","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3927df091a7e249f18bbe82f18208ac0d7fb27c0","revisions":{"83ca4bf1a0c85b94bc0129051fd310b88caa695f":{"kind":"REWORK","_number":1,"created":"2023-02-03 10:09:29.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/872638/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/872638/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/872638/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/872638/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/872638/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/872638/1"}}},"commit":{"parents":[{"commit":"094852e3c6d4b0652abf620368e0a1251d624708","subject":"CI: add Rocky9 upgrade jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/094852e3c6d4b0652abf620368e0a1251d624708"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-03 10:07:54.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-03 10:07:54.000000000","tz":60},"subject":"CI: lower memory for baremetal compute test","message":"CI: lower memory for baremetal compute test\n\nRL9 baremetal tests often fails with OOM on qemu process\nCirros should be enough with 512MB\n\nChange-Id: I09076ff9daf3e45f2c91f57319c661278aa2f9ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/83ca4bf1a0c85b94bc0129051fd310b88caa695f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/83ca4bf1a0c85b94bc0129051fd310b88caa695f"}]},"branch":"refs/heads/master"},"007e0213d8de12f267100363bbdcabc69f028242":{"kind":"REWORK","_number":2,"created":"2023-02-03 10:16:37.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/872638/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/872638/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/872638/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/872638/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/872638/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/872638/2"}}},"commit":{"parents":[{"commit":"094852e3c6d4b0652abf620368e0a1251d624708","subject":"CI: add Rocky9 upgrade jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/094852e3c6d4b0652abf620368e0a1251d624708"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-03 10:07:54.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-03 10:16:31.000000000","tz":60},"subject":"CI: lower memory for baremetal compute test","message":"CI: lower memory for baremetal compute test\n\nRL9 baremetal tests often fails with OOM on qemu process\nCirros should be enough with 512MB\n\nChange-Id: I09076ff9daf3e45f2c91f57319c661278aa2f9ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/007e0213d8de12f267100363bbdcabc69f028242"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/007e0213d8de12f267100363bbdcabc69f028242"}]},"branch":"refs/heads/master"},"381824fa55c33ad5c76d19484ac6dbca1f930b43":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-02-07 10:48:43.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/872638/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/872638/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/872638/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/872638/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/872638/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/872638/3"}}},"commit":{"parents":[{"commit":"47d675558dc373083bf0d733c16e9aa272c02e84","subject":"CI: Use nested nodepools for jobs that are spawning vms","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47d675558dc373083bf0d733c16e9aa272c02e84"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-03 10:07:54.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-07 10:48:43.000000000","tz":0},"subject":"CI: lower memory for baremetal compute test","message":"CI: lower memory for baremetal compute test\n\nRL9 baremetal tests often fails with OOM on qemu process\nCirros should be enough with 512MB\n\nChange-Id: I09076ff9daf3e45f2c91f57319c661278aa2f9ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/381824fa55c33ad5c76d19484ac6dbca1f930b43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/381824fa55c33ad5c76d19484ac6dbca1f930b43"}]},"branch":"refs/heads/master","description":"Rebase"},"e2641c5901ac42fc34076a636530861b795f5f76":{"kind":"NO_CHANGE","_number":4,"created":"2023-02-07 11:38:49.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/872638/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/872638/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/872638/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/872638/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/872638/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/872638/4"}}},"commit":{"parents":[{"commit":"dc618e139dac9caddd2c3ab441b9de3ad6cd6e9a","subject":"CI: Use nested nodepools for jobs that are spawning vms","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dc618e139dac9caddd2c3ab441b9de3ad6cd6e9a"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-03 10:07:54.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-07 11:38:49.000000000","tz":0},"subject":"CI: lower memory for baremetal compute test","message":"CI: lower memory for baremetal compute test\n\nRL9 baremetal tests often fails with OOM on qemu process\nCirros should be enough with 512MB\n\nChange-Id: I09076ff9daf3e45f2c91f57319c661278aa2f9ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2641c5901ac42fc34076a636530861b795f5f76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2641c5901ac42fc34076a636530861b795f5f76"}]},"branch":"refs/heads/master","description":"Rebase"},"3927df091a7e249f18bbe82f18208ac0d7fb27c0":{"kind":"REWORK","_number":5,"created":"2023-02-07 14:45:37.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/872638/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/872638/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/872638/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/872638/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/872638/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/872638/5"}}},"commit":{"parents":[{"commit":"094852e3c6d4b0652abf620368e0a1251d624708","subject":"CI: add Rocky9 upgrade jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/094852e3c6d4b0652abf620368e0a1251d624708"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-03 10:07:54.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-07 14:45:20.000000000","tz":60},"subject":"CI: lower memory for baremetal compute test","message":"CI: lower memory for baremetal compute test\n\nRL9 baremetal tests often fails with OOM on qemu process\nCirros should be enough with 640MB\n\nChange-Id: I09076ff9daf3e45f2c91f57319c661278aa2f9ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3927df091a7e249f18bbe82f18208ac0d7fb27c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3927df091a7e249f18bbe82f18208ac0d7fb27c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
