)]}'
{"id":"openstack%2Fkayobe~873373","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I666d7011bde0050ebc509b427c1d4f5a66b6231a","project":"openstack/kayobe","branch":"stable/yoga","topic":"systemd-networkd-vlans-stable/zed-stable/yoga","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-04-12 17:19:43.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-04-12 17:19:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I666d7011bde0050ebc509b427c1d4f5a66b6231a","subject":"Support configuring VLANs with systemd-networkd syntax","status":"MERGED","created":"2023-02-10 14:29:01.000000000","updated":"2023-04-12 17:20:59.000000000","submitted":"2023-04-12 17:19:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873373-systemd-networkd-vlans-stable/zed-stable/yoga","cherry_pick_of_change":869128,"cherry_pick_of_patch_set":1,"meta_rev_id":"933b886ddb925698a697a0d273b832d3d09b127f","_number":873373,"virtual_id_number":873373,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-12 17:19:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-04-12 14:18:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-02-21 14:20:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-02-21 13:55:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-04-12 14:18:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-10 14:29:01.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2023-02-10 17:58:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-21 13:54:32.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-04-12 14:18:15.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"4ca18c019030e359693bd9ddf4b7fe1747c19dcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-02-10 14:29:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a6419d01aa465bdb361560d35f5a577b8adaa4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 17:58:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/234809542491441097e83964026c112f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48b6b55dc65e4294866f9933ff4a9f3b : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d7938cfb6734a508dafbbad9c711ee6 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fc074a452b54b7397894b260ca4fbd2 : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e886faf789a436d81e6b1f1b296e931 : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca1eab13a0ad4aad9fa9584c9596bfb2 : SUCCESS in 10m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/540349f729234414a841bf56065b1b84 : SUCCESS in 3m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f891597580a7405ba68213e8b3aa444e : SUCCESS in 6m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7cc0975a79b840cbab283c5aacb1066a : SUCCESS in 17m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2b8f4dd5511144cfb65b3f86209bbe13 : SUCCESS in 25m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8c8a77ff72ed485b99dceacbf425cb69 : SUCCESS in 1h 08m 42s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/bd0bed2ea8ac4170aada3c882a3dcba6 : SUCCESS in 1h 09m 38s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad841e5af18c4387a3876fbbe06519de : SUCCESS in 1h 05m 48s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/edeae451681b4c1e8a2989ccdf01a072 : SUCCESS in 54m 04s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1a56320845da4d32b74e3fde58a8f7b9 : SUCCESS in 24m 46s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/e6601bafbaef4e8ab229d8222a471a47 : SUCCESS in 19m 37s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f88d9c1025a449ba9658aa264775838f : SUCCESS in 15m 32s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42b387a3b13247549365ed3d6a6d61ba : SUCCESS in 27m 05s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/41a0f6feb0db4ba49b1d0c962d9c1dd4 : SUCCESS in 1h 12m 58s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a8f00158d34d4b3b854d43f901b9a812 : TIMED_OUT in 2h 39m 43s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e47bb15bbdf4e89b2e3368f0983d294 : SUCCESS in 1h 25m 52s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6050c55c607546fbad9c1f22f877b178 : SUCCESS in 22m 11s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/b6946b49a6de4f808a134f7dc1d6180b : SUCCESS in 19m 31s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6acada66deba4e2fba08dcdcc308da10 : SUCCESS in 18m 02s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e0588fde2e64003b7b9756d07e658b7 : SUCCESS in 27m 07s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/006de39b4bbc40c8998c4beda8eebf5e : SUCCESS in 38m 26s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28867c5ff4e64681a78ddd580b505922 : SUCCESS in 31m 44s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ed2e025844b1469db67310acfc6a2bde : SUCCESS in 15m 47s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/d0d9f32102bd4baebff929e2bfa89afc : SUCCESS in 17m 29s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88500f6d3a57441598d0011b49e93a3c : SUCCESS in 11m 36s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4fc0096fcb84762a575f3d7058e11ec : SUCCESS in 13m 00s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/942915c41f0a4e53acb2caf941c9fcf7 : SUCCESS in 20m 33s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/7767db79a18d4d3db505f75baa643dd9 : SUCCESS in 16m 11s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d2fefcd453e4f4fa07ab8bde67e9c8f : SUCCESS in 14m 19s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe9e66be131544e1abd8f3c9539263ec : SUCCESS in 13m 50s","accounts_in_message":[],"_revision_number":1},{"id":"ea492e0837dee26ae727dfcb44e209af99615a57","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-02-10 22:07:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"70d6ec18bf34fcea5f959b5c10e0af39a0de7130","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 23:56:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c1a07c028811485582e2adbda792b21c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/764b68332dd740f29364321e3fe0af5c : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b68e10475694c27a571533bffca9327 : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/594210b9a56c482b99138065aafcd566 : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/426f8ce2f85e432b9ad8f28180d01c2f : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12a0f2008d694db8b1e7949ffca63117 : SUCCESS in 9m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c888645e14145d397e47aa0ae2ac59a : SUCCESS in 3m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b56d43dc4e834c569f64840e8ecf208b : SUCCESS in 5m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2c75673f51454fbfa92bd0ca155860d4 : SUCCESS in 14m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/591752cbae6148a6b1397d781216fa4f : SUCCESS in 18m 11s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a01fff251c374ce6a7a699c06c0d41d1 : SUCCESS in 1h 02m 59s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/dfd53bd07679455fa18d63d95aba8e4c : SUCCESS in 1h 18m 58s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca7084daded74595a56860f2a18360d3 : SUCCESS in 52m 00s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/227f5a5ca21249768ee98426a7a593dc : FAILURE in 43m 55s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d664a4e11a1548eaa72ba888c4b56c99 : SUCCESS in 23m 31s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/56dfcb22e0314cf19e7ea58e2bb21737 : SUCCESS in 19m 16s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ad47b2c1d194fbc849da0869179b308 : SUCCESS in 17m 09s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e744ccd153b4a3da5fc0df60830c124 : SUCCESS in 17m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9d4b4ee7652447aeb0e6f94e7d49d5c3 : FAILURE in 52m 31s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5d4429ae0f76497eb9caec63ee2e1761 : SUCCESS in 1h 43m 04s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1b04d9922e44a9abce8f8c41685aa04 : FAILURE in 55m 25s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7591fd75d7f04429859c91047979f0cd : SUCCESS in 22m 03s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/2dc49d02d86f4591a881170c735b47a8 : SUCCESS in 13m 35s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6e398f86389457b893be149b7dc7374 : SUCCESS in 16m 42s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/535d6eed36024aac9d286fca87264d37 : SUCCESS in 17m 33s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6a60e6256b1e460eb204582411e5c3ef : SUCCESS in 38m 54s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42ac33f416f4405e9104c3aae1c839e1 : SUCCESS in 29m 24s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4112298890524f298d1238e9ce4b3f14 : SUCCESS in 18m 48s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/ded3da85cfb24c74b30e091caab356fa : SUCCESS in 15m 51s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ec5ab6c887640dbab1fd4e93ba4ba80 : SUCCESS in 13m 26s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd8ce71c3b974cb0aa04df66480cc6c3 : SUCCESS in 13m 51s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/95754bf13ece445ebba7c91a9466ebeb : SUCCESS in 18m 40s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/cf6bf5c3742541c7aca592c0a002b90f : SUCCESS in 18m 26s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c62d06326e94422ad1ba74f70afe82d : SUCCESS in 15m 05s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e6929d58fa045dda408e858f424a228 : SUCCESS in 13m 25s","accounts_in_message":[],"_revision_number":1},{"id":"fa98eb169da206530c60e7685dd606e0b6b3a613","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-02-15 14:08:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"00206d17b9ab60b5aac2d4a47b7e4f4d8992f345","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 16:13:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0b77ffab4b54e16b6b875090c704ec4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e2bb4cea6e3444a857de3e44b90e3c0 : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37dabc58b06b4e2fb83ac62094864fb0 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d29f526dabb44e8bf3df2ca081ac5ca : SUCCESS in 10m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42258f535c804704bc708a82d259ba11 : SUCCESS in 12m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98dcc8d46fd74569bf459b9f90a60e84 : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64d6f9d1471a4aadbd7c9e58980504c8 : SUCCESS in 7m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a943a3703b174bd7a665d31f0a62231f : SUCCESS in 7m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/56fdc268ec4c4b279e6de9bd7ca272bf : SUCCESS in 14m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/25e20d41887342e39aa6a01de8bd963c : SUCCESS in 17m 57s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/dbdc43073baf40569a78521227139e6b : SUCCESS in 1h 11m 16s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/f18684235c9a4803afaf93333a413cd2 : SUCCESS in 1h 18m 27s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bdbef2c3c134146959a71851fe923af : SUCCESS in 58m 34s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e84ecd31933c4d73bd1ddb8a3aa5a67d : SUCCESS in 57m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/724fa4e73a6042b98ed3dc128ca5995f : SUCCESS in 25m 45s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/fed1690391c04590ab9f890b5c15d605 : SUCCESS in 24m 59s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06aa8db4a1ed46479c998cb0d70745a0 : SUCCESS in 15m 44s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc06c83f0e244aefa881efeb83d9bf7d : SUCCESS in 15m 50s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f7a8efb04f81497db1643521565b4298 : SUCCESS in 1h 11m 33s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a2b2b48f21fb478fbaa305669e7b5fa6 : SUCCESS in 2h 04m 30s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7191868c7a944026b6e7a7ba7ce23e20 : SUCCESS in 1h 11m 19s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ef405ab14e1a411d9aa15b5ceb8598f8 : SUCCESS in 29m 01s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/f351d4aa71914bfeb2d583b28e499589 : SUCCESS in 16m 46s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96e56abf4e924e968d7a35ce5afce43b : SUCCESS in 14m 56s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9b6ba6807634743b09582b70d69e0b8 : SUCCESS in 19m 27s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/171f498bbf4c4c019ac9c0b3cda047fe : SUCCESS in 41m 37s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a4ba5b02bde4825bf3543c050c1380e : SUCCESS in 32m 11s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8ad1e07d4f2042dbb0d67624fcc3e470 : SUCCESS in 17m 05s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/1fe279e9056f4712be79f3018d9f8968 : SUCCESS in 15m 52s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/892ae41b76c0440296dfefb935e99a3b : SUCCESS in 11m 54s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/071ce88af4b54de78c515a3cd7c09191 : SUCCESS in 23m 42s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/a77554da3e2f4e3cad530c713cdb5914 : SUCCESS in 18m 10s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2107d80300d34c73b803d233b56d3d38 : SUCCESS in 20m 23s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89e1f73bd40b4732bc45f01fd8db0bce : SUCCESS in 14m 09s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c8354e3089b47e4818ad7d8e24a4ede : SUCCESS in 12m 52s","accounts_in_message":[],"_revision_number":1},{"id":"cc9d0a2fbd6c5ff92ce8ea7b1b801840fa30b55a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-21 13:54:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"72fa3e7fd94ec6b2e2f879541a7636f4624c5c8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-21 13:55:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d76ac583107d17348ce81c6a3014e4db73039852","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-02-21 14:20:16.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"57da5b3b2f35087a0bfd97f27b4fa527d9f63081","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 15:42:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9815db430054b62b139870e07f9168d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3081ab8f23f84b9c97c622f702a05a7a : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/541458ab71dd4efa99923656626d582f : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4be962192c34425aa98bf2c4c3fd0ee0 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/418cbdef24ed417eb47c49560f435403 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9691a63afb444a3935f3e0f15808b43 : SUCCESS in 12m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ce26414b1c347bf93992f7b707e1982 : SUCCESS in 4m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1205c822e9f7480181ee712d29292dec : SUCCESS in 6m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1037344fa250478baa9203abcbe6f60a : SUCCESS in 11m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b730fee7cf1a446ab8e6df8655b3bb54 : SUCCESS in 15m 57s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/17e7ff07cbf84425b60ff40d54a40f28 : SUCCESS in 1h 14m 11s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/dfa7379c58d74a76862a9717ed213a32 : SUCCESS in 59m 01s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8df7eb5944334751b4c0cec8a00ccd8d : SUCCESS in 50m 15s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df970e4e205741388cfaaa33011c2fea : SUCCESS in 54m 47s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d357b0e046ac4e3ba6db522c1b73fece : SUCCESS in 20m 55s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/2da02e3844ce4b9fa313a5c07a537885 : SUCCESS in 20m 15s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d098ccd17a140a29f7f83f99e89122a : SUCCESS in 12m 29s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b931cf761c644bf4b982e32bfc2e53e2 : SUCCESS in 17m 15s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/127d3822cf944505841eeb3049cef199 : SUCCESS in 1h 15m 00s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1d8c5bde3de24a93b14fc04ba1c6ada8 : SUCCESS in 1h 46m 29s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0dd26e917adf47e293206a6aa9736a0d : SUCCESS in 1h 04m 41s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/937a62770cd147ec97d006259dd14065 : SUCCESS in 21m 51s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/1ebd13af29da4bd695c7e1b45a4fa781 : SUCCESS in 19m 25s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/faba6dab1a564c82ab6941f7e45a1517 : SUCCESS in 13m 08s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9bba4750de38445f9f85607e45373b0f : SUCCESS in 20m 01s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5c10543b918b439eb9b61c556b246160 : SUCCESS in 38m 39s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9be7a01ec31d4d3a8208528a326bb356 : SUCCESS in 32m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/211fa1c726fe4d919d2c34e572162430 : SUCCESS in 18m 31s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2deebf60d7d14759a996bd352d92d073 : SUCCESS in 14m 23s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b73c4b2b37c49d48f697b031e77712e : SUCCESS in 14m 07s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50256f34dcd54191bcfac2431688d218 : SUCCESS in 10m 29s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/ac40809d10e24dadbb97e5a58627962e : SUCCESS in 17m 16s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/ff13d449da3d49fab1554b0487403bc2 : SUCCESS in 15m 12s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6929665c71e044acb8745d77b2f6196a : SUCCESS in 11m 51s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/433f13525e0744b18189b0bc8f4b8af4 : SUCCESS in 12m 52s","accounts_in_message":[],"_revision_number":2},{"id":"6cedd32a66415e4e25d5e843511fda4035ce9aab","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-12 14:18:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"83322796afeb1981f9102cfe21c4b30175de4104","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 14:18:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"93cb03b911863d142156ba828308e37e3e47f377","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 17:19:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a037d2eb47064c83a3306cb575a4a8bd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e826ad3ad2f4e1393de05a4a3d95384 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77d83ddd8381404dbd0f5c5dbb757068 : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3f5216ccda24df6be37c9dbfa6d6f00 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5877174cd153463fa008934a1741a661 : SUCCESS in 10m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70c594791974444d8e3534b8be6b6faa : SUCCESS in 3m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3e8d89146a75404ebf4d97e502112dc8 : SUCCESS in 6m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/086640356db54893b7ee40c951d12705 : SUCCESS in 14m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/804fb2bf6e144e0484ab6d39c47dccb8 : SUCCESS in 19m 15s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/500c4a8fab8e45d7b56c78c7416c0d51 : SUCCESS in 1h 12m 33s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/3a062479077249218f04ffc535e13500 : SUCCESS in 1h 03m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5de8f155e47a46a791a0197cc4f18543 : SUCCESS in 1h 03m 39s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ed2489f282a4f749e672b24717a2885 : SUCCESS in 57m 10s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/353e928643f648eea9a6014b94b92ab7 : SUCCESS in 1h 02m 43s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a406c1f0ad9041de8078362aebf8677e : SUCCESS in 24m 09s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/91ec459556214b9c8e3e2d5139cc3923 : SUCCESS in 21m 35s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/bc5cdb5a84164404a113272e510b2620 : SUCCESS in 21m 41s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0b483f082ad41c68b908ac9b3431b8d : SUCCESS in 21m 13s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8392c727ecd54a5ba42ac17df117770a : SUCCESS in 16m 54s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/88eb552769db429da5479304e7f08f58 : SUCCESS in 1h 20m 31s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a659d17b569f4b948f4d5ae3fde4cd91 : SUCCESS in 1h 07m 26s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0fdb44cd977c4ea2829b4aa929c79097 : SUCCESS in 1h 43m 09s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/556f5a70aa2f4648b9e5d472701be51e : SUCCESS in 1h 15m 56s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d9c8fedba1844c49a2e0ca291629bb3b : SUCCESS in 25m 10s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/1a2a8be73f8e40bba000d8c0ac0faa2f : SUCCESS in 20m 15s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2a9cbe4051a045b79fc939d2b4a8ce58 : SUCCESS in 21m 44s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a0bfaf83f4e41e6837cd7f921b72bc4 : SUCCESS in 18m 06s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3673ec95ab3b400db9f7db8c285f4eba : SUCCESS in 18m 19s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/18b329302bfa4762a25724170b54b6fb : SUCCESS in 29m 39s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afdc2760189b4915916f850db5bdec34 : SUCCESS in 26m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d975aea750c941f3a0ba0b4ce1e9dd73 : SUCCESS in 16m 42s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/bdc30d4bcece44ca83b3b8d8e2b82ebe : SUCCESS in 16m 28s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d388023635164ab6983bb9be5accfe89 : SUCCESS in 16m 56s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cfe61a7f75e453589e473dac1c4366a : SUCCESS in 14m 30s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1f3151dde17421c9ca8f441daf30b8c : SUCCESS in 14m 37s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/0363a2ef7ee146c6b58240c0bcda78bc : SUCCESS in 19m 39s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a73baa277deb4aa3bb83291b746d763b : SUCCESS in 20m 22s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d3feadba25d84c70b8ffa0a993421536 : SUCCESS in 19m 11s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75c747489f83496796a8f2117affb706 : SUCCESS in 17m 57s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc84191f7d084e59b46d11470654a915 : SUCCESS in 15m 12s","accounts_in_message":[],"_revision_number":2},{"id":"3bc2e7cd2a11f3a201312b5c12364920b6ed4979","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 17:19:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"933b886ddb925698a697a0d273b832d3d09b127f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 17:20:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8ea173ee08648f19dee940f525563f0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cb5e638f54f41bc86a3a54b5f1698fb : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbfb9fbdbfe544a889b7e464b24e441a : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8c2dec1229bbdf5f13a31ae959b74be982ff10dd","revisions":{"c97d27b35e346407a3d23af014e38a242f8dfdaf":{"kind":"REWORK","_number":1,"created":"2023-02-10 14:29:01.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/73/873373/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/873373/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/873373/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/873373/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/873373/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/873373/1"}}},"commit":{"parents":[{"commit":"6ee465288c0177388b802a4866470bf403fbb028","subject":"Merge \"Replace references to Docker Hub by Quay.io\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6ee465288c0177388b802a4866470bf403fbb028"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-09-09 15:57:11.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-02-10 14:28:20.000000000","tz":180},"subject":"Support configuring VLANs with systemd-networkd syntax","message":"Support configuring VLANs with systemd-networkd syntax\n\nThis allows operators to configure arbitrarily named VLAN interfaces\nusing systemd-networkd.\n\nStory: 2010266\nTask: 46178\n\nChange-Id: I666d7011bde0050ebc509b427c1d4f5a66b6231a\n(cherry picked from commit 6d7b8812ae82b481519818a00293bb1aec32d058)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c97d27b35e346407a3d23af014e38a242f8dfdaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c97d27b35e346407a3d23af014e38a242f8dfdaf"}]},"branch":"refs/heads/stable/yoga"},"8c2dec1229bbdf5f13a31ae959b74be982ff10dd":{"kind":"NO_CHANGE","_number":2,"created":"2023-02-21 13:55:26.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/873373/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/873373/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/873373/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/873373/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/873373/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/873373/2"}}},"commit":{"parents":[{"commit":"6ee465288c0177388b802a4866470bf403fbb028","subject":"Merge \"Replace references to Docker Hub by Quay.io\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6ee465288c0177388b802a4866470bf403fbb028"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-09-09 15:57:11.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-02-21 13:55:26.000000000","tz":0},"subject":"Support configuring VLANs with systemd-networkd syntax","message":"Support configuring VLANs with systemd-networkd syntax\n\nThis allows operators to configure arbitrarily named VLAN interfaces\nusing systemd-networkd.\n\nStory: 2010266\nTask: 46178\n\nChange-Id: I666d7011bde0050ebc509b427c1d4f5a66b6231a\n(cherry picked from commit 6d7b8812ae82b481519818a00293bb1aec32d058)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8c2dec1229bbdf5f13a31ae959b74be982ff10dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8c2dec1229bbdf5f13a31ae959b74be982ff10dd"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
