)]}'
{"id":"openstack%2Fkayobe~873374","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I666d7011bde0050ebc509b427c1d4f5a66b6231a","project":"openstack/kayobe","branch":"stable/xena","topic":"systemd-networkd-vlans-stable/zed-stable/yoga-stable/xena","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-04-12 18:56:20.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-04-12 22:27:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I666d7011bde0050ebc509b427c1d4f5a66b6231a","subject":"Support configuring VLANs with systemd-networkd syntax","status":"MERGED","created":"2023-02-10 14:29:30.000000000","updated":"2023-04-12 22:28:55.000000000","submitted":"2023-04-12 22:27:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873374-systemd-networkd-vlans-stable/zed-stable/yoga-stable/xena","cherry_pick_of_change":873373,"cherry_pick_of_patch_set":2,"meta_rev_id":"24a3cd07938b81e92f94dbc33a1bbae1688fc127","_number":873374,"virtual_id_number":873374,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-12 22:27:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-04-12 13:09:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-02-21 14:19:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-02-21 13:56:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-04-12 13:09:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-10 14:29:30.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2023-02-10 20:02:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-21 13:56:09.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-04-12 13:09:58.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"029de5b4375fa073ebd54024768d7130fc4b9642","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-02-10 14:29:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5396cb31830c856c1e38c0fd59d611dc94fe1501","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 20:02:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42187a3f26f34cb5a3a6a196ea2d22bf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e29a7bf81ca4a1fafc2f8e48c63d5c2 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81b6c90681884370b1d309e5990dcea5 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a5e01cd39884a30ba14ae2da697ce6d : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7207a8ba47746afae6a6b8875212838 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3530fc60ec03491a95e1ee8c9d23b596 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa0d438ae96048a78ec11e96ea279e60 : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6274728713446d79fe672a0a1485b1c : SUCCESS in 3m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9f9fb4664bff4abab8cfba71f386f107 : SUCCESS in 5m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b33c42be8b824a5b9b83322956737cc0 : SUCCESS in 11m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/557dddf53337446eb84ae84d5022d68e : SUCCESS in 18m 11s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b9349e50bf0441319caca12b8b2ad298 : SUCCESS in 1h 21m 47s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/d9cbbd84502b4f00a81fb68fca8cec06 : SUCCESS in 1h 12m 56s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35198b03bb0d4ff294002472d49e7e4e : SUCCESS in 52m 22s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/64c5557a28cf48b1a6a303e660d10f44 : SUCCESS in 1h 03m 11s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9f67a7b6884242eb833f9c8b60378aa5 : SUCCESS in 25m 27s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/77685395464b457f8c73c0695c5722a1 : SUCCESS in 24m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62596722fba3454291b4755647467727 : SUCCESS in 14m 00s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/198b2df5b90e4c539102b96b32dbce53 : SUCCESS in 2h 01m 46s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/791bf5cbfc294aa8a35d96b9f9ac7fb0 : SUCCESS in 1h 13m 20s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e63cb76c3fd44f598d1b26ffad2a9008 : SUCCESS in 18m 59s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/488e369288d640d9b71530bca0283007 : SUCCESS in 18m 01s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2825bf747d44b81a4cc7939f1b1d2e1 : SUCCESS in 14m 07s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/8228e29a589d4487b407246c77c797c8 : SUCCESS in 45m 12s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/39961f12204344359f54e427b17f65e1 : SUCCESS in 28m 29s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0278d9dcdef847e196eebda46187128a : SUCCESS in 20m 11s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/83fefeb89e5e467497753ef23805f723 : SUCCESS in 17m 54s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/9a2d19bbf8554b4b9a78927b86a73dca : SUCCESS in 14m 46s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee9051a887204701bc956d9f6df20a98 : SUCCESS in 10m 55s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/58160f6b77e74d05b481dc2f6e68462f : SUCCESS in 20m 35s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/5de1159715a74cda8ff5cc91ad3ee67c : SUCCESS in 16m 19s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bbeca2e94a94db4b544fd3cd2c1a2fc : SUCCESS in 12m 22s","accounts_in_message":[],"_revision_number":1},{"id":"32b5c147e8a53b7dd4e7173ba017bd7b1331d526","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-21 13:55:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8b529449f2003e3a455879bfc04446729309df5b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-21 13:56:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"aab12c5c19f64fbf3479f90833d1e84ceb301e90","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-02-21 14:19:07.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8da4944f6ec16b7053c32c00c2fed959fa2323d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 16:20:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/820fff45eb1542d8b086d7132006ad98\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/428251c132f24bb8bcc3b99ed80b92e5 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e4cca52f0e643f79ca55aace72311f0 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/421dac7a768c45918d31da3c0a0e8c74 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ce8fa08779b4f25bbfa84d57ff9e632 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09f49aa6031a4041903a9fc85091dc8e : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d342684264544e1985f44b4acd5b8542 : SUCCESS in 8m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8926b2d94cde4099a79c055fc53ab368 : SUCCESS in 3m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a0b5433607c241f491ae4e1abeca6ce1 : SUCCESS in 5m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/69b633a29d784a029a121471881520e4 : SUCCESS in 11m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5040d6f9af9f4753a3c2e62dd34f7ff6 : SUCCESS in 18m 03s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/816dc74c33d242459f11a35a3261e802 : SUCCESS in 1h 07m 34s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/d1b9fdc4f0e44db0b4f9c9db6cb71690 : SUCCESS in 1h 01m 24s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db540f207b424f1caca6d43d04b6513a : SUCCESS in 1h 00m 04s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e33261c605fd4dacb8b636327fac16af : SUCCESS in 1h 08m 24s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a51171c71f6748139f7c0b63207911ce : SUCCESS in 28m 42s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/90e72014ba55471daf0a0b25999b96f1 : SUCCESS in 26m 50s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e70c23dfc6c74d0cba7b01071ee25eb4 : SUCCESS in 15m 29s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e2cad73d550b47d1909bebfc5b9d3cbd : SUCCESS in 2h 08m 43s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbb48ace1dee4a358043e55de6930c53 : SUCCESS in 1h 07m 04s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/539b9b291ef24d2e8ca83d1b977aecff : SUCCESS in 19m 20s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/f67597ba90194f47a4b5e94a22f15b4e : SUCCESS in 18m 07s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94acca6ca6c847f6bf43335d035cddbb : SUCCESS in 13m 19s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/6f81c37ec6c94d1da5a178ffd4dda35a : SUCCESS in 46m 24s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3bb0136fad494a5aae5a6372d33df323 : SUCCESS in 27m 30s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cf68d71f510420cb8333b44a0d06e93 : SUCCESS in 21m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b352ea4fc2564a7e9138cf281aeca926 : SUCCESS in 18m 37s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/1e49a744f61c41acbcec5d9c2bfaafca : SUCCESS in 20m 41s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/663caa963d22428f8daccf06ab336ec2 : SUCCESS in 10m 05s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/231c73bfb8fa42febd3e393230e48f0f : SUCCESS in 20m 31s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2d8926b2d9b645d5b741837084507c19 : SUCCESS in 16m 33s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba8b199994054ba79049c2bc3af4e35e : SUCCESS in 11m 17s","accounts_in_message":[],"_revision_number":2},{"id":"72e693d4d4e97e6aee0046f84950e198792f3525","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-12 13:09:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d741e424b0d3b159f2b5ad830b5eae147810503f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 13:10:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"899569c5c5fdb0ef5a0ce75160a1c5eee077eac8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 15:14:50.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a258901e2b43489d8d6f173671528c54\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/228410894e694406babb3689cf3ff2e6 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da618624256b429aa6b104c601dbeb01 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdaaf1fdb043440aaab6c7c678d1b15d : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4b367db6a1e4e43a51e22ded4abf200 : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/056c748eac4541bba911a3b9f9439663 : SUCCESS in 3m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2cf5b5ada68a4dc88b0eb817efca1c07 : SUCCESS in 5m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/813d61b2274e478db33e09cafed48e85 : SUCCESS in 10m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ac5ea02252184218b5bdd3b3d9dd1226 : SUCCESS in 18m 35s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/542244317e3b441ebfa39a850bc55499 : SUCCESS in 1h 20m 34s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/10b697a2fef64ea3ba25de244465a057 : SUCCESS in 1h 02m 56s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9bae054708074c3c80e993f7cb96d516 : SUCCESS in 54m 54s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f397db9bbd6d441f8b8fdc94a1ec5476 : FAILURE in 1h 57m 06s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0ec78469ed7e499f9c651b5b5488f018 : SUCCESS in 26m 04s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/5d794586d6fc4e0db995ee18ce4d38fa : SUCCESS in 24m 26s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b331d56c2544439a2a5e43b12b425fb : SUCCESS in 15m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/24341cf963234aa5bafcb483438631bf : SUCCESS in 1h 55m 23s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0759e0fd299a4a26b8815ee6feef7c42 : SUCCESS in 1h 05m 17s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/477d554659f04bf0a794ae27252d1b4e : SUCCESS in 18m 01s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/0a3f0282f2f04d84a3aee38b6cfa95df : SUCCESS in 21m 34s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21e3cbaf668e4236a1032f1450d02c6b : SUCCESS in 16m 33s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/23692bfd741a4ef2bc707cbdc129eb9e : SUCCESS in 32m 37s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5df1fc91a5c14b6f95ac947806b8aca1 : SUCCESS in 20m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2b3b5fa0d57a428187b3b6050f6aada8 : SUCCESS in 15m 57s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/95c0118e79ef4c8dae9d0c3bd9add507 : SUCCESS in 16m 31s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e855c7dccc8c4bb9a74f5bf0d9264bb2 : SUCCESS in 10m 45s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/cd878bb262ae448990817ab305d0b272 : SUCCESS in 17m 51s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a1353380125b488a88bd93600ce583dd : SUCCESS in 18m 41s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7cd4f8c1084143e5a7b9da3bf3a53566 : SUCCESS in 14m 33s","accounts_in_message":[],"_revision_number":2},{"id":"9fbfdc9906448d8c79f20b0f1c361e989b7b20a8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-12 15:29:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d4e8165d6c3d105acc982c01836144b3668113e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 17:38:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1d0f7facab341bab3f9c34cb1bad22b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2b0762f5ca94d92b9c46856764c0250 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e32ecb92de740dfb6d81699f35d7248 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c2f554f869e45d29dcf9d02959d81f8 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d299319ebb84c2da265dfb8f1a54ee1 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44a11561cc4240e4a81060c0dd022208 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2359d9e5f27b4c62a90c5b618e83ef63 : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd89759449d3461882e6060673bed819 : SUCCESS in 3m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/812629a184b64846be93e8561cc77cfe : SUCCESS in 5m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/16da84fdf6654451b094b61cb68cb8bb : SUCCESS in 12m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7b9830d8a9fe4912b2ac2bda1da04403 : SUCCESS in 16m 29s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f4f520b71abf469fa65dc9a84e4e01c7 : SUCCESS in 1h 06m 14s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/2b3e9b6469aa4e78923bf4c2f581a640 : SUCCESS in 59m 26s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33898725526a4205b46fc1e699d8d77a : SUCCESS in 52m 13s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1424d167b4f9435f95d3e019ed86f95d : TIMED_OUT in 2h 04m 00s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1f9b48a541764ff9ad8c74a310cd887e : SUCCESS in 27m 44s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/3d3b01af2b1842b389a3eddecf8e5208 : SUCCESS in 26m 05s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddd2ff04ed31473c8feddde2d3809dca : SUCCESS in 12m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f26ce2cd32aa46ba8218be8a4c279ab3 : SUCCESS in 1h 50m 46s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea4d6fe1a7bf46ee931de14f385f5368 : SUCCESS in 1h 12m 57s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a311936ee9c7400dbb6a9212e1b0e572 : SUCCESS in 17m 46s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/84a4dbf6abbc42be864e7ae92380d2a7 : SUCCESS in 18m 02s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f1161c0cdd747ceafdd2f763c22cc7f : SUCCESS in 16m 41s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/adfe944a0b8a4cb1b2a4b63886d130ec : SUCCESS in 42m 01s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6088dd8978334470a652cc1cf16b4a33 : SUCCESS in 26m 51s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3be1ab5f953046b69cb34febe2c2e582 : SUCCESS in 21m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e62d8c79c9b946b9927e53e58117037f : SUCCESS in 19m 13s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/43c3e40b2e324b62845530bbd788f22f : SUCCESS in 13m 45s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6f8034618b04bf78a4b29fdaa584e7b : SUCCESS in 12m 53s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/588e268301f64444b37fa98c0b5db3a3 : SUCCESS in 20m 21s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/187f1d6ebae1443a99adb460bafe2db1 : SUCCESS in 17m 48s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fe928bbd9f949c0bff0eb88f74cdb24 : SUCCESS in 14m 09s","accounts_in_message":[],"_revision_number":2},{"id":"c7d16e042b1f9d13283d882cbb56283c65034f9d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-12 18:56:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5c6b2b3ac23ad06e7ef8b0b38ca45bcf37f3cb05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 20:37:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f049dcc36920495f9222f1fce50e07b4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aaf6e2d7db1c4bb98f00b7340544e355 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7cc6d0622364ae5b113a960c297850b : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1d63d0d888b4afe87c3b1c57fa148fc : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f985242f77dc40ea95d1bfe4a80d097b : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17040448a5834d4aa53ece331e8fcbed : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/033e71537492437f82fc379e67441d78 : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db53cad1afec437e9dcf52c09f8e5811 : SUCCESS in 3m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a3a5d32900224500a6f6b1feab7bf8ed : SUCCESS in 6m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/71d077b18330445aa2568022b1a259b6 : SUCCESS in 10m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5ab7da31d9bc471ab53c63e444122bd1 : SUCCESS in 21m 51s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ffde2333d8594d1aa5449553654a1c5a : SUCCESS in 1h 06m 33s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/f4cabbb9f35646aca4f63e54622e4648 : SUCCESS in 1h 03m 06s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5365908c0d8417783654339d0763dbf : SUCCESS in 56m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d8cc8cf2efdd49f0bc40d24d1d0f51b5 : SUCCESS in 1h 09m 37s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0e1068dda4a14e8e942b92250a67e717 : SUCCESS in 20m 34s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/5291f543d8574d2c9c89412c3ed71085 : SUCCESS in 30m 13s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aaa6ea2dc2094e60ac2c0df44dae532e : SUCCESS in 16m 09s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0d51c26ef3a84bb69a4a854014344e6c : SUCCESS in 1h 35m 25s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/688658e3cd554feaad2bde8ab2063d1d : SUCCESS in 1h 23m 40s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9e34bb1156e341d2b39ba1c8a0812aef : SUCCESS in 20m 14s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/689a60d68a30463d926eebb795ed7adb : SUCCESS in 17m 03s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8014a28e52448069da7c941128dd5de : SUCCESS in 24m 52s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/eb8a73273f1f4721b91d9ca27d81b0db : SUCCESS in 37m 06s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2b4a7e3f71cf4e14ad19b6fd691260e8 : SUCCESS in 34m 45s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/851ef100edcf4287820eb579bdc060c8 : SUCCESS in 22m 38s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0307c84b284048faa7459147e99e1840 : SUCCESS in 15m 42s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/93a0069bc2f44e8caad495c8a53f96cf : SUCCESS in 16m 17s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d32acb736eb24a22b86a9f9ac134d66a : SUCCESS in 12m 01s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/b7764d3bb72548f096c89429f0f7ae54 : SUCCESS in 19m 22s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2735c70abab34a10b5ce13327b4fc8c9 : SUCCESS in 17m 17s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b82e0d06fff64b988e629390757419e4 : SUCCESS in 15m 31s","accounts_in_message":[],"_revision_number":2},{"id":"d7a870b94fb0f596c179f149c517b2e58fdc0c26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 20:37:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3af3d8828067972f967810a20fcdd5d150d6c9d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 22:27:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/937af5687c9b4937ac724cf4127dcaab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5c4e932bf454c788c8728ac35cbdfb8 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10b1a086034d4b35bdef1095a9aeded1 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94afc6b20be84c6389ee7acd00198100 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e297bed4b634996904dc25c8942f016 : SUCCESS in 8m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a60d43526a45437b83bdab3291eed62f : SUCCESS in 3m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c9f00963779b421a91d746fb62f01729 : SUCCESS in 6m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8a184e5d7f984e77850db4ece09a62fd : SUCCESS in 11m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a9044594dfc04ec6ab508e1656ebbc66 : SUCCESS in 19m 15s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d027a937517240cabf53a3bee81abd6a : SUCCESS in 1h 12m 56s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/4a129c13f68e47b5aa7700135dd37d09 : SUCCESS in 1h 09m 50s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ecf7537954a40adbc3f5bdbde7e8db4 : SUCCESS in 56m 43s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e0e97edcda9c425f99c9557c90453ed1 : SUCCESS in 1h 12m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6098825957334ffcb6d38546d973fe18 : SUCCESS in 25m 18s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/4c09f6437a9e4496b5412317c85baadc : SUCCESS in 23m 58s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b3fc720d2924d4194c0a9430732517b : SUCCESS in 16m 41s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fdd3f29eae0f46e393dfd2b4fedd5657 : SUCCESS in 1h 46m 18s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b1d2cdcfb2e4beeadd94448f734d8a9 : SUCCESS in 1h 17m 01s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/14dc74fe85b94afa9aa7e67becb90a29 : SUCCESS in 22m 04s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/649337b6a70a42a48018d701a33b14b8 : SUCCESS in 15m 14s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/daadcd88ac3b4c5f942712ed7a941655 : SUCCESS in 14m 03s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/32217967c18d4f4d9acb3c709f18d94e : SUCCESS in 33m 08s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15bcd30928bf47b2b17ebdf1599a8209 : SUCCESS in 18m 38s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/788f38146787458d830c2f56fa5d8589 : SUCCESS in 16m 41s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/814b2896784c456e8d1154ef58a0065e : SUCCESS in 12m 17s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cfe28745487a48b5a92e26930bd16fbb : SUCCESS in 11m 49s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/36c0680684e34096acd503099c451dbe : SUCCESS in 20m 10s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/3f3b543681994d6b9dd3da9ccad37332 : SUCCESS in 14m 06s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c32a035e0b6475a9c0ce2d14b4e37ec : SUCCESS in 14m 30s","accounts_in_message":[],"_revision_number":2},{"id":"bff644fb78407d080663b300bb89a0475093d7f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 22:27:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"24a3cd07938b81e92f94dbc33a1bbae1688fc127","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 22:28:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9ba8fb91de04e7b9d91fbea69b76bd4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2882440645e64c9b80ff1f920d113863 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4236e2049f51454595e86e2e54c1bd7e : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3fb24c5d4f3ce3fb4f7483e9497b56dd26569683","revisions":{"1e2d74da06ca1adeb884235369cc7d5910a10d99":{"kind":"REWORK","_number":1,"created":"2023-02-10 14:29:30.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/74/873374/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/873374/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/873374/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/873374/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/873374/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/873374/1"}}},"commit":{"parents":[{"commit":"931c5f594c1cd3336fe083f0a1604b73b9eb0e92","subject":"Replace references to Docker Hub by Quay.io","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/931c5f594c1cd3336fe083f0a1604b73b9eb0e92"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-09-09 15:57:11.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-02-10 14:29:09.000000000","tz":180},"subject":"Support configuring VLANs with systemd-networkd syntax","message":"Support configuring VLANs with systemd-networkd syntax\n\nThis allows operators to configure arbitrarily named VLAN interfaces\nusing systemd-networkd.\n\nStory: 2010266\nTask: 46178\n\nChange-Id: I666d7011bde0050ebc509b427c1d4f5a66b6231a\n(cherry picked from commit 6d7b8812ae82b481519818a00293bb1aec32d058)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e2d74da06ca1adeb884235369cc7d5910a10d99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e2d74da06ca1adeb884235369cc7d5910a10d99"}]},"branch":"refs/heads/stable/xena"},"3fb24c5d4f3ce3fb4f7483e9497b56dd26569683":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-21 13:55:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/873374/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/873374/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/873374/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/873374/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/873374/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/873374/2"}}},"commit":{"parents":[{"commit":"f698b9b5b47c9d8a3d535824e36d2353a06a4ca7","subject":"Ubuntu: Fix networkd config for multiple VLANs without MTU","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f698b9b5b47c9d8a3d535824e36d2353a06a4ca7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-09-09 15:57:11.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-02-21 13:55:59.000000000","tz":0},"subject":"Support configuring VLANs with systemd-networkd syntax","message":"Support configuring VLANs with systemd-networkd syntax\n\nThis allows operators to configure arbitrarily named VLAN interfaces\nusing systemd-networkd.\n\nStory: 2010266\nTask: 46178\n\nChange-Id: I666d7011bde0050ebc509b427c1d4f5a66b6231a\n(cherry picked from commit 6d7b8812ae82b481519818a00293bb1aec32d058)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3fb24c5d4f3ce3fb4f7483e9497b56dd26569683"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3fb24c5d4f3ce3fb4f7483e9497b56dd26569683"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
