)]}'
{"id":"openstack%2Fkayobe~874119","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I8e25fdc459102ca6e13bd3d38191554ba32c1f48","project":"openstack/kayobe","branch":"stable/xena","topic":"multiple-vlans-without-mtu","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-02-20 12:05:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8e25fdc459102ca6e13bd3d38191554ba32c1f48","subject":"Ubuntu: Fix networkd config for multiple VLANs without MTU","status":"MERGED","created":"2023-02-16 17:46:27.000000000","updated":"2023-02-20 12:06:10.000000000","submitted":"2023-02-20 12:05:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874119-multiple-vlans-without-mtu","meta_rev_id":"8a385b5df23a26f91cc984f628ae3f8838b96508","_number":874119,"virtual_id_number":874119,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-20 12:05:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2023-02-20 09:37:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-17 08:04:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2023-02-20 09:37:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-16 17:46:27.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2023-02-16 20:39:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-17 08:04:24.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-02-20 09:37:20.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"31dbb312bb9a70887b093f8b07de7bb63a395a71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-02-16 17:46:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f93da6c93bea377860d134a6edebeb1eb1cafd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 20:39:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/946c534b4e20472cb552f0834340fba2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38a530d1aa694d6b89205adbd6d9d07f : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7901d7d3a544a9d9bf843f83af71657 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65f3607d0a7a4bfe8b91c299ef2484f1 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c22ef991a0f94e7cb9e17bbd0d3098dc : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1e39aa964c0446fb1450a9afa3d3c40 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62c7117bdc1c4bacbea5e64aa873ca89 : SUCCESS in 9m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0caa53709ada4fce88c7bda073815a92 : SUCCESS in 3m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ff1f304110344a49bcc27ca2a0d45a31 : SUCCESS in 6m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f690e7831f5e44a8b2fc3b156383ffd0 : SUCCESS in 13m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3fa59af74cbd48c080e2e480c5f79b32 : SUCCESS in 18m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/351b8c1543c64e2e93bb85d11e74c798 : SUCCESS in 1h 20m 58s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/f4df20f11d044ee7b38dbbfae14eed96 : SUCCESS in 1h 09m 23s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a8500c0f3bd40789f37f0f0bfc63909 : SUCCESS in 56m 43s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d52b4df0afa14801882015b11664ea79 : SUCCESS in 1h 15m 00s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2e350abe5d91417a8dcc1231f9503e9d : SUCCESS in 32m 23s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/e1e72eb6b37143398214120e8c5cfc1a : SUCCESS in 25m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/732f721a872e46d09f4c4447dc35eeeb : SUCCESS in 15m 41s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b7e8aaf019df44e6a50218d251bea9b2 : SUCCESS in 2h 10m 40s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b23388b09e734427b3b0412ded4c1c56 : SUCCESS in 1h 10m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/fb6377d92d7c482f8a5fbf60d241be11 : SUCCESS in 20m 02s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/5c8691d42aba4ca0ad37345b05566eeb : SUCCESS in 15m 35s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a07db640ac004da48f141bd4b42e6b94 : SUCCESS in 15m 39s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/702978c9e1104ff38b22f6070d576bb1 : SUCCESS in 53m 29s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/71827f2e93fa4007813fa9d38985c4e2 : SUCCESS in 33m 36s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c66a3aaa63e34656acb740e8eca32ea7 : SUCCESS in 23m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/68c6fff291bb4ad69bc91b833462e363 : SUCCESS in 16m 46s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/3faea356b9774c24a97da7e24af9b1e0 : SUCCESS in 18m 29s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/798c24a7bc484a3ea24ee4fd674d3e4f : SUCCESS in 13m 02s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/7cfe167f98514f728e9804f4177ee1be : SUCCESS in 22m 02s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/683980b5dcb54e33ab061cda7dae2609 : SUCCESS in 19m 10s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5080c067f7a440b783fb9778d6b495a7 : SUCCESS in 14m 23s","accounts_in_message":[],"_revision_number":1},{"id":"a989a46efefa77690845868a20c577a335f9c093","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-02-17 08:04:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5016dc563bba60d6fc7661785b7ce606c922475d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-20 09:37:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"58c43915883c9ef1bc84e1e751750e24a0b035aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 09:37:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9de540a88a0fe4d882e8ff057d66a662bda7a6aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 12:05:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/982f776e09c04ccc966f73d4d3acfddc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77b2c3f2fd0c40cc915ea6bbd4d241d2 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2d7177df8ad4359ad8dd219ac82e032 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f90c37581e7c432ba244763eec0a6912 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b52f4de80f3846869c673be9b806213f : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63fc0136376e49c6b990ef2f9520a2ab : SUCCESS in 3m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2ad97c3bd3744d33880648d823936e4f : SUCCESS in 5m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/74677cc19f1946eeb476dfc3d5464b65 : SUCCESS in 12m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f6eea2b3a265483c92362ba73f54be9c : SUCCESS in 18m 55s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/dba09f2fe1264eaabf849601f37ada56 : SUCCESS in 1h 14m 49s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/0c4cd06db341438582612d1fa0dd61e2 : SUCCESS in 1h 05m 21s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bf085e22fdc4997aca287d1109ac976 : SUCCESS in 44m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/64c8c800001c402d96e78ff602ec2c38 : SUCCESS in 1h 15m 38s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/92545efe86694c9b8a9fe06ad98b5622 : SUCCESS in 25m 43s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/d0f4cea5eda54a2c8be43f8f0439212c : SUCCESS in 19m 24s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91a9dc9a39754b4286745ea88fcd44a1 : SUCCESS in 20m 07s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8a34533c004f4151a3dfcba342a770d5 : SUCCESS in 1h 58m 03s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32820565676e4f55935ec0b67fe97e14 : SUCCESS in 1h 11m 51s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/76eb67386db74a59b9371785ac7a8e26 : SUCCESS in 21m 42s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/46da680724bd4d0b8490c365ec3078a4 : SUCCESS in 17m 07s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0c7232d41794ab7aa510df85fbda5d6 : SUCCESS in 15m 17s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9d160045b85b43d78c2185b64c0c3689 : SUCCESS in 32m 52s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25b9bf9e2f494988a7378d92b4fa5c87 : SUCCESS in 22m 45s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4091f5de5257427b970e3f3267a016c8 : SUCCESS in 18m 11s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/ec1c477ea632475b9ad81d0d7cb7d14d : SUCCESS in 15m 28s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0b0dbf0eea248e780ed366ffa5eeec1 : SUCCESS in 9m 41s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c6761a7fb9a34c47937ce0003b179d9d : SUCCESS in 21m 54s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/36c0d0603ad847209c4de165071432e8 : SUCCESS in 18m 17s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4e16e641b5341f7b5f9c6a380a1608d : SUCCESS in 12m 56s","accounts_in_message":[],"_revision_number":1},{"id":"9d8ff2a86e9c1bae38c6b70972cdd9b01d26844d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 12:05:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8a385b5df23a26f91cc984f628ae3f8838b96508","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 12:06:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b1c0268343d4182aa1210ed0fe67c55\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9074fe67f864d068d659597d8bf12c3 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fec9a79d2cf64558916d9df7f74f898a : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f698b9b5b47c9d8a3d535824e36d2353a06a4ca7","revisions":{"f698b9b5b47c9d8a3d535824e36d2353a06a4ca7":{"kind":"REWORK","_number":1,"created":"2023-02-16 17:46:27.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/19/874119/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/19/874119/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/874119/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/874119/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/874119/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/19/874119/1"}}},"commit":{"parents":[{"commit":"931c5f594c1cd3336fe083f0a1604b73b9eb0e92","subject":"Replace references to Docker Hub by Quay.io","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/931c5f594c1cd3336fe083f0a1604b73b9eb0e92"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-29 15:58:42.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-02-16 17:46:14.000000000","tz":180},"subject":"Ubuntu: Fix networkd config for multiple VLANs without MTU","message":"Ubuntu: Fix networkd config for multiple VLANs without MTU\n\nTypeError: \u0027\u003e\u0027 not supported between instances of \u0027NoneType\u0027 and\n\u0027NoneType\u0027\n\nStory: 2009013\nTask: 42729\n\nChange-Id: I8e25fdc459102ca6e13bd3d38191554ba32c1f48\n(cherry picked from commit ead017e1969b70fddabe7390f2727eb0dc34e620)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f698b9b5b47c9d8a3d535824e36d2353a06a4ca7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f698b9b5b47c9d8a3d535824e36d2353a06a4ca7"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
